)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~800163","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I4dbc598ef0d085ac8c5656c52b9239ee69a1f4d1","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"spell_error","hashtags":[],"change_id":"I4dbc598ef0d085ac8c5656c52b9239ee69a1f4d1","subject":"Fix spelling error","status":"MERGED","created":"2021-07-09 07:42:48.000000000","updated":"2021-07-15 09:25:57.000000000","submitted":"2021-07-15 09:24:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800163-spell_error","meta_rev_id":"a09b409e4222ce3f3da65c4eaf93c5982cf63e0e","_number":800163,"virtual_id_number":800163,"owner":{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-15 09:24:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-07-09 07:48:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-15 07:48:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-15 07:48:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-09 07:43:57.000000000","updated_by":{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-07-09 07:43:57.000000000","updated_by":{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-07-09 09:13:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a46b29828db228dadb05e6f7f8dd0f7271ebf04a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},"date":"2021-07-09 07:42:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ebb8448e592d30fd8ed537ac3788fa66523989a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-09 07:48:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ebb2c51b1bd2fab60b90da087b662d4d337c8d06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 09:13:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16f315fe6acf4cebb9abe647cd417eb1 : SUCCESS in 16m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c0540586eb18463c8343d6f1d985c6a1 : SUCCESS in 1h 21m 45s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/dc102f78b4894f69b237ec005f7ac2de : SUCCESS in 1h 11m 59s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/88bb591619844ae8ba22f178315148af : SUCCESS in 1h 29m 08s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e2592fc92ca450186c24e11520b83e0 : SUCCESS in 1h 26m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efacfd454e144a179c80d6e3f1ca4068 : SUCCESS in 1h 28m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4553837c0f243a8ae454e7ec27a6cc6 : SUCCESS in 1h 19m 42s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fccd63e27b524864964c46819b31572f : FAILURE in 1h 09m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4a0855a62c84dd3a8ccbf1b37487c5e : FAILURE in 1h 02m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adcbc678fa0741a7857385f4c43540b6 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"da56cadfab8757fc6c4827f1dd8c4d45762b922c","author":{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},"date":"2021-07-09 09:20:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"00ea7d322bdec04b68049f1572b1e6fb364372c3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-09 09:36:35.000000000","message":"Patch Set 1:\n\nhm, looking at error feels like we do smth wrong now in upgrade jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27d2ce1aad923b53af9f4178265edf3b73f52c2c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-09 09:37:44.000000000","message":"Patch Set 1:\n\nah, we need to merge backport first...","accounts_in_message":[],"_revision_number":1},{"id":"998923ff397a0a7c827c446335aaa0dcab9bf67b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 11:22:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e99807d9243848f68492c26f6b70c59f : SUCCESS in 18m 33s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8293863b69bd45729bcfbbed0eeb6f91 : SUCCESS in 1h 18m 40s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3705cbd66eff4c90bedb4b2c69d6375c : SUCCESS in 1h 17m 38s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ae4f358fdea046bfb4b129722c37ab8c : SUCCESS in 1h 57m 58s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0338775e643d41acbe6f5a7bb031b9ff : SUCCESS in 1h 24m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/767ba3cf2fc144fb87075157ac0566ae : SUCCESS in 1h 24m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bced74e087744e28d169f74e60cfb06 : SUCCESS in 1h 36m 39s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1ad77bf061944c44ac860065fa23ef73 : FAILURE in 1h 10m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73f21474e8a54193a286d9c0386f6d82 : FAILURE in 56m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e389291645994bd89cbdd0c794773dbc : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1e8ec685e771b7785f0f3d73675184288c563f09","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-10 04:35:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"07e77bb6fabbbe4556bc1a80df9f7b865587ca9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 06:19:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18dc60b86c1545ae8b7bcf130b697788 : SUCCESS in 11m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f7244830f33a4c5fa21a9fbc634ca85f : SUCCESS in 1h 18m 20s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/90c7f488dff54787bd5301fce22fb2de : SUCCESS in 1h 01m 10s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/520f3e0628794c6c8c0931310b4d6cd9 : SUCCESS in 1h 33m 30s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/10a667a4fc6b476a94c3019350694792 : SUCCESS in 1h 16m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b8700e3caef478495c9f720ff71c43e : SUCCESS in 1h 18m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a757d6c8d864e4b99c921dd85ece50f : SUCCESS in 1h 15m 22s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/eaeae25b76254b3dadbe430a0e967512 : FAILURE in 1h 08m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1842eab7b3a47ed9f2ca127e40f6cc4 : FAILURE in 52m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0328c2a40b3446648a6f0a5e4c60cdc0 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":1},{"id":"5400daf7b9ef9b192d1250d0d0636c0e907cf360","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-12 11:07:48.000000000","message":"Patch Set 1:\n\nfeels like we need W bump as well 😞","accounts_in_message":[],"_revision_number":1},{"id":"466701faea9a6e02185cb14f759becb4bfd67e1b","author":{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},"date":"2021-07-12 12:20:09.000000000","message":"Patch Set 1:\n\n@Dmitriy Rabotyagov can you please elaborate the last comment. Thanks!","accounts_in_message":[],"_revision_number":1},{"id":"074992e77d8ef78a03e80144a02d05e9941736bd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-12 13:33:25.000000000","message":"Patch Set 1:\n\nSo, we have SHA fixed in W https://opendev.org/openstack/openstack-ansible/src/branch/stable/wallaby/ansible-role-requirements.yml#L169\n\nSo to get upgrade job working, we would need to update that to include https://opendev.org/openstack/openstack-ansible-os_octavia/commit/19bf040b67f03111cb205cb41bf03b5f5f2c205c","accounts_in_message":[],"_revision_number":1},{"id":"1690f456d577856d527e710b4502bbe32431a781","author":{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},"date":"2021-07-12 13:39:57.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e So, we have SHA fixed in W https://opendev.org/openstack/openstack-ansible/src/branch/stable/wallaby/ansible-role-requirements.yml#L169\n\u003e \n\u003e So to get upgrade job working, we would need to update that to include https://opendev.org/openstack/openstack-ansible-os_octavia/commit/19bf040b67f03111cb205cb41bf03b5f5f2c205c\n\nACK.","accounts_in_message":[],"_revision_number":1},{"id":"8e71f04148bf1e9da90e339187e845b94bedeb9d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-14 11:24:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"33fb98edc19b9ea8eced9160c1f37b4cc2527cd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 13:33:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/399912608336436da3b9e68da5b2f36f : SUCCESS in 17m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e0057d0c485d4922aa3e79d0923d597b : SUCCESS in 1h 17m 01s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f246d3175360488d8d70ec19f3acd453 : SUCCESS in 1h 12m 09s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/86e088f2f26d40798d14d419930e8022 : SUCCESS in 1h 26m 41s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a00e4d2be1941c0a13b77928cb37838 : SUCCESS in 1h 28m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eba788f5c339417bba8b959c020c10f4 : SUCCESS in 1h 27m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b04a3ffbc2564fcea1038773c8d9bd86 : SUCCESS in 1h 16m 22s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2a37d525f5f345ebac86cd9d96de7234 : SUCCESS in 2h 08m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8753c4f237c24d3aaa30128bc908c33f : SUCCESS in 2h 00m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bceac563f6a64e4093e2326fbacab696 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3929a25211628ff9baa384c0019ae3f7f67fc012","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-07-15 07:48:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b900e2663b082f06a223a86bd59c3d838b0e3e65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 07:49:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"532b1573a9f1c786b63546be3794b99893361fee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 09:24:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7300f65cacff418ba863c0b4aa150693 : SUCCESS in 12m 15s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8f5f5e204ded449ab8ea3a1cbabce43c : SUCCESS in 1h 30m 14s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0ed02f56868742cbb09b44d8d63cd6d1 : SUCCESS in 1h 28m 14s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c988fa880ac949de9574e9eb771834d8 : SUCCESS in 1h 25m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d64564072bdb42dd847fcdff1924dcc5 : SUCCESS in 1h 25m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/234a570f4df64938a8e2f5e8e9239b11 : SUCCESS in 1h 06m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46ff45d85636451cb4fae10df4cdfd04 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":1},{"id":"163ac408edc9b33887f3d54527543f85923802ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 09:24:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a09b409e4222ce3f3da65c4eaf93c5982cf63e0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 09:25:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f196879d83c1453aa8a8587c8642db51 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2de42c7bb85edb8a5ec07bf99418b6db8fd027e","revisions":{"a2de42c7bb85edb8a5ec07bf99418b6db8fd027e":{"kind":"REWORK","_number":1,"created":"2021-07-09 07:42:48.000000000","uploader":{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},"ref":"refs/changes/63/800163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/63/800163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/800163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/800163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/800163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/800163/1"}}},"commit":{"parents":[{"commit":"b37ab7592e0d19b3c2f97e5cf5acc751313ce9bb","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b37ab7592e0d19b3c2f97e5cf5acc751313ce9bb"}]}],"author":{"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","date":"2021-07-09 07:36:47.000000000","tz":330},"committer":{"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","date":"2021-07-09 07:42:30.000000000","tz":330},"subject":"Fix spelling error","message":"Fix spelling error\n\nThis patch fixes the spelling of octavia which is misspelled as octaiva.\n\nChange-Id: I4dbc598ef0d085ac8c5656c52b9239ee69a1f4d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a2de42c7bb85edb8a5ec07bf99418b6db8fd027e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a2de42c7bb85edb8a5ec07bf99418b6db8fd027e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
