)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~820093","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~stable%2Fwallaby~I89649503d5918c33f0d1e4200be67be5e0ed8a9e","project":"openstack/openstack-ansible-os_octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2021-12-02 12:57:13.000000000","reason":"Andrew Bonney replied on the change"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2021-12-02 14:48:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I89649503d5918c33f0d1e4200be67be5e0ed8a9e","subject":"keypair: copy key to deploy host rather than setup host","status":"MERGED","created":"2021-12-02 10:20:46.000000000","updated":"2021-12-02 14:50:40.000000000","submitted":"2021-12-02 14:48:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820093","cherry_pick_of_change":816997,"cherry_pick_of_patch_set":2,"meta_rev_id":"936db453dc55c01646738931ac58bf4448770a00","_number":820093,"virtual_id_number":820093,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-02 14:48:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-12-02 12:57:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-12-02 11:57:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-12-02 12:57:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-02 10:20:46.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2021-12-02 11:57:20.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-12-02 12:37:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7f11131b307602d461389206bf7a88a562737887","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-02 10:20:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"08d2c671396a645d44363271d038971fbfbaad4f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-02 11:57:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2b047ecd29b3f1d60c55ab35a94910952c023a18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 12:37:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07b084e36ab74a3ca7c9e7e16a4c4534 : SUCCESS in 14m 13s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c1017362be3c4ecd8d05595a70663572 : SUCCESS in 1h 21m 29s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/25da323e52c74ed89df66286817291e5 : SUCCESS in 1h 13m 10s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/48b0b7a6823c487ab38dc2176bb5c7ca : SUCCESS in 1h 43m 08s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b13ada0312804a27b8581ae3669adae8 : SUCCESS in 1h 28m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e19ea15a427a422b9cdd744cd38000f4 : SUCCESS in 1h 26m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f13be555807a48168f56961726c7630b : SUCCESS in 1h 27m 39s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/24f26352a45d465b8a4df1bc3d78e8ec : SUCCESS in 2h 09m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fde1de8175d48dd9132f96b13ff82aa : SUCCESS in 1h 50m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4d2854aad0b44fe8cb95def9e348422 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3843de0b2cc0335204fe3c39909f143bbaa65a94","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-12-02 12:57:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"473fe9c58bddd16435d146ef2b4021b86f69a156","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 13:01:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d15e2fa34e30ccb4924b7bb6053425d708f72037","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 14:48:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/188ba1cbbc364bafad3b055edee5e1fd : SUCCESS in 13m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/18c96667c06e4b8ebd58b925a68fd972 : SUCCESS in 1h 22m 16s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/52b1b7815a404a4786ece233f8904655 : SUCCESS in 1h 41m 24s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f6a5e98a4c0143f28f6321e50557cacf : SUCCESS in 1h 37m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9719b6d667f040efb9ce181b12c7ac55 : SUCCESS in 1h 29m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6b40d2613a246aea09954c24072d62d : SUCCESS in 1h 25m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e0778c6d18f4e53a5e0023c2ceb3948 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"11faac167228126212db2b88de66b0c78e3bf312","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 14:48:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"936db453dc55c01646738931ac58bf4448770a00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 14:50:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b8257a7899a44739ff3480c871d7094 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29287119f58fadf5f0122fa8f023b9d92d2636c9","revisions":{"29287119f58fadf5f0122fa8f023b9d92d2636c9":{"kind":"REWORK","_number":1,"created":"2021-12-02 10:20:46.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/93/820093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/93/820093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/93/820093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/93/820093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/93/820093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/93/820093/1"}}},"commit":{"parents":[{"commit":"99c246e80963f13bf6c21f0e53200a6fb793c21d","subject":"Fix self-signed certs distribution","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/99c246e80963f13bf6c21f0e53200a6fb793c21d"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-11-08 11:00:38.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-12-02 10:20:46.000000000","tz":0},"subject":"keypair: copy key to deploy host rather than setup host","message":"keypair: copy key to deploy host rather than setup host\n\nThe env lookup in the key copying task refers to the deploy host.\nIf your deploy host user is not root then this task will fail\nas this directory does not exist on the remote setup host.\n\nAs the original intention appears to have been to copy the result\nof the keypair generation to the deploy host, the delegation is\nexplicitly set to \u0027localhost\u0027.\n\nChange-Id: I89649503d5918c33f0d1e4200be67be5e0ed8a9e\n(cherry picked from commit ffac17201a735115267648b657399ad9fc4be9ae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/29287119f58fadf5f0122fa8f023b9d92d2636c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/29287119f58fadf5f0122fa8f023b9d92d2636c9"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
