)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~820220","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I7804ec93d6ec82249f4d81ccec3ab02c4bc8a233","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"db-pooling","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2021-12-04 21:26:07.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2021-12-04 17:04:46.000000000","reason":"Dmitriy Rabotyagov replied on the change"}},"hashtags":[],"change_id":"I7804ec93d6ec82249f4d81ccec3ab02c4bc8a233","subject":"Database connection pooling improvements","status":"MERGED","created":"2021-12-02 18:15:34.000000000","updated":"2021-12-04 21:27:37.000000000","submitted":"2021-12-04 21:26:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"820220-db-pooling","meta_rev_id":"ed891c2b6e181c375a336f614de31715c39a6da1","_number":820220,"virtual_id_number":820220,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-04 21:26:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-12-04 07:36:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2021-12-04 12:38:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2021-12-04 12:38:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-02 20:10:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-03 16:31:00.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-12-04 12:38:06.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"}],"messages":[{"id":"2def98f5236d29c61547b22f04b91e35236b3d8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-12-02 18:15:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"affed204011ae15ecfa5a1b3935884d039b80183","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 20:10:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce7adeaee192415baf914574f8734195 : SUCCESS in 14m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e397ce296ddc4f85905a26ddd94965a1 : FAILURE in 1h 20m 48s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a79188f19f8444f182fd40f72005b477 : FAILURE in 55m 32s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/61f2b6d0a4e4463ab807e6f1857c68a5 : FAILURE in 1h 09m 06s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/647d3b9439c24d06a83a22b0f87e3fe0 : FAILURE in 1h 13m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72b0e2f32d94459ba44f085b3893be08 : FAILURE in 1h 27m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2c82f06152d4aaab45dd3238756f589 : FAILURE in 1h 23m 43s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a60a7225bde4dfea60de3fe96c631ee : FAILURE in 1h 51m 02s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/895442af036f4771a3306251648a1f2e : FAILURE in 1h 37m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b33e8207bf148978d35efca4a0be595 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":1},{"id":"1af9f8cbb031735911630a289a32e6d606f71e26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-12-03 10:41:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e7ced8af18ada1bb5c5e1aa3c953f756abad8a56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 13:16:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13ca863805c745fb9c1fd4c572d0f389 : SUCCESS in 14m 04s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b6cd89ed92a5474db9c0c0f9b57a603e : FAILURE in 1h 30m 30s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/57914ec5083545c8af2b87aac1c5c57b : FAILURE in 1h 14m 06s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1200a821774e40a78bf9b5216361c4cb : FAILURE in 1h 26m 07s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8d4bb524a99545e0bafd19e958d14ade : FAILURE in 1h 33m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb966b27d06b4d248ab204fa3a995142 : FAILURE in 1h 40m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc197d8b74ba4dc0b7e1277fa363bcce : FAILURE in 1h 32m 36s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/bdad888fde4e4359bf6bb3803faf0e77 : FAILURE in 2h 12m 18s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/542994065caf42d7bea3db3a4b6cc1b1 : FAILURE in 2h 13m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aa93f833ba947fb991229ce91aaa831 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":2},{"id":"a1230499fe321535ce33183598dee551d37b1c24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-03 13:47:18.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"fcea552edfe44865165063ce0ca53d2680e6b078","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 16:24:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65255a2e26624a73a2e8f5afe62e450b : SUCCESS in 18m 18s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/37c2292e26c64308bd3fc33e77dd1830 : SUCCESS in 1h 39m 50s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/11fb66468f21452992ddc0a1577e31d5 : SUCCESS in 1h 22m 34s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/39f654cba4594bde9838005537294139 : SUCCESS in 1h 31m 55s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b07a6e1469c44c26a89ca80da7d7401d : SUCCESS in 1h 25m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3235ae99bfcb41c0b89e7381799f66b7 : SUCCESS in 1h 51m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1d186b3db8343dd9c3b37b254626af3 : SUCCESS in 1h 58m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5fe1bdc4a84446fd9252a82973335c1d : SUCCESS in 2h 29m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c34ae9725f854394bf4e05773bc9327a : SUCCESS in 2h 18m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59453033e2754251ac0b0a787531681d : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":3},{"id":"1b1d898057881a4f6d07355bbe4e15b2be1098ae","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-03 16:31:00.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c541946071b4640155b87f946bfb3e9335585891","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-04 07:33:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"696f70f1a0159e908b8db3de508f45da8ec49b64","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-04 07:36:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3abe75a6f0a7088945b4a28a891998cff00bf271","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 10:08:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06bd1cbfeba34cb8a7b656991b3f18d5 : SUCCESS in 11m 19s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/12f111e25d35468a82549e9f7854e6dd : SUCCESS in 1h 45m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/966aad257a5e4cf99c2399a61641f392 : SUCCESS in 1h 15m 43s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1d0e8fd5924f44d4ac9e622f46f677a7 : SUCCESS in 1h 37m 05s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e777fd519c5c425ab90318239503dac7 : FAILURE in 1h 35m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2cdf0c020a6049898649bae34d590f33 : SUCCESS in 1h 40m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44b063d541b94335abc1ac9578d4c5cf : SUCCESS in 1h 38m 05s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/be52817c9886499fb8b26636e03a1eb3 : SUCCESS in 2h 27m 43s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3052da277364c89b1d197d46e4a6219 : SUCCESS in 2h 15m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/360fbca6b1ce42258272326a684bc9e5 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/492fb8c905614f4ba73f4c91e263b4fe : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":4},{"id":"d031fb4ab290e48432ce7c3d776456e7120878e9","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-12-04 12:38:06.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4009948d771f1409611faa2f5ff83004129b5c27","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-12-04 12:38:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f9d484686c3f4655bba15852ae7dbe917646c398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 15:02:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/927cfd6d63ba49b0b4fcff5e32f89f20 : SUCCESS in 13m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8bb8e6e4433c46ef960e83251f9c6ef4 : SUCCESS in 1h 45m 11s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4825e830b6bd49c181da95082d0cfb36 : SUCCESS in 1h 31m 57s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/76f957677afb4a058d44b7145dfdfd5c : SUCCESS in 1h 35m 20s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6b534e88453347b19c8b84f57bcd0798 : SUCCESS in 1h 38m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e05c08401268436c89a4c774014db8f7 : SUCCESS in 1h 49m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09ddc6d020ee4dbab0514abad425f3f7 : SUCCESS in 1h 31m 25s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/bce35c760f3844629c16b5c8ebe409c8 : SUCCESS in 2h 17m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e40c0fd11c194f4d92af799ecf02622e : SUCCESS in 2h 21m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5271018928949d6b8c5cb7966dfb736 : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3f0cc4dacb44ca089fe54315e77f1ff : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":4},{"id":"2267e9297f8d7ce2d36cbcd0149fc0fe730959c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 15:05:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4e8e81f76459067c57f649108a19a908cfff2f40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 16:43:16.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eab12b4483914430ba62efb1a20687d9 : SUCCESS in 17m 52s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c4a6dc0fa055497baa77b209251e8792 : FAILURE in 1h 23m 54s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ea647d536fa04668965d077ed6bd3686 : SUCCESS in 1h 07m 45s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/dd4f55481cb749f8bed97ffa0a57ecf1 : SUCCESS in 1h 20m 15s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/39fa84c543eb4ea4b08f25e01988e9d2 : SUCCESS in 1h 28m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e193585c11ab4cb3b91b07ad7dab08ff : SUCCESS in 1h 37m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dfd7af2117446a2a47e3e3a251161b9 : SUCCESS in 1h 27m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/103737555edd4059bd8f8fb72c3dfcf9 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b17caa718fc146ae8eec2e34c3db2333 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":4},{"id":"c223e79cea38503d9b3d0d10bf285855dd566852","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-04 17:04:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3be9421cf751b83673a2ce3b86c7a68c1028983c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 19:36:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5095fb7e72de43b597762eaa69ba1daa : SUCCESS in 20m 06s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/facc3e6f160048e7988b3eeca4322645 : SUCCESS in 1h 44m 57s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/69ee11d0240747dfadca63696f796460 : SUCCESS in 1h 17m 12s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c03e55a7335340cdb122abf92d3c2173 : SUCCESS in 1h 30m 54s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b2291ccb25104818af0b70f7d2570d24 : SUCCESS in 2h 04m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe786b2bd3b646d0a0aaf135a38c4406 : SUCCESS in 1h 43m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32764a50e84f4e12bf39e30e9532b69a : SUCCESS in 1h 35m 01s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d730ebe28d44612bce9b660b729d058 : SUCCESS in 2h 24m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0fdacb53458472d8991f3adbcbf9085 : SUCCESS in 2h 19m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d701e773ca34185b6d327a9d2ebc544 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d42d09c4648e44b9a684fe62471c320b : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":4},{"id":"bef84bddc8a61349f54d0a895a84e2b650b84be7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 19:38:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9fd40cd5ec03fd814382a4793c8ca0cc8210b11c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 21:26:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a43c34e01524067b7e433a4e29d94ca : SUCCESS in 13m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/36eeea9d055e4538a2aac8305b1787b0 : SUCCESS in 1h 47m 53s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/901ec5d4e94941d086a12158a86e5a78 : SUCCESS in 1h 18m 37s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bc4c85dc27404a30ab6d619564da557d : SUCCESS in 1h 32m 11s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c2a327012b242409cd5299e47843009 : SUCCESS in 1h 32m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26e164a24e1241c180fa6c53bfb2efe9 : SUCCESS in 1h 45m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d7154eec9a54c7c809ea189ac2a120f : SUCCESS in 1h 35m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/400b7ff8e6094d7681f67d715ba5370a : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcc58c1002d544d0896106344594b9be : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":4},{"id":"3b8c19acd00db34f4bcb1907bf680b56743c0360","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 21:26:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ed891c2b6e181c375a336f614de31715c39a6da1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-04 21:27:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b601b298d164eadb660373141439a2b : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6353f2f7479c1fcac3dc056045e870942f99d2b1","revisions":{"848602cfe182aea36540c67b8eb86c04fa2f5e05":{"kind":"REWORK","_number":1,"created":"2021-12-02 18:15:34.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/20/820220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/20/820220/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/20/820220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/20/820220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/20/820220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/20/820220/1"}}},"commit":{"parents":[{"commit":"ffac17201a735115267648b657399ad9fc4be9ae","subject":"keypair: copy key to deploy host rather than setup host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ffac17201a735115267648b657399ad9fc4be9ae"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:11:28.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:15:13.000000000","tz":60},"subject":"Database connection pooling improvements","message":"Database connection pooling improvements\n\n- Implemented new variable ``connection_recycle_time`` responsible for SQLAlchemy\u0027s connection recycling\n- Set new default values for db pooling variables which are inherited from the global ones.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/819424\nChange-Id: I7804ec93d6ec82249f4d81ccec3ab02c4bc8a233\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/848602cfe182aea36540c67b8eb86c04fa2f5e05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/848602cfe182aea36540c67b8eb86c04fa2f5e05"}]},"branch":"refs/heads/master"},"eaeb42eb141f04636a59eb7125c92cbb2954a8c8":{"kind":"REWORK","_number":2,"created":"2021-12-03 10:41:34.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/20/820220/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/20/820220/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/20/820220/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/20/820220/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/20/820220/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/20/820220/2"}}},"commit":{"parents":[{"commit":"ffac17201a735115267648b657399ad9fc4be9ae","subject":"keypair: copy key to deploy host rather than setup host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ffac17201a735115267648b657399ad9fc4be9ae"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:11:28.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-03 10:41:29.000000000","tz":60},"subject":"Database connection pooling improvements","message":"Database connection pooling improvements\n\n- Implemented new variable ``connection_recycle_time`` responsible for SQLAlchemy\u0027s connection recycling\n- Set new default values for db pooling variables which are inherited from the global ones.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/819424\nChange-Id: I7804ec93d6ec82249f4d81ccec3ab02c4bc8a233\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/eaeb42eb141f04636a59eb7125c92cbb2954a8c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/eaeb42eb141f04636a59eb7125c92cbb2954a8c8"}]},"branch":"refs/heads/master"},"f718c276096c148304758097775619e233db28e6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-12-03 13:47:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/820220/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/20/820220/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/20/820220/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/20/820220/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/20/820220/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/20/820220/3"}}},"commit":{"parents":[{"commit":"1a7f8b2a57bf41c85beab323fab454744826efd7","subject":"Refactor definition of lock path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1a7f8b2a57bf41c85beab323fab454744826efd7"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:11:28.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-03 13:47:18.000000000","tz":0},"subject":"Database connection pooling improvements","message":"Database connection pooling improvements\n\n- Implemented new variable ``connection_recycle_time`` responsible for SQLAlchemy\u0027s connection recycling\n- Set new default values for db pooling variables which are inherited from the global ones.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/819424\nChange-Id: I7804ec93d6ec82249f4d81ccec3ab02c4bc8a233\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f718c276096c148304758097775619e233db28e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f718c276096c148304758097775619e233db28e6"}]},"branch":"refs/heads/master","description":"Rebase"},"6353f2f7479c1fcac3dc056045e870942f99d2b1":{"kind":"REWORK","_number":4,"created":"2021-12-04 07:33:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/820220/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/20/820220/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/20/820220/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/20/820220/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/20/820220/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/20/820220/4"}}},"commit":{"parents":[{"commit":"1a7f8b2a57bf41c85beab323fab454744826efd7","subject":"Refactor definition of lock path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1a7f8b2a57bf41c85beab323fab454744826efd7"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:11:28.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-04 07:33:28.000000000","tz":120},"subject":"Database connection pooling improvements","message":"Database connection pooling improvements\n\n- Implemented new variable ``connection_recycle_time`` responsible for SQLAlchemy\u0027s connection recycling\n- Set new default values for db pooling variables which are inherited from the global ones.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/819424\nChange-Id: I7804ec93d6ec82249f4d81ccec3ab02c4bc8a233\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/6353f2f7479c1fcac3dc056045e870942f99d2b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/6353f2f7479c1fcac3dc056045e870942f99d2b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
