)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~824379","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~Id2621d0945b0f92d7ca14a0d4db66be5908d1722","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"osa/common_tasks","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2022-01-28 15:44:54.000000000","reason":"Andrew Bonney replied on the change"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-01-28 22:24:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id2621d0945b0f92d7ca14a0d4db66be5908d1722","subject":"Use common service setup tasks from a collection rather than in-role","status":"MERGED","created":"2022-01-12 10:48:41.000000000","updated":"2022-01-28 22:25:35.000000000","submitted":"2022-01-28 22:24:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824379-osa/common_tasks","meta_rev_id":"b6ac90d1a0f0ca7db44e814d72d5d5840a170af5","_number":824379,"virtual_id_number":824379,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-28 22:24:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-01-14 08:47:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2022-01-13 11:31:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-01-14 08:47:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-12 13:03:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-13 11:02:56.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-01-14 08:47:27.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"80fc4fd9a67d0f9ea90b364f0fd39aa43e7c0b8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-12 10:48:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12c1a8188bc6f888b3110fcb9d7d38595597d1bf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-12 11:31:48.000000000","message":"Topic set to osa/common_tasks","accounts_in_message":[],"_revision_number":1},{"id":"933657c86f95aded3dabd4fe3fe34ad735103024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 13:03:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11deccf8144a4e5493ee1f7e16ce4f28 : FAILURE in 19m 57s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9445028e500f4f0c9926a7346d4f964d : FAILURE in 38m 03s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d118baf94e1549c3903dafbe4d111284 : FAILURE in 41m 50s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ffd63ec214046d5863298472cbbead6 : FAILURE in 38m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7cc43f1eb444b6495a640d44ced7420 : FAILURE in 37m 58s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b240027ae854d6aacc47c6f5d769e46 : FAILURE in 1h 44m 13s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6656cda354e240cea2995326cc1e290d : FAILURE in 1h 45m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d3b4f7d88f64c379423e7519b3a1952 : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":1},{"id":"1eef8acc425f37d34ae50eb08902bb874874b6a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-12 13:14:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"204d417e948fb8c10c23dfb72c23524f11226a27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 16:47:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85a8bd897c7d4b2a8740d763ef992557 : SUCCESS in 26m 42s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/379bdf35fdf94e09ac2e8358e17a17af : SUCCESS in 1h 17m 17s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ed56e49bbec74a76ab671ea7add01017 : SUCCESS in 1h 29m 24s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/3319671ea0a1414c88818aacffeee378 : SUCCESS in 1h 30m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e2cc4544bd54f5b905c641b578c0ddc : SUCCESS in 1h 41m 00s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/3fb356ff24fc454782b0dd4497224088 : SUCCESS in 2h 24m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/727983b3a09d40b19f814217f34030fa : SUCCESS in 2h 34m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3f645728ba94b9fb76b0b17cb4161ab : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":2},{"id":"160cb2ef647b6f83860b1379bd1781311149c56b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-12 17:42:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"11649c57792d73426a3439cd015f7b0d037ddc37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 20:34:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61505121005c410ba3ba5dd9ede25a79 : SUCCESS in 19m 55s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9018e0c87738456d996bb6edb7388549 : SUCCESS in 1h 18m 17s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e2d76581ae1e4e79abbe656cb093d2f5 : SUCCESS in 1h 30m 18s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/fbc7462379aa40fbb733df8b84638061 : SUCCESS in 1h 49m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/589edec2168b4f1c8ca3ab9fde7e97f1 : SUCCESS in 1h 43m 51s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c8248e0a4a842bfa8797887f08e28d5 : SUCCESS in 2h 26m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b8c9cf495e049f39a79075c76f1d594 : SUCCESS in 2h 33m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ad51beedb9f48c783ce731503f047db : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":3},{"id":"c58f9c28b81c191b56a4ebe9c830997e085e2a1b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-13 11:02:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0ed74c23db3d20721ea01c1fac147f25e0ed4cb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-01-13 11:25:35.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"0311d2ba22dc08a617cadec845f30b825215efff","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-13 11:31:53.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d1f534e7d2ab9221641437c077cdd50418ca787c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 14:22:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d83f68c3f81472bbd502cd2d9f771be : SUCCESS in 23m 27s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2039b296919e4438b3fed23969fb3c4e : SUCCESS in 1h 23m 17s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2fc30e8fa76d487f9e5dba0124fbbff0 : SUCCESS in 1h 29m 17s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/47a14ce1e3be450d9f9fda8444110b10 : SUCCESS in 1h 29m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37c7229ea6f549298a640f58d4915faa : SUCCESS in 1h 35m 03s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d10e47cd9e1429686e777a886eb968d : SUCCESS in 2h 15m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcba085946294ccf86f7eed88dfa21dc : SUCCESS in 2h 29m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d6171cc867d43079bbc77a39246d564 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":4},{"id":"8e3d998c411e287f289df2610f09acb9403dcd34","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-01-14 08:47:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e6212fa15cff300eb7f1fde8f3d3a82fc71f0259","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 08:49:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e020ce892bdd7f70fb02dcc32312b9b1c2831ac6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 10:21:05.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dfcb7aa6b164499abc934146569d10c : SUCCESS in 16m 10s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/31e578b0e99847478b2c2f3187352eff : SUCCESS in 1h 14m 47s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8df2262ae4f6449990e913dbfaf49d12 : SUCCESS in 1h 31m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/580a83cbbe4749c7a55dffdcce4f4a0f : FAILURE in 1h 16m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3b93608dd5d42c490fbe73d970b47d5 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":4},{"id":"18efc3a15e1341ba33f26dd4a4c68e18e8c7381a","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-01-28 15:44:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a1c8c16fc5ac914c801232a2422075d71d133e82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 20:23:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05e5f49d75ee4f428fdba2d9417a0a1d : SUCCESS in 14m 43s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/798f0c215d9e461581cbce6653b2bbb6 : SUCCESS in 1h 07m 29s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e8d805359bf47f2b671578aa8207668 : SUCCESS in 1h 27m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d38a0e0575346ca9765b590ed68753c : SUCCESS in 1h 23m 29s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a268ca16c11c418f871ca1722aa069b1 : SUCCESS in 2h 22m 19s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebac91c7023d44809b63636845e8f739 : SUCCESS in 2h 09m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de644888a066406292b19d97e2dcd93e : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":4},{"id":"43e272b3c3cd9da49c643dbdd2d9566954e54390","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 20:28:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ce8240ae3a03970c427b7f0a274c8c1689624af8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 22:24:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1d86b68739f4f8cb56b795113f26efb : SUCCESS in 22m 46s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ee39ad012ed64127a9f4f9ea6697e7ce : SUCCESS in 1h 25m 10s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/797f8b57fdd043d3befa816cda5ca0ec : SUCCESS in 1h 51m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6441812212bc4046a2b9b477013dae30 : SUCCESS in 1h 32m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c5ede3602a0499cacd662779a4b5b53 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":4},{"id":"e3ae96a9c3380e0ac44db4c281530c57fc8b7147","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 22:24:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b6ac90d1a0f0ca7db44e814d72d5d5840a170af5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 22:25:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/868d79a12be445808d3f35b2663a95e9 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4775014ff95598f63949e3cacec6b59980e33b14","revisions":{"fa30d1fba7327c3b528ef37d85e1d84b09d9ab49":{"kind":"REWORK","_number":1,"created":"2022-01-12 10:48:41.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/79/824379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/79/824379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/79/824379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/79/824379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/79/824379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/79/824379/1"}}},"commit":{"parents":[{"commit":"c7a7a14f7bee4d46336ab8a3cc82d29628f32c01","subject":"Use focal amphora test image by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/c7a7a14f7bee4d46336ab8a3cc82d29628f32c01"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 10:48:12.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 10:48:34.000000000","tz":0},"subject":"Use common service setup tasks from a collection rather than in-role","message":"Use common service setup tasks from a collection rather than in-role\n\nChange-Id: Id2621d0945b0f92d7ca14a0d4db66be5908d1722\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/fa30d1fba7327c3b528ef37d85e1d84b09d9ab49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/fa30d1fba7327c3b528ef37d85e1d84b09d9ab49"}]},"branch":"refs/heads/master"},"e44ee5d553cd0fa5f979738c91ad0fa0319214df":{"kind":"REWORK","_number":2,"created":"2022-01-12 13:14:02.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/79/824379/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/79/824379/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/79/824379/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/79/824379/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/79/824379/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/79/824379/2"}}},"commit":{"parents":[{"commit":"c7a7a14f7bee4d46336ab8a3cc82d29628f32c01","subject":"Use focal amphora test image by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/c7a7a14f7bee4d46336ab8a3cc82d29628f32c01"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 10:48:12.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 13:14:02.000000000","tz":0},"subject":"Use common service setup tasks from a collection rather than in-role","message":"Use common service setup tasks from a collection rather than in-role\n\nChange-Id: Id2621d0945b0f92d7ca14a0d4db66be5908d1722\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e44ee5d553cd0fa5f979738c91ad0fa0319214df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e44ee5d553cd0fa5f979738c91ad0fa0319214df"}]},"branch":"refs/heads/master"},"7c5bbc4f7b1f29b6cfae1427e5a0c2da3aefe9f9":{"kind":"REWORK","_number":3,"created":"2022-01-12 17:42:58.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/79/824379/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/79/824379/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/79/824379/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/79/824379/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/79/824379/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/79/824379/3"}}},"commit":{"parents":[{"commit":"c7a7a14f7bee4d46336ab8a3cc82d29628f32c01","subject":"Use focal amphora test image by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/c7a7a14f7bee4d46336ab8a3cc82d29628f32c01"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 10:48:12.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 17:42:48.000000000","tz":0},"subject":"Use common service setup tasks from a collection rather than in-role","message":"Use common service setup tasks from a collection rather than in-role\n\nChange-Id: Id2621d0945b0f92d7ca14a0d4db66be5908d1722\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7c5bbc4f7b1f29b6cfae1427e5a0c2da3aefe9f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7c5bbc4f7b1f29b6cfae1427e5a0c2da3aefe9f9"}]},"branch":"refs/heads/master"},"4775014ff95598f63949e3cacec6b59980e33b14":{"kind":"REWORK","_number":4,"created":"2022-01-13 11:25:35.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/79/824379/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/79/824379/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/79/824379/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/79/824379/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/79/824379/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/79/824379/4"}}},"commit":{"parents":[{"commit":"c7a7a14f7bee4d46336ab8a3cc82d29628f32c01","subject":"Use focal amphora test image by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/c7a7a14f7bee4d46336ab8a3cc82d29628f32c01"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 10:48:12.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2022-01-13 11:25:35.000000000","tz":0},"subject":"Use common service setup tasks from a collection rather than in-role","message":"Use common service setup tasks from a collection rather than in-role\n\nChange-Id: Id2621d0945b0f92d7ca14a0d4db66be5908d1722\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4775014ff95598f63949e3cacec6b59980e33b14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4775014ff95598f63949e3cacec6b59980e33b14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
