)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~839381","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I8644d4999d3b5c10825d84e3f3d28ecd331a60b2","project":"openstack/openstack-ansible-os_octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29605":{"account":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"last_update":"2022-04-27 11:24:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8644d4999d3b5c10825d84e3f3d28ecd331a60b2","subject":"Add support for system scope openrc","status":"ABANDONED","created":"2022-04-26 13:39:00.000000000","updated":"2022-04-27 11:24:22.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9c011d75b9c7e8e83492f2d23c9afded5fb932f8","_number":839381,"virtual_id_number":839381,"owner":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-26 15:45:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":-1,"date":"2022-04-26 15:16:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-26 15:16:56.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-04-26 15:45:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a07827a9f1f84125c6452f84b3f9311c2ad1230d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2022-04-26 13:39:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d06f98e31c048fde819c52b1834b532016e9de8b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-04-26 15:16:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66b81820b468d3c2759761610415c1562ab89024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 15:45:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61f01af19c274863b253af71c1fe7b98 : SUCCESS in 13m 14s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ed262fe8798748d8a00013aafb09f475 : FAILURE in 59m 54s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8fe0de65f7a240f7b25b398dae887a72 : FAILURE in 1h 24m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d16c61f6c5d74b2da504e533c80a969c : FAILURE in 1h 15m 30s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/a6de0441b3de48acbd9a510dc6022710 : FAILURE in 1h 10m 19s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/bd60578a37494028ae0af858026f3d10 : FAILURE in 1h 52m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/beeb9897c853447195ee41d57f91b64a : FAILURE in 1h 58m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/857ade4e66874f938134d1c4ab77632f : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":1},{"id":"9c011d75b9c7e8e83492f2d23c9afded5fb932f8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2022-04-27 11:24:22.000000000","message":"Abandoned\n\nalready resolved in openstack collections","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b8f03687d6074616189687e6d7dcf47e72cb11c","revisions":{"3b8f03687d6074616189687e6d7dcf47e72cb11c":{"kind":"REWORK","_number":1,"created":"2022-04-26 13:39:00.000000000","uploader":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"ref":"refs/changes/81/839381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/81/839381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/81/839381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/81/839381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/81/839381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/81/839381/1"}}},"commit":{"parents":[{"commit":"4b8ba186f83383fb4acc13d9d18a0882f4588a00","subject":"Cleanup setup.py config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4b8ba186f83383fb4acc13d9d18a0882f4588a00"}]}],"author":{"name":"siavash sardari","email":"siavash.sardari@gmail.com","date":"2022-04-26 13:37:29.000000000","tz":270},"committer":{"name":"siavash sardari","email":"siavash.sardari@gmail.com","date":"2022-04-26 13:37:29.000000000","tz":270},"subject":"Add support for system scope openrc","message":"Add support for system scope openrc\n\nadd support for https://opendev.org/openstack/openstack-ansible-openstack_openrc/commit/fdc640ddcbc13de17609005f4ca34cc4067cd5f8\n\nChange-Id: I8644d4999d3b5c10825d84e3f3d28ecd331a60b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3b8f03687d6074616189687e6d7dcf47e72cb11c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3b8f03687d6074616189687e6d7dcf47e72cb11c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
