)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~855074","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~Ic22ec379983e43aa5f2b55fd4543b4aa70762354","project":"openstack/openstack-ansible-os_octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-11-29 21:29:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic22ec379983e43aa5f2b55fd4543b4aa70762354","subject":"Adopt output structure to new collections version","status":"MERGED","created":"2022-08-29 18:14:52.000000000","updated":"2022-11-29 21:30:58.000000000","submitted":"2022-11-29 21:29:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855074","meta_rev_id":"4f0d0268be44249014c46eff68394f4a6ee624f7","_number":855074,"virtual_id_number":855074,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-29 21:29:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-11-29 19:46:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-25 16:49:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2022-11-29 19:46:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-29 20:22:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-18 04:16:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-11-25 16:49:48.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-11-29 19:46:05.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"}],"messages":[{"id":"7d645247dd5b27d40f6e4adbe10b735afec4617b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-08-29 18:14:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf61e77047063ed6022f21441619680f7c0fb766","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 20:22:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7d8130c07d041c7a66a5db62f1ac179\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/095a347d1b72414eb009ed96ef659e0a : SUCCESS in 36m 33s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7cd63bfd09d148cb8c5c864e18632c07 : FAILURE in 1h 11m 21s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b070b2a6d5634d1f98e11b5972f9a790 : FAILURE in 1h 19m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fa33960cd034465969f9ed97cdab385 : FAILURE in 1h 15m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1695e009403460ba9624ab45b018602 : FAILURE in 1h 16m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b02576f79d6e4296a08e52b271c517e5 : FAILURE in 1h 59m 44s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c8de5a568ac4806b74f044fd740abdf : FAILURE in 1h 49m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae7738528d2644639705a582a58f7b24 : FAILURE in 32m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbdc4bf39dbf4bcbac6dc8ff3f4f5eca : FAILURE in 29m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8913604a5f09475a9dfcad4d386c2935 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"6d06b89677f23b0fefbb4fb49dbc21b610c509f1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-11-18 04:16:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3d67610fc8c0e5dd91d4cc884790eb45cba35fbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-24 17:18:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bb9c9669a1559e5461cd2ce729b7d8a140532070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 19:19:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d8b86b7c19548ef9e8c9978d55c608f\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb6f8a07145442528d78b4ebb5036d9a : SUCCESS in 17m 46s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/378d717c8afc4990b0717a3bb3d7ae2e : FAILURE in 1h 20m 22s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2de64a8ea0bb4eb1b63c8c6b261d8123 : FAILURE in 1h 22m 37s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/74c8cfe3627a45b3af2490ad533b0bf1 : FAILURE in 1h 24m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d96005175ccb4e668ee7c42bd8dd6477 : FAILURE in 1h 33m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7bde4f076e74da5803a0c4b001e13a0 : FAILURE in 1h 26m 23s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc05fd3ac45b403a91841e9a927853da : FAILURE in 14m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e5ba68fdf844b84b3466cf653575edd : FAILURE in 1h 57m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5b8955e56a34111a1fcdaa79d9bac9b : FAILURE in 46m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b06b3ac8225a47ce86a810bb115ad319 : FAILURE in 41m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d14f683c857c4e6f850f294dca6cfdd3 : FAILURE in 43m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0538e03dd3814a17a9ad0f0d5fbdf4f9 : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":2},{"id":"deeb45d25e8c41e1193b72294c5e4023ed7b5609","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-25 11:57:50.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d14712c47d6f4b13d990dac6a5627aa28400ecc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 14:18:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0613a54d24104e1da45877f3a31f1512\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3d340c135b94f8bb6344cf7780830a4 : SUCCESS in 14m 12s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3a3a6a849c0845b5b860d3b9a83aff61 : SUCCESS in 1h 32m 41s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/df78278946d645258a82cbd07f07e9d9 : SUCCESS in 1h 35m 08s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9f216d0909454077a4af8e2d57dd6b4a : SUCCESS in 1h 29m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbfb78bb17274892a54b67428e672e5e : SUCCESS in 1h 46m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af7025cb2a4d4e278bd7280aaf0fd5d2 : SUCCESS in 1h 39m 59s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f77b0273a9ac49d1a5dca92b6a7a2ffd : FAILURE in 9m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff14a8de11b34599abe38716a030fb64 : SUCCESS in 2h 13m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f7a6bebb23c4967b9df5896ef0e17ed : FAILURE in 45m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a2e401cb5cd24c06b32ab07845d809aa : FAILURE in 34m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6820b21c4014b6db04eef0f5e85f617 : FAILURE in 39m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4172a8d8bc4d44799196c523a5eeef39 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":3},{"id":"e759035771f91bf1a6af8a8726b8c67db23a8360","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-11-25 16:49:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"19a07fce7b5dddf27fc46b1b12ee30146ac7dcd5","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-11-29 19:46:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e38425720f8e72e74b1104cc165b98c4b77d7f27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 19:48:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"825b7e08d82d788b4c136e0d19bbd01ea1e8ad27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 21:29:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14646374d68647859c867025fdfb08af\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bf95aa7d5cf43ad991271a5473b5f67 : SUCCESS in 11m 49s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e8fd36396a064b2abccc289cb20aac6a : SUCCESS in 1h 36m 16s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/bab9ecd7d55e4776a945d042c47cbd40 : SUCCESS in 1h 31m 42s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/17ba4694b7bd4648be87d38c435163cf : SUCCESS in 1h 35m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e42a5d5f1604367b226086b074e1fd2 : SUCCESS in 1h 35m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00d185b340884d5ea48e4abdea1b5bab : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":3},{"id":"26f130ce03d1bfb5c4cb00a475bb528cc482513d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 21:29:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4f0d0268be44249014c46eff68394f4a6ee624f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 21:30:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7babf7beda2b4b45b6c281814dd0a0e9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30ef8dbb30194a159c25969e009533bc : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c31bbc215c2f69f9738350b9e9a592b728d8e12b","revisions":{"a2193d6e34357b19915f80e86e3cac67d645571b":{"kind":"REWORK","_number":1,"created":"2022-08-29 18:14:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/855074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/74/855074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/74/855074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/74/855074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/74/855074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/74/855074/1"}}},"commit":{"parents":[{"commit":"87e78ee34c7f0126ba0d6a1eb8252d9141726681","subject":"Support service tokens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/87e78ee34c7f0126ba0d6a1eb8252d9141726681"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-08-29 18:14:42.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-08-29 18:14:42.000000000","tz":120},"subject":"Change networks_info output structure","message":"Change networks_info output structure\n\nWith commit [1] to collection output structure of networks_info module\nhas been changed. With that we adpot to the new format.\n\n[1] https://opendev.org/openstack/ansible-collections-openstack/commit/9272146cf7a7486f6004b52bec420a998e710f45\n\nChange-Id: Ic22ec379983e43aa5f2b55fd4543b4aa70762354\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a2193d6e34357b19915f80e86e3cac67d645571b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a2193d6e34357b19915f80e86e3cac67d645571b"}]},"branch":"refs/heads/master"},"7930f234e88fe7ab709455b76d39d53033236f30":{"kind":"REWORK","_number":2,"created":"2022-11-24 17:18:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/855074/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/74/855074/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/74/855074/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/74/855074/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/74/855074/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/74/855074/2"}}},"commit":{"parents":[{"commit":"87e78ee34c7f0126ba0d6a1eb8252d9141726681","subject":"Support service tokens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/87e78ee34c7f0126ba0d6a1eb8252d9141726681"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-08-29 18:14:42.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-11-24 17:17:20.000000000","tz":60},"subject":"Adopt output structure to new collections version","message":"Adopt output structure to new collections version\n\nWith commit [1] to collection output structure of networks_info module\nhas been changed. With that we adpot to the new format.\n\nReturn values for keypair have also changed\n\n[1] https://opendev.org/openstack/ansible-collections-openstack/commit/9272146cf7a7486f6004b52bec420a998e710f45\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/864553\nChange-Id: Ic22ec379983e43aa5f2b55fd4543b4aa70762354\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7930f234e88fe7ab709455b76d39d53033236f30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7930f234e88fe7ab709455b76d39d53033236f30"}]},"branch":"refs/heads/master"},"c31bbc215c2f69f9738350b9e9a592b728d8e12b":{"kind":"REWORK","_number":3,"created":"2022-11-25 11:57:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/855074/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/74/855074/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/74/855074/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/74/855074/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/74/855074/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/74/855074/3"}}},"commit":{"parents":[{"commit":"87e78ee34c7f0126ba0d6a1eb8252d9141726681","subject":"Support service tokens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/87e78ee34c7f0126ba0d6a1eb8252d9141726681"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-08-29 18:14:42.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-11-25 11:57:49.000000000","tz":0},"subject":"Adopt output structure to new collections version","message":"Adopt output structure to new collections version\n\nWith commit [1] to collection output structure of networks_info module\nhas been changed. With that we adpot to the new format.\n\nReturn values for keypair have also changed\n\n[1] https://opendev.org/openstack/ansible-collections-openstack/commit/9272146cf7a7486f6004b52bec420a998e710f45\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/864553\nChange-Id: Ic22ec379983e43aa5f2b55fd4543b4aa70762354\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/c31bbc215c2f69f9738350b9e9a592b728d8e12b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/c31bbc215c2f69f9738350b9e9a592b728d8e12b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
