)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~864419","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~Ie7a6dd78d5af82a546e4d6a19ec5b256df8f64e6","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"osa/healthcheck","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-12-01 15:06:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie7a6dd78d5af82a546e4d6a19ec5b256df8f64e6","subject":"Enable /healthcheck for Octavia API","status":"MERGED","created":"2022-11-14 15:23:58.000000000","updated":"2022-12-01 15:07:13.000000000","submitted":"2022-12-01 15:06:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864419-osa/healthcheck","meta_rev_id":"09be201188a4122f3cd89e42511bb34176f91ff8","_number":864419,"virtual_id_number":864419,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-01 15:06:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2022-12-01 13:22:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-28 20:29:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2022-12-01 13:22:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-14 17:31:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-28 20:29:29.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-12-01 13:22:21.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"8c8e4d43c0a6fe0c0ec4433bc6bda1300e81b649","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-14 15:23:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3327b54db6a3482d2b1e979c3414c4b506725196","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 17:31:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/29c79fa06c654da28732ccd36721a4a7\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed7a18b3be1f4135aa19754aa194093b : SUCCESS in 15m 12s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/bd6387e87df743439c323feabcbbe0ee : FAILURE in 1h 24m 17s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffbc4aa69942455bab4fca8661e04182 : FAILURE in 1h 22m 14s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5df737d69b514317b2181f22be931cf6 : FAILURE in 1h 22m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f59fd2740fe451d83d92acaa34ec992 : FAILURE in 1h 31m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea569ff2065e40c5a2554bfc0ec58268 : FAILURE in 1h 30m 16s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/abf7a3304f214e1b953393e3e986ff8a : FAILURE in 1h 17m 13s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02e337c416424a1a8da298fbdbe0afa4 : FAILURE in 2h 03m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbc41a765f754e849c71222c3a699d37 : FAILURE in 43m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/211b26e9212b44bb8524c30acaf4c81f : FAILURE in 37m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/510f1e01e0e449e3a7a443b531ee71f0 : FAILURE in 42m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1e92de1b877470792eb3ffa0c5abb5c : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"ef108a6fb47571081d0052bed90078fc37a009a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-15 14:53:15.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"63bd4ac6b4e9a151b0a6e1126998f03e26df87f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 16:56:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee023776eeb34d7997662540d0fb472d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a035c5ac28914e66b1530efec6da89b4 : SUCCESS in 12m 45s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d34483fe20f14b5289320c88829496ed : FAILURE in 1h 23m 33s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fbfbc2812cf4ccf94e2e7db19853949 : FAILURE in 1h 34m 04s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4b4a0065b2544697836f91645690dcc6 : FAILURE in 1h 24m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c1eb20f338a48d6a68945fd717857b1 : FAILURE in 1h 33m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a52470543fd467b97b3521baa49f995 : FAILURE in 1h 30m 31s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5dff164e56c4d5297ef075290762f4b : FAILURE in 1h 56m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5404abdd61944359597d04b80d94280 : FAILURE in 1h 55m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/583651585bac492cb8a4ff6ae2fb4845 : FAILURE in 39m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5f5da2dc0b324af5ac9fc8cda6ff74c5 : FAILURE in 38m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18338f6eed6148f8a823a0f4c7909a1b : FAILURE in 40m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2944bf6b23a44430a912b360b39f8342 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"e7d2d7774e6a011dcb536b461c93d9094e8fa49c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-24 17:24:29.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"15af935138f36da362b98b762cd637f604898653","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 19:44:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/939e07612d094f8482828e1edd911c5d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ace745fccf104246891080d72f9e1529 : SUCCESS in 14m 14s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/bf0f2781789141298dc30b694fc0b5ad : FAILURE in 1h 27m 16s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3cf2c0d586834fb7bc690045aa46de36 : FAILURE in 1h 24m 42s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1a3a6cfad5074f92898a8c632defdd9a : FAILURE in 1h 16m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab022b1ae91847009a28d57e87f2c8aa : FAILURE in 1h 30m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14e11e1cc84d4e99a00c795eb3a9cf73 : FAILURE in 1h 28m 48s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd6d4b3e55384e32aa844d0b65b2aced : FAILURE in 10m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1d76d9927f148a58b8fcb8128ecbc84 : FAILURE in 2h 12m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0622082315594c179c4b776ddc244eae : FAILURE in 42m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7f027c262f3f4aa7b27f56c241dc9379 : FAILURE in 36m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20e48d2cdee340588dd5ab4cd043e3c8 : FAILURE in 41m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/645f3a6e793649c891af79d98f3c204c : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":3},{"id":"8c3e67b8ab7f65373708c555381b310f8c81fac8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-25 16:32:43.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"adf523f44bb739921d98de65ae8178f8341527ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 17:28:09.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/855074, https://review.opendev.org/864553 are needed.","accounts_in_message":[],"_revision_number":4},{"id":"e6fee3a3490e5cc3f3394184761e873afcbbed13","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-28 11:50:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"481a0abc89de7332b52b2a382ef4bd6203d3a273","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 14:34:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b09fa7f3a9d54b688be143ffa7e1734d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b5cfce5051d4e61874043a029a914a6 : SUCCESS in 13m 04s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/71a84015b60c44309d8d1eae26168b96 : SUCCESS in 1h 32m 34s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef3c3ae316ba4dd8a0b4e0db4683f8dd : SUCCESS in 1h 41m 58s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/975bd97923c74fad8b54c5977701e0ec : SUCCESS in 1h 38m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dca5939818934cb7930d7def563e9b43 : SUCCESS in 1h 49m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44eed270f4734f8b9c2e91f629b36f8b : SUCCESS in 2h 02m 06s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/07a6ccfde0f2479e967974e44a67fa85 : SUCCESS in 2h 07m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/699c13abe74b4f95b8bbbf6e0650e684 : SUCCESS in 2h 39m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/465a4093063d4ff589b4c07548d390c0 : FAILURE in 50m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/94803f63a2074177a4e85e99161fccc3 : FAILURE in 40m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fde8da6dee0497aa66bb3bdadebf323 : FAILURE in 39m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6a7fa23b9264bf88c5b4114f5a11737 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":4},{"id":"3d522992e9038820141873ff90780c07347e64d6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-11-28 20:29:29.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7d8183c4b0d03f9a8dde3732b6cacb20c92961b4","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-12-01 13:22:21.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"aa2f9975c667bc703f8e7c8d8f817e42f1a86138","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 13:24:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"85c67d787e7c3411805c701ff23d3d5d49c503d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 15:06:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33a1f9c3f6f24c4a8f0ec1d9b61f493c\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b42e810f291d48fc8eb2f4fbca571169 : SUCCESS in 13m 40s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d4246747b5a14332a848e5d0ac6611dd : SUCCESS in 1h 35m 41s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e44419c87fe9478db092a3a69e12eeae : SUCCESS in 1h 31m 18s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/897bd9cb032e42df8f93bf542438b4a0 : SUCCESS in 1h 33m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02ae37ea354d4bebb4c9ab4989fcfe12 : SUCCESS in 1h 38m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/628ed5a853b8405da08d40c58cf8e75c : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":4},{"id":"8201a60f03aa0456ff40d6d1add941bbac9d83cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 15:06:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"09be201188a4122f3cd89e42511bb34176f91ff8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 15:07:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ceab0722e5ce4409b64f6b6cf9eaa5e5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8faa745beedf4f6fa367eaeca6471c18 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b08e95a972e24beb135b80a3f9bd711a3b49b8ca","revisions":{"61cda3ffa164e14c881bfac131af4f9d4312fc2e":{"kind":"REWORK","_number":1,"created":"2022-11-14 15:23:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/864419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/864419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864419/1"}}},"commit":{"parents":[{"commit":"87e78ee34c7f0126ba0d6a1eb8252d9141726681","subject":"Support service tokens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/87e78ee34c7f0126ba0d6a1eb8252d9141726681"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-11-14 15:23:50.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-11-14 15:23:50.000000000","tz":60},"subject":"Enable /healthcheck for Octavia API","message":"Enable /healthcheck for Octavia API\n\nChange-Id: Ie7a6dd78d5af82a546e4d6a19ec5b256df8f64e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/61cda3ffa164e14c881bfac131af4f9d4312fc2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/61cda3ffa164e14c881bfac131af4f9d4312fc2e"}]},"branch":"refs/heads/master"},"d9ce0a02176c73fab6ad631e6400fd5504b6c541":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-15 14:53:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/864419/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/864419/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864419/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864419/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864419/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864419/2"}}},"commit":{"parents":[{"commit":"87e78ee34c7f0126ba0d6a1eb8252d9141726681","subject":"Support service tokens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/87e78ee34c7f0126ba0d6a1eb8252d9141726681"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-11-14 15:23:50.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-11-15 14:53:15.000000000","tz":0},"subject":"Enable /healthcheck for Octavia API","message":"Enable /healthcheck for Octavia API\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/864553\nChange-Id: Ie7a6dd78d5af82a546e4d6a19ec5b256df8f64e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/d9ce0a02176c73fab6ad631e6400fd5504b6c541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/d9ce0a02176c73fab6ad631e6400fd5504b6c541"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8c1445b68c5055e4e0b098e9ed3811777dc6c6b0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-11-24 17:24:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/864419/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/864419/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864419/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864419/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864419/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864419/3"}}},"commit":{"parents":[{"commit":"7930f234e88fe7ab709455b76d39d53033236f30","subject":"Adopt output structure to new collections version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7930f234e88fe7ab709455b76d39d53033236f30"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-11-14 15:23:50.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-11-24 17:24:29.000000000","tz":0},"subject":"Enable /healthcheck for Octavia API","message":"Enable /healthcheck for Octavia API\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/864553\nChange-Id: Ie7a6dd78d5af82a546e4d6a19ec5b256df8f64e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8c1445b68c5055e4e0b098e9ed3811777dc6c6b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8c1445b68c5055e4e0b098e9ed3811777dc6c6b0"}]},"branch":"refs/heads/master","description":"Rebase"},"b08e95a972e24beb135b80a3f9bd711a3b49b8ca":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-11-25 16:32:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/864419/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/864419/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864419/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864419/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864419/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864419/4"}}},"commit":{"parents":[{"commit":"c31bbc215c2f69f9738350b9e9a592b728d8e12b","subject":"Adopt output structure to new collections version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/c31bbc215c2f69f9738350b9e9a592b728d8e12b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-11-14 15:23:50.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-11-25 16:32:43.000000000","tz":0},"subject":"Enable /healthcheck for Octavia API","message":"Enable /healthcheck for Octavia API\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/864553\nChange-Id: Ie7a6dd78d5af82a546e4d6a19ec5b256df8f64e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b08e95a972e24beb135b80a3f9bd711a3b49b8ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b08e95a972e24beb135b80a3f9bd711a3b49b8ca"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
