)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~864819","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I9fa33c5ee76197191f1e66b7a70a4c1c0a5fa394","project":"openstack/openstack-ansible-os_octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"14552":{"account":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"last_update":"2022-11-29 21:29:57.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-11-29 21:29:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9fa33c5ee76197191f1e66b7a70a4c1c0a5fa394","subject":"Adding octavia_provider_network_mtu-parameter parameter","status":"MERGED","created":"2022-11-21 17:32:19.000000000","updated":"2022-11-29 21:31:59.000000000","submitted":"2022-11-29 21:29:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864819","meta_rev_id":"92f4bfc609ad2210cf30296106d603cbc5fc8d23","_number":864819,"virtual_id_number":864819,"owner":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-29 21:29:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-11-28 11:50:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-29 13:52:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-29 13:52:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-21 17:42:48.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-11-21 18:54:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-29 13:52:12.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"f5327f06ebebb80623f456096a86b988a9d0d308","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2022-11-21 17:32:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19d76b3929043fed412ef8cfb0d00e5c1bc8e5f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2022-11-21 17:34:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cacd42327e288cd49d77380eef2b6f056ad34a08","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-21 17:42:48.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6f81a549d6ab73ca2b548314a799325c24ec742e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 18:54:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d3323f04a7440cb9a58e8f429701b87\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ecfdc6ae1a54509beaa746a8e121b6b : RETRY_LIMIT in 8m 15s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3ae3445050a4465386f12fb7ebbcf32e : RETRY_LIMIT in 9m 37s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/eea257f26f7e4def9cabb1252b29da29 : RETRY_LIMIT in 12m 24s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/03d5a783c0274bf29b852bfeb5126c17 : RETRY_LIMIT in 9m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af42173724cc4370a712a8b92e6ffae9 : RETRY_LIMIT in 9m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03b3a44fa05e4fcf9f917a595d5a4e04 : RETRY_LIMIT in 8m 59s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3392354961e742598253cdd4c8e071b9 : FAILURE in 11m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4123e7b84a1e4585950e9ccace4091e3 : FAILURE in 1h 15m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/296f62bf9db9447d829d16897e29f2cd : RETRY_LIMIT in 7m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/041a38d389864213903f1419a35df0c4 : RETRY_LIMIT in 8m 25s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57a35368d4294f989ef223d0b2622ba3 : RETRY_LIMIT in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d5b8d629fc14d258e465dc10106c4f9 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cafd7718c7fb441680f36b326f886840 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":2},{"id":"ccd218459c99e706b6dd2dfe84f99fbbd81848af","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2022-11-22 04:50:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b0c9a3945e25e5008cdef33b232bf1c5e336881","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 05:56:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e4fce69b0684934ae101d446ed8eb33\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50198042422648e798d124a8a42d0b29 : RETRY_LIMIT in 7m 05s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/90742c913e85497b93b1dbddfef01474 : RETRY_LIMIT in 7m 22s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8fa1c9126c3f4ae987a9b8ac50b4a79f : RETRY_LIMIT in 13m 29s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d8c44904934e4c73ab2d332c863bee1e : RETRY_LIMIT in 9m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56e2142cb1ef464d8702697e0bb82f4c : RETRY_LIMIT in 9m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03ddd16a11f6464c967016e745c06035 : RETRY_LIMIT in 8m 27s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/365c4a0fcdab496d95d280b7eb6a0834 : FAILURE in 9m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9967ed68a24b463e935bd08dfad64a25 : FAILURE in 1h 02m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/964864c989bb46cc9472f941a7a1884b : RETRY_LIMIT in 5m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2616f0f0f8134c74b2a336f2a5c4d2ee : RETRY_LIMIT in 5m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35dca03cae5743c0b7838b772fcf6bfb : RETRY_LIMIT in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9cbb5e6621f4bceb20e3284c0c7d418 : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9c668fc294c4dbb84a43ce77befc389 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":2},{"id":"893e4700e7133ccd940d4a2a3e44ce91ccabe516","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-24 17:23:53.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"dc063be2914b51691b2671c27d8c92b008d3fa97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 19:14:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/71cf2eb7a79e4ba2a2bc41486ae7de85\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bcbedd2064544aa855f6698265c870c : SUCCESS in 14m 12s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6f1dac64254f4aef9591edae6bd44e97 : FAILURE in 1h 22m 03s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf85010de2c64172acf9101dfacd3e08 : FAILURE in 1h 11m 46s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ebe3b9a173ef4dd498d56439bf1a07d1 : FAILURE in 1h 16m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1d64df40f184e11855d40689b702dff : FAILURE in 1h 26m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93fce90e0b424f40af3a6a2fcd0478c7 : FAILURE in 1h 21m 57s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/fcd9c57b0fe74be188d8571f79153caf : FAILURE in 12m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa08f5db03924f32910755d39219045d : FAILURE in 1h 44m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a3d8f427e614cafb361590cb4b4801b : FAILURE in 46m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/694b9358a01e4b4b8e06f2b4c49232fb : FAILURE in 40m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6e2f142c4f84c008c1e82cb99a085e0 : FAILURE in 40m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1485cd0f3b94b6493753c4dcac648f1 : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94f5efcecae447aca0aee803f0a8f14b : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":3},{"id":"006a0a42bc7c975af1e5753e3455b03c7551e5fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-25 16:32:24.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"9874820f54d82a7df3b91a37e51c15254aa8769b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 17:28:06.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/855074 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"0d83463c339443e9e153beb43ca28d6f668794cf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-28 11:50:44.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"829a9e776611aded1d56ee7f2291dbe95a5eb5f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 14:08:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7d9e6754e1f4de2b9b794909589f4fa\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ecab7e597f84bb9889fea6db297942b : SUCCESS in 19m 27s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1297ea12878541fb94e156fd34af40be : SUCCESS in 1h 38m 50s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/65fd726985624372ad3c135a53f958e5 : SUCCESS in 1h 39m 59s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/06be6e9b02b4495f9a654f930a873359 : SUCCESS in 1h 38m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5436be56d764d7fbd996b57ef472e18 : SUCCESS in 1h 52m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d3d8508e1914b69975713d123c0d8b3 : SUCCESS in 1h 39m 26s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d55a60f4ba348469727c78efafe1e61 : SUCCESS in 1h 45m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cc95fa4f52040ed8a6ccdae13dd6e3c : SUCCESS in 2h 14m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b159931c2d184ebe9e1ac6ae7727ba3b : FAILURE in 40m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cc6be561faaa4a789a5384bc7823f831 : FAILURE in 31m 56s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24d74be2f9424e9089621791b82bc040 : FAILURE in 41m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b8bc379b1b24c728bc802bc289f7ebc : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/194ba11fe168454d89beda05d3efb3f4 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":4},{"id":"5eb31b617bf6c75fc6ad59de2aac7bfafd26c35a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-11-29 13:52:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ce1f29908ce594c26b26f35b44479320817b0f41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 19:48:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9d06eb0be67d8727efcf9aa9ce5a202b1936d29c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 21:29:57.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bb5903398d84e5f92c8d7f8901fc9a3\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e6d3a1772574161a04c34dbd6100cb0 : SUCCESS in 14m 47s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/36900b3172f3484f9807f25b9231d1e6 : SUCCESS in 1h 17m 54s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b9293872ac04dd6b8edb229dea871ae : SUCCESS in 1h 26m 52s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ae5f9d5c16e24acaaaf6291cf36b21e0 : SUCCESS in 1h 35m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5bcbca33d644fcebed3c7ef95d63aaa : SUCCESS in 1h 34m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a270a9a6f1654d96bc28abf48df66220 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c271ac9ffb14366a16f686976bf0041 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":4},{"id":"dae3c627c5e4efdb8b41986dc19afbe14871081a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 21:29:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"92f4bfc609ad2210cf30296106d603cbc5fc8d23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 21:31:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8aa32818cc94975a2d55ea0b7b98892\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4cf6c96e11640bca9551d17004715d2 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2f1a3b4db1664cd9492bd32d1bad1793d5e61368","revisions":{"968180fa638b8a534c489147806a2850571ae5b8":{"kind":"REWORK","_number":1,"created":"2022-11-21 17:32:19.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/19/864819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/864819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864819/1"}}},"commit":{"parents":[{"commit":"87e78ee34c7f0126ba0d6a1eb8252d9141726681","subject":"Support service tokens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/87e78ee34c7f0126ba0d6a1eb8252d9141726681"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2022-11-21 17:27:21.000000000","tz":-360},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2022-11-21 17:27:21.000000000","tz":-360},"subject":"Adding octavia_provider_network_mtu-parameter parameter","message":"Adding octavia_provider_network_mtu-parameter parameter\n\nThe `octavia_provider_network_mtu-parameter` defaults to 1500\nto not accidentially use `global_physnet_mtu` on deployment with\nlarge MTU settings\n\nChange-Id: I9fa33c5ee76197191f1e66b7a70a4c1c0a5fa394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/968180fa638b8a534c489147806a2850571ae5b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/968180fa638b8a534c489147806a2850571ae5b8"}]},"branch":"refs/heads/master"},"3942f4b2a9a9204079f4d9f9f3cae477b118a282":{"kind":"REWORK","_number":2,"created":"2022-11-21 17:34:19.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/19/864819/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/864819/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864819/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864819/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864819/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864819/2"}}},"commit":{"parents":[{"commit":"87e78ee34c7f0126ba0d6a1eb8252d9141726681","subject":"Support service tokens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/87e78ee34c7f0126ba0d6a1eb8252d9141726681"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2022-11-21 17:27:21.000000000","tz":-360},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2022-11-21 17:34:14.000000000","tz":-360},"subject":"Adding octavia_provider_network_mtu-parameter parameter","message":"Adding octavia_provider_network_mtu-parameter parameter\n\nThe `octavia_provider_network_mtu-parameter` defaults to 1500\nto not accidentially use `global_physnet_mtu` on deployment with\nlarge MTU settings\n\nChange-Id: I9fa33c5ee76197191f1e66b7a70a4c1c0a5fa394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3942f4b2a9a9204079f4d9f9f3cae477b118a282"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3942f4b2a9a9204079f4d9f9f3cae477b118a282"}]},"branch":"refs/heads/master"},"e563ca0c68a9da4c9be6834b8f2a3c80ca09fcc3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-11-24 17:23:53.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/864819/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/864819/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864819/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864819/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864819/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864819/3"}}},"commit":{"parents":[{"commit":"7930f234e88fe7ab709455b76d39d53033236f30","subject":"Adopt output structure to new collections version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7930f234e88fe7ab709455b76d39d53033236f30"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2022-11-21 17:27:21.000000000","tz":-360},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-11-24 17:23:53.000000000","tz":0},"subject":"Adding octavia_provider_network_mtu-parameter parameter","message":"Adding octavia_provider_network_mtu-parameter parameter\n\nThe `octavia_provider_network_mtu-parameter` defaults to 1500\nto not accidentially use `global_physnet_mtu` on deployment with\nlarge MTU settings\n\nChange-Id: I9fa33c5ee76197191f1e66b7a70a4c1c0a5fa394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e563ca0c68a9da4c9be6834b8f2a3c80ca09fcc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e563ca0c68a9da4c9be6834b8f2a3c80ca09fcc3"}]},"branch":"refs/heads/master","description":"Rebase"},"2f1a3b4db1664cd9492bd32d1bad1793d5e61368":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-11-25 16:32:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/864819/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/864819/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864819/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864819/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864819/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/864819/4"}}},"commit":{"parents":[{"commit":"c31bbc215c2f69f9738350b9e9a592b728d8e12b","subject":"Adopt output structure to new collections version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/c31bbc215c2f69f9738350b9e9a592b728d8e12b"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2022-11-21 17:27:21.000000000","tz":-360},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-11-25 16:32:24.000000000","tz":0},"subject":"Adding octavia_provider_network_mtu-parameter parameter","message":"Adding octavia_provider_network_mtu-parameter parameter\n\nThe `octavia_provider_network_mtu-parameter` defaults to 1500\nto not accidentially use `global_physnet_mtu` on deployment with\nlarge MTU settings\n\nChange-Id: I9fa33c5ee76197191f1e66b7a70a4c1c0a5fa394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2f1a3b4db1664cd9492bd32d1bad1793d5e61368"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2f1a3b4db1664cd9492bd32d1bad1793d5e61368"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
