)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~869038","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~stable%2Fwallaby~I5a05f583631ef7b6429da4a6a3a4e895e9c75163","project":"openstack/openstack-ansible-os_octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-01-09 12:06:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5a05f583631ef7b6429da4a6a3a4e895e9c75163","subject":"Use focal amphora test image by default","status":"MERGED","created":"2023-01-03 08:01:13.000000000","updated":"2023-01-09 12:07:32.000000000","submitted":"2023-01-09 12:06:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869038","cherry_pick_of_change":822834,"cherry_pick_of_patch_set":1,"meta_rev_id":"c78acd638f92d2fe6b1dbfb931e3768fa256d1b5","_number":869038,"virtual_id_number":869038,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-09 12:06:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-01-03 12:13:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-09 10:25:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-09 10:25:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-03 10:04:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-09 10:25:23.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"c6f34646feeeae2426670c0ed7b65ea0087a4999","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-01-03 08:01:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"00cbdca8698eb1ddeda6d9a2825dc0baa07f34aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 10:04:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad80fc4888034e449cc8768ad27644b8\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8136b2202cef461faca5c5458e145726 : SUCCESS in 10m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f2f5d11a3416462d9edfdfc93f698062 : SUCCESS in 1h 19m 58s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5d5613143e504304bc320703a48acdfe : SUCCESS in 1h 11m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7840988f8fbd421d8b311b207d2e388d : SUCCESS in 1h 29m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/219b3adfa8834b9aa756e1aad83a1d49 : SUCCESS in 1h 25m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/511a8fc98ec841db80f16d93a38397ba : SUCCESS in 1h 31m 20s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8a81b22a88b44809c8ad77bc7dad42f : SUCCESS in 1h 56m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1613e41c3a1b4a5fb90b84a3f570d85c : FAILURE in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"950550a53543b2d6241baed6d80d4c2532d7343a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-01-03 12:13:27.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4f90f6559aad6c42e99fb050b5c3c314d4bf8220","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-01-03 12:13:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2d9c7fe2695ce7de92b0062cf4e46898ad4d0107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 14:01:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b51737d5c684452babe568aab7e39bec\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b070e12f35364fdb949da341f8f0e0d6 : SUCCESS in 9m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0974ba64f28041e7aba7810395489e95 : SUCCESS in 1h 12m 25s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d253add1888943bb9cd1902bf2f586e4 : SUCCESS in 1h 04m 08s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7cda215c33e447a0af67dbaf64e19a84 : SUCCESS in 1h 26m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a66e738dcf4641939fae35ff38b65d39 : SUCCESS in 1h 24m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28d64057bcd64a6b90a5467b9f63af20 : SUCCESS in 1h 05m 51s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbc72df7a1074fa4a465d89f33f7bfd5 : SUCCESS in 1h 41m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6764577f0e274e47bf4817d5ef97f080 : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c95b58839ca4ec382c112274f42a9c9 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"2c3d86c26acd218b897308679fb533dbe2ac38ed","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-01-09 10:25:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d5eabe566a7e62698e639cf454125444d56db056","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 10:27:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"24cad262a53ecffd034244de85336deeafc82755","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 12:06:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6102a95427a846afa62101fc7379949b\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49934777181342a5b60e51749a1a2138 : SUCCESS in 11m 11s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7c809f99b8244973b6ce0d723ce4b656 : SUCCESS in 1h 20m 25s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/be77b86cea4745dd969d1e071bbb9a53 : SUCCESS in 1h 32m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0a446c9a05749328f7bc3988463d87b : SUCCESS in 1h 29m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ce72e31fa8542aa84f1520bc5646f2d : SUCCESS in 1h 13m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68f042b7cfd1474cb455d44841c6752d : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/566e148adefb45e6bbe4387c419706b2 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"d5159171312c1380e0b3c2a997217c59901574e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 12:06:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c78acd638f92d2fe6b1dbfb931e3768fa256d1b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 12:07:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c3279e1df2e4f3dbce3e3e089694ec4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16861053eee34587b44228cd89c0a129 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e8e9d1d11da960a91a116eef0c48b099e86ff9ef","revisions":{"a4003124170226ce3046bc716590bc485d1242b9":{"kind":"REWORK","_number":1,"created":"2023-01-03 08:01:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/38/869038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/38/869038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/869038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/869038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/869038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/869038/1"}}},"commit":{"parents":[{"commit":"c90138f2e975b27fcc8bf91ba5b7b70266791d9f","subject":"Adding octavia_provider_network_mtu-parameter parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/c90138f2e975b27fcc8bf91ba5b7b70266791d9f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-23 14:19:59.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-01-03 08:01:13.000000000","tz":0},"subject":"Use focal amphora test image by default","message":"Use focal amphora test image by default\n\nSwitch used default amphora image from bionic to focal\n\nChange-Id: I5a05f583631ef7b6429da4a6a3a4e895e9c75163\n(cherry picked from commit c7a7a14f7bee4d46336ab8a3cc82d29628f32c01)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a4003124170226ce3046bc716590bc485d1242b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a4003124170226ce3046bc716590bc485d1242b9"}]},"branch":"refs/heads/stable/wallaby"},"e8e9d1d11da960a91a116eef0c48b099e86ff9ef":{"kind":"REWORK","_number":2,"created":"2023-01-03 12:13:27.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/38/869038/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/38/869038/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/869038/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/869038/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/869038/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/869038/2"}}},"commit":{"parents":[{"commit":"c90138f2e975b27fcc8bf91ba5b7b70266791d9f","subject":"Adding octavia_provider_network_mtu-parameter parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/c90138f2e975b27fcc8bf91ba5b7b70266791d9f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-23 14:19:59.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-01-03 12:13:27.000000000","tz":0},"subject":"Use focal amphora test image by default","message":"Use focal amphora test image by default\n\nSwitch used default amphora image from bionic to focal\n\nChange-Id: I5a05f583631ef7b6429da4a6a3a4e895e9c75163\n(cherry picked from commit c7a7a14f7bee4d46336ab8a3cc82d29628f32c01)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e8e9d1d11da960a91a116eef0c48b099e86ff9ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e8e9d1d11da960a91a116eef0c48b099e86ff9ef"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
