)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~881999","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~stable%2Fyoga~I4a3346c90825a4bf0b416943286696fa529f526d","project":"openstack/openstack-ansible-os_octavia","branch":"stable/yoga","topic":"cherrypick-osa/systemd_restart_on_unit_change-nq9phb886h","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-02-21 09:21:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4a3346c90825a4bf0b416943286696fa529f526d","subject":"Ensure service is restarted on unit file changes","status":"ABANDONED","created":"2023-05-02 11:01:59.000000000","updated":"2024-02-21 09:21:22.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":880339,"cherry_pick_of_patch_set":2,"meta_rev_id":"0e76bdd1fc8f0da3f07f979db4b519dc7b46be33","_number":881999,"virtual_id_number":881999,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-11 12:33:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-10-02 14:22:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2023-09-28 17:30:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2023-10-02 14:22:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-02 14:05:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-02 14:22:43.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-02-21 09:21:21.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"9cc65ee60a149103d7f6335828016aa1547051f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-02 11:01:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7439abe6f1eb230eb1ffbcc811acc7f4dad586fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 14:05:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcd4b79731b14457a0c67b6af7a5ea89\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ed38f6f21704540bf10c67f83116fc6 : SUCCESS in 11m 35s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e7fa03f76a744de8b65937683c1073ce : SUCCESS in 1h 27m 29s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/67ac931c072c490280cc274bcfc1ef8e : SUCCESS in 1h 36m 04s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c52da7d9f64d4f6fa49fbd175d49ff6c : SUCCESS in 1h 31m 22s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2771122455eb41caa3b72cfa7db8ec19 : SUCCESS in 1h 32m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e648b101e9d14d30b6c84240a5d38ce3 : SUCCESS in 1h 32m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17adb8a3bbc3467aa7eaae363075c2ca : SUCCESS in 1h 29m 28s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/fa72dc5916294fb2a5400e1480d3606e : FAILURE in 1h 57m 22s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb4b56bb595e454aad2a30bae6b66935 : SUCCESS in 2h 44m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64be69c009f74a39922112a6f9d5e854 : SUCCESS in 2h 16m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a4b16ef4146947d59366e7c0832a062a : FAILURE in 1h 10m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/27126d414b9d4791bcc7935562e02f46 : FAILURE in 40m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c82e66afd8b34d0f88d6de56a1702510 : SUCCESS in 1h 03m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50516fbdfa0346b5823009ce3e53f965 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a3035f96109c4e0a98a4a31fa455559417ac389f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-02 17:44:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d80008fee277a8acc10eafe5f3dc17f01abdb84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 20:06:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/be73eb6c30eb4402bc6533250464372a\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5777cb4167a24fb3bc25a5e5eb4065a1 : SUCCESS in 11m 09s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e6cd6dd4c72f4418ac2e56565d943b19 : SUCCESS in 1h 25m 04s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e436a4e6c8f4b62ba4b7622826d4a0c : SUCCESS in 1h 37m 52s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ce6d464968b475fbcd13ba6de5c4af1 : SUCCESS in 1h 35m 22s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/42853161da29484f8bd55e745398bc2d : SUCCESS in 1h 26m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b02a67a164f4688b646abb77c9029ba : SUCCESS in 1h 38m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc077aa6dc9f4b3eba33da7774983c94 : SUCCESS in 1h 23m 25s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/65dfe70bbca94218a7b9de56df35f1ff : SUCCESS in 1h 25m 23s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d09ee98bfee34f3ab9e7afa58dd9ed1a : SUCCESS in 2h 12m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ca0bcf253e840909aeddf63904675a4 : SUCCESS in 2h 09m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/19cd5a20efb74a42a4f17797c7d2c95f : FAILURE in 1h 09m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bc86d7f89a6a450ab62e450d8ec03c64 : FAILURE in 37m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d112cd1b76d477f92b658a0d1752b9c : FAILURE in 42m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acee4943aa0445e8b2f30777960eee00 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f6f4f5ca2936d3fe1bfad3916db293fffb21e96d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-28 17:30:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2543acf636f871dbb64c1699224556b33183f7a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 19:39:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a58ff1316dd4fcfa2ae386ae41c0010\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2600282daf24b74ae238b48d310bb0e : SUCCESS in 13m 01s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/820ce825c300498586d176c1402c9c77 : SUCCESS in 1h 20m 35s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9b56338ea4f248e39d166a6d9abeee60 : SUCCESS in 1h 44m 46s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ad9439f524349cea945e94c344434bb : FAILURE in 55m 30s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fe1f7414674040b389da2eb235ca5736 : SUCCESS in 1h 27m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ba44d1fa55c4b9ea667eb2c5caee1ee : SUCCESS in 1h 31m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cbdd613573f4b44ac3d727959f94558 : SUCCESS in 1h 31m 38s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/7b6288933cc143b098f434962a6c4193 : SUCCESS in 1h 35m 49s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a50abb86c40440ffa61cca4988234b87 : FAILURE in 1h 07m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/489d20fab97a4073b367be84917590ea : SUCCESS in 2h 03m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5a5ea341417749da8f0ac2cd93807910 : FAILURE in 1h 09m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/66bb8a65c9024db3b5371f46a6130c70 : FAILURE in 39m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1dc2cd09e6ad4dbd9e148b1264a10e48 : SUCCESS in 1h 04m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47046cca386f498f97f72894e7d4db8e : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"1c2aa48f6571bf2822dcf3d06e195d5cc4daa8e0","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-10-02 14:22:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1fa84468329b9b6d30a83c2c2a2a412421c59c0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 14:24:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6496adc47c030b223583b420e2ef99e6268572a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 15:38:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/170967573b4e4012b92b50685a144bfd\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3696905014d4b1f90b27b7c113f3949 : RETRY_LIMIT in 16m 20s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ea1ee4482ed3488abbfe7d2478aab14a : RETRY_LIMIT in 9m 51s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0f89c068453435e82a8884f6b3669d2 : RETRY_LIMIT in 19m 11s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2ff5e860f39942c191e8dff2babbb226 : RETRY_LIMIT in 16m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12171aacbc4a44f98379b82a838c1dc0 : RETRY_LIMIT in 10m 56s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/24b32200929647239d8e55dbc3bc8b39 : RETRY_LIMIT in 10m 30s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fda582f409c9485ab54fb5e424c1d734 : RETRY_LIMIT in 7m 26s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8214deb5b77b4c13a5f12d5e853d64c3 : RETRY_LIMIT in 10m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe22a2b7d6c0404f9af10bfd96bee21b : SUCCESS in 4m 08s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/95cbc561853545ec8939a5537b087d44 : RETRY_LIMIT in 13m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6753db2492c39fc0fb33c38556fff324ec2bba0b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-06 16:07:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3998b82335e5a0de8808353d6fe3839c072aa82f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 18:10:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/01f2cd00838a4c47bf1945a41ff3ee5b\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acb3166f1b7f4fa79db0e3e6febc0495 : SUCCESS in 13m 11s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d4362ce7db874e75809928bd878efacf : SUCCESS in 1h 08m 04s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7fc56a30d2dc4fdb9c40731d577702b6 : FAILURE in 1h 53m 13s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff1f3774fe424d9da964c4ff4e5179b0 : FAILURE in 59m 57s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1680c125bd2f4999ae3fc300088d47d6 : SUCCESS in 1h 30m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa4958dd91f94ef4bdacd0377f1c2166 : SUCCESS in 1h 35m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25449a64a3af442499e56a4841a5669a : SUCCESS in 1h 28m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/dda7e77fa3aa41fab153d04345afbe3c : SUCCESS in 1h 28m 11s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba65930f8768414bab7160cb8702ee95 : FAILURE in 17m 58s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/308e1ba714f44f2bb33cd5ad280258c9 : FAILURE in 14m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e64771a05274f9a84b0d14f47c5d020 : FAILURE in 56m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/465422ccb7054ff68f8296599e441979 : FAILURE in 39m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85c86fe916034eab8ad2207c4bdded11 : SUCCESS in 1h 03m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f63ef701518649f6bcc51a246315089d : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"734d856b2e39f8ac3fa39351eb0de3ed2cffb2de","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-09 08:07:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"27b80714e98ec2af2c3663a7c4fd924f5f5d269c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 09:56:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1abf9a30b7cb4079a5ebb47afb33f465\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6c0bdcfe7e745b5be95940f2a05eb5f : SUCCESS in 12m 24s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3a9f256a29ab47ec9021d6cfee2e102c : SUCCESS in 1h 09m 37s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b7a1562e232c4441b9cc17a455a4580f : SUCCESS in 1h 41m 25s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/031db85bd28d4d3bb069f96a9d4caf9e : FAILURE in 1h 03m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/98c4d5eac230476ba39429326aa7562b : SUCCESS in 1h 30m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b9b7ed165884a2abaa1295712f24dc2 : SUCCESS in 1h 30m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ade0067208554b15a4cf884641b1f569 : SUCCESS in 1h 29m 13s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/f2ba1eee892b421f86d0b59b8d87c9d4 : SUCCESS in 1h 39m 25s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e78428676cef4b71a9d8d81c29536561 : FAILURE in 13m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca529a6b03ef4ca29e0ade92b4b642d7 : FAILURE in 12m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0db0195efc24a64b2e221f00a8b528a : FAILURE in 1h 11m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d46e0c0edf01462698261be8d8784bed : FAILURE in 38m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/047c400e279e421886ea824b599a2d42 : SUCCESS in 57m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/200dd61b3a0b4c68abda884f818c38b5 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"9aa22e9c94f2bf219ca773aa07d58054a5422dce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 09:58:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a061902ab55cacdbfff8dcefcd350f0af73d9554","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 11:52:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/293f0ec82af24f20807f338d99b988bd\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53a36230479145d88249182617f90828 : SUCCESS in 10m 00s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/12de24a272c6409c81af50f2632ce829 : SUCCESS in 1h 30m 52s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/3b0b18fe601549b58328ec3a9f916ad6 : FAILURE in 1h 47m 26s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/06e6c238edbf474d88d59565cb052672 : SUCCESS in 1h 25m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab3ef3430d9d4440bcabf239001219f2 : SUCCESS in 1h 30m 29s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/78735be6b20d482e803bc83806dff545 : FAILURE in 1h 48m 07s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffc8b6656cb24c6f9c9ca7ca57156780 : SUCCESS in 1h 02m 01s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7e18619fd56a488e8033ec106c076eca : FAILURE in 45m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69f704d8b156491e8e98ddb4c07722fa : SUCCESS in 4m 33s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f7a972fb43b439d933a593d59943539 : FAILURE in 1h 00m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3dd1ed2ce18ff50e5b22e1ea01f5eeb499a9d516","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-10 17:23:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b53d1bb30f9b663145aba42a552e10e13481234","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 19:48:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f237fe6e97934a4998004e80ecfc8a2d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce88ddf22859498f8585167cc32113aa : SUCCESS in 16m 07s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7654704408a845bfafa340aff68dce56 : SUCCESS in 1h 20m 02s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc959a052a594b719ce3cc0b7adf9ce3 : SUCCESS in 1h 42m 17s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f09d980631e24422bea004e0a567f724 : FAILURE in 59m 23s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3ba1887eb3324dc5a58a9c77d8468c99 : SUCCESS in 1h 27m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/343a40722454465a97f740db95eb4899 : SUCCESS in 1h 41m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15ff0981056149d09e04170ec1459ec4 : SUCCESS in 1h 32m 14s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/128e70e21510424788389970edd0d114 : SUCCESS in 1h 32m 41s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7ef33d7fdb5f4270bfbf80be479525b1 : FAILURE in 1h 20m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be25c99a786f4d66978f78d568892a8d : SUCCESS in 2h 17m 37s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/76bd52d05a22457ab71c036d50f73b42 : FAILURE in 57m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e0343cf175904cd382b8a8626067bbca : FAILURE in 39m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb670f5864724137be54f99098b556c5 : SUCCESS in 1h 02m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0da367c5044243a0911d7e5eb137a3b5 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2f214633040fcd3e2020984bc3ceaa6d5a286545","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 19:50:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6eada756d4700f1ae550c8d1281be0b4a74cebad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 21:42:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0fb14c62b83475ca15164b90d72d2c9\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0aeee6038684ee3a890f189d1c935dd : SUCCESS in 10m 48s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/085d8ac05945494ba75eb68481fad172 : SUCCESS in 1h 00m 43s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4df976e510548ce89cb57647e55f963 : SUCCESS in 1h 42m 32s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5b0008e102a14be69b7ac0fc8a2b2017 : SUCCESS in 1h 26m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58f744b33df74544b69b2ad6078a14b0 : SUCCESS in 1h 12m 12s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/1647dd61d79a41bbaa1cc8b4000ce527 : FAILURE in 1h 47m 43s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee00fed585774119a29bd814d720df8a : SUCCESS in 1h 02m 13s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ae7b5e1b0deb4070bb7c7f72c766b0fc : FAILURE in 47m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc3a7f23c0cb4d6ba70fff9149ef2616 : SUCCESS in 4m 35s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3cbfca02eb8e46c5965254351d1c71ff : FAILURE in 49m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60647dbfea4f1d872894b885f3dce291a3be8b34","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-11 10:15:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"854a3e5f98ba509d240dbf09f0ec4fed226fca14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 12:33:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9049cf01854241af9af7c0482ed563b9\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d11ff9f72afc4b07ac79be84cd077510 : SUCCESS in 11m 29s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d49ae8ff3efc453ba02d75071c591b53 : SUCCESS in 1h 21m 54s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/138c13b9d6584fad927889fc87f2b9d4 : SUCCESS in 1h 52m 48s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d47270d9c654d90b2f2b79a7a6ad080 : FAILURE in 1h 01m 19s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ac0ced7d0ed244528cf1bddcd3c82ee0 : SUCCESS in 1h 26m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/450729558caa4e088302315d53c38cdd : SUCCESS in 1h 39m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c98b45615d384f1eab85dbf469612caa : SUCCESS in 1h 22m 15s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/202c163e06d44dcab88a927edd39a8aa : FAILURE in 1h 47m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0b8fd59842f4ed68d5b42b28899d6d7 : FAILURE in 1h 13m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c87f596741fe4a7ba8fcc18cb98fa209 : SUCCESS in 2h 00m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb47f8ba159a4a0d9340204831c71f83 : FAILURE in 1h 03m 56s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/745f6d1c11104d4fa43d7fae08575389 : FAILURE in 37m 40s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f6a8c8a95c84130adea807b6cd73612 : SUCCESS in 57m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc39158641846b6b044b82dd92014a1 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"0e76bdd1fc8f0da3f07f979db4b519dc7b46be33","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-21 09:21:22.000000000","message":"Abandoned\n\nstable/yoga branch of openstack/openstack-ansible-os_octavia is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/yoga if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dccb3d78e371901f92e0045176bca9200fb69de8","revisions":{"dccb3d78e371901f92e0045176bca9200fb69de8":{"kind":"REWORK","_number":1,"created":"2023-05-02 11:01:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/99/881999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/99/881999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/99/881999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/99/881999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/99/881999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/99/881999/1"}}},"commit":{"parents":[{"commit":"f8b8e0bfabebc06558cca1fa3231bb5cc411b297","subject":"Adding octavia_provider_network_mtu-parameter parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f8b8e0bfabebc06558cca1fa3231bb5cc411b297"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-13 16:10:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-05-02 11:01:59.000000000","tz":0},"subject":"Ensure service is restarted on unit file changes","message":"Ensure service is restarted on unit file changes\n\nAt the moment we don\u0027t restart services if systemd unit file is changed.\n\nWe knowingly prevent systemd_service role handlers to execute\nby providing `state: started` as otherwise service will be restarted twice.\nWith that now  we ensure that role handlers will also listen for systemd\nunit changes.\n\nChange-Id: I4a3346c90825a4bf0b416943286696fa529f526d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dccb3d78e371901f92e0045176bca9200fb69de8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dccb3d78e371901f92e0045176bca9200fb69de8"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
