)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~885519","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I7afeaca12dd65e5455359e22c21a77191381bc73","project":"openstack/openstack-ansible-os_octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-01-03 10:24:46.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"34411":{"account":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"last_update":"2024-01-02 18:33:33.000000000","reason":"\u003cGERRIT_ACCOUNT_34411\u003e replied on the change","reason_account":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}}},"hashtags":[],"change_id":"I7afeaca12dd65e5455359e22c21a77191381bc73","subject":"Remove obsoleted provider drivers","status":"MERGED","created":"2023-06-07 16:02:32.000000000","updated":"2024-01-03 12:21:00.000000000","submitted":"2024-01-03 12:19:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885519","meta_rev_id":"fa5ae2c2919fc67e84874045efb58b78c0508eff","_number":885519,"virtual_id_number":885519,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-03 12:19:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2024-01-02 18:33:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-29 22:46:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2024-01-03 10:24:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2023-12-28 13:41:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-07 18:32:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-29 22:46:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-12-31 01:54:38.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"}],"messages":[{"id":"bbccbe8275bf7f7ef9a5aac5ab6658aae86af4ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-06-07 16:02:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9add8a3241b2d783ff096878e779368c1354df0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 18:32:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8af813f1b9c640aaba33b651dbc26beb\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43bcaad1582f49a382b23913cf00bffb : SUCCESS in 15m 22s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2254c8b8cce04bde95be89c358610a45 : FAILURE in 1h 28m 43s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a6fee79b2414e03b31ea8893a5e7db0 : FAILURE in 1h 01m 41s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e139765a5f3249f8b27bdd5d681377e8 : FAILURE in 1h 33m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1149103b67b54d7ca47f0f5cde23caf6 : FAILURE in 1h 36m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6c1239cdc2f4fb1965464cc59abf35d : FAILURE in 1h 25m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/465b23c02277483ab0dbc41ed86b1f10 : FAILURE in 1h 43m 45s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9fb81beaef944e2e9f56a141bf23be82 : FAILURE in 1h 28m 06s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/26ecd2552bfd4103827a560406b27a65 : FAILURE in 2h 23m 08s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed5e4772a5ac483a89d2e0773b0d3f90 : FAILURE in 2h 25m 00s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/43b54ada57fd41e1a9f51c837a97090b : FAILURE in 57m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a527d89288c4081b2e176faa219446d : FAILURE in 52m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/723d009df5484dceaf375d546b5f1fcc : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5a486869481889febcc2b114cdfc5a34d5914ae0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-18 13:47:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6e37039fa8bf0276c86588993d5e99d16a3cac3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 16:48:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/06278425f5ba49f59022130561b641c4\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/095176cc14244795b74fbe64d15a9fd6 : SUCCESS in 11m 13s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6b18e1b4a3c4498f8285d6ab19f4e557 : FAILURE in 1h 28m 32s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/bfb46a5f4fa44fee8f561bb94fccfadd : FAILURE in 1h 37m 38s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/de9fa5d25bba43a1801d6f967d3aa247 : FAILURE in 1h 44m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e79304ec0f1e41ad9c70330c6f1895db : FAILURE in 1h 32m 29s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7a7c157e3c8247049fe49155ae86f08e : FAILURE in 1h 47m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35f0cd88d2d54428ae81ca95a42a1956 : FAILURE in 1h 36m 02s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f83e65bd9c504837a097cc211a5023c4 : FAILURE in 2h 51m 04s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfb749e0d10c481f9f817e6ba1aee4a6 : FAILURE in 2h 41m 24s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aea2dc009eca45b0aa5ed52e284b10d6 : FAILURE in 1h 00m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72d80a616d5b45b1a9a2fc5dc4d76d8d : FAILURE in 1h 01m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82575a54f19c48a3a868127c0cc23628 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"2ec22d665f55a812aab6954726b6110de27cf072","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-07 19:15:35.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a34982150eb62fd46d1308d308865b6e80fc25e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 21:33:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4902b91095045cdb6c8b7941847da88\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/635c31a10ed342d2b7cd066a2664e381 : SUCCESS in 13m 15s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/50f3cd322d6b40c886b566e3a5475254 : FAILURE in 1h 17m 46s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2e3140125ffe47b68baf1e753800b136 : FAILURE in 1h 25m 32s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d025f4c745344fd3a302d1320a73abdb : FAILURE in 1h 32m 31s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e66e1991643540baae0523034c3e2f97 : FAILURE in 1h 46m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c369e4ac09074e639fed18c2e10ae57c : FAILURE in 1h 35m 52s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2e3a875c253a4ba2b9fdb41afb2c4836 : FAILURE in 1h 34m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b735c4607fff4686ba860453548fbf84 : FAILURE in 1h 33m 44s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b9baae4d004f4f4284bb6c09fdb16309 : FAILURE in 2h 06m 52s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2cc91dd30cc482baa8b96860c3e48d5 : FAILURE in 2h 01m 19s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/794ac16218584542a6635914dc13e069 : FAILURE in 2h 10m 17s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd98f220dd25424aa5168ff7ad38b046 : FAILURE in 2h 09m 58s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a5ad5d86f3da4b80a2d015a89d5dad66 : FAILURE in 56m 42s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72b38cfcde5e47bd9c361cc46a821b41 : FAILURE in 1h 07m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c5d4e279b6d4821af08c0beee54aa1f : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":3},{"id":"b69fee41c96b91228567c587d40687d315e79a81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-08 08:44:10.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"798d0ae21e1ceff72ad4a6c6eab07237b664ce9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 11:37:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/659c974629304e96b7a0068f8de993f7\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66faf39735854e49a5a2476e72075b73 : SUCCESS in 12m 42s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6ea4a4604927406bb97a9c9466820bfc : SUCCESS in 1h 10m 14s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d91e8d8abf8344c3a0cc310e482de531 : SUCCESS in 1h 24m 38s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/af005fb1edfb40be91173d9814b83397 : FAILURE in 1h 41m 27s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7cdd496c14554a8299a156423d505b7b : SUCCESS in 1h 19m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d35e70145f04692abbaac41f31a3675 : SUCCESS in 1h 17m 04s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b1d2c92f8b3b439b923d6224a0a175c4 : SUCCESS in 1h 29m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/629455cac3f64e56a4985bd24b0c721a : SUCCESS in 1h 22m 40s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b6ad19df6b6149b2b3e5fe6279c27ff1 : FAILURE in 2h 14m 50s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57c2e986de244a0faff9beb1d5b63881 : FAILURE in 2h 29m 07s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/967c7c5aadfc42d79feeaaf1b4cae101 : SUCCESS in 2h 51m 15s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/307f0037c66645dc9a751e2352697dd1 : SUCCESS in 2h 29m 29s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/62dae23fbca34f33b6d3fe0db04f985e : SUCCESS in 1h 05m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ce89fb933aa4a2e8da2dc82d16ad579 : SUCCESS in 1h 07m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f04c66e65da14cb4a18bb71b61463be5 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":4},{"id":"78bb3524ce479ae028b44fa71bd4b08240d48a51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-12-05 14:06:47.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"833bb8ac8a397a5414ada4e82e5c07e88e87332d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 17:18:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e55b50e22dd46459fadabda57680d2a\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed0c68ac36fa465ea6f14d6bde1ac623 : SUCCESS in 10m 25s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8e28d1df63a8436aa32a8d8ac002e981 : SUCCESS in 1h 19m 37s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/a0e2eb742ffd47928eaabd805292597e : SUCCESS in 1h 25m 17s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/15d3e4a04edd4f8eba2fef122838328b : FAILURE in 1h 41m 10s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bd4da39d258547478df6d310456b4403 : SUCCESS in 1h 27m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42297685280b451cbb2da5a20b9dee88 : FAILURE in 1h 29m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0eeefdff35484367bffd17235005d695 : SUCCESS in 1h 32m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f2bb244fb2c44ca990682c2c2d733d1 : SUCCESS in 1h 27m 36s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97141fe72500478eb1aeb69c509bad03 : SUCCESS in 1h 23m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b4be3e949e474bf89c5411c893846375 : SUCCESS in 3h 02m 12s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4078013d5dbd4bdb9ed51469f5fd0bbb : SUCCESS in 2h 32m 53s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/df9dc43dce8b41409d99b2d1912d8197 : SUCCESS in 2h 36m 49s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdf43da788fb46cc82e982989921e920 : SUCCESS in 2h 30m 31s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3d10837d7f564444a1a27e0a3c758e08 : SUCCESS in 1h 04m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96f3fb5dd2b1497c84e4e5ccf6bbb0f5 : SUCCESS in 1h 05m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be874689689c497daf0b0eead9c58996 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":5},{"id":"783fbf0108233a0d7833644df08b82b0cb64c1a7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-12-28 13:41:09.000000000","message":"Patch Set 5: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7a7a04859f221655ac2c06c5e2650e8f3e0e6b4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-28 16:26:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b6baf1bf7d64c5e8423aeb823856afc\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9fa8de531a549dc973ce96313057060 : SUCCESS in 10m 24s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4f40e8d1a4d549b8acf639738c121155 : SUCCESS in 1h 22m 57s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d887080836c54ee38708aca0309e0731 : SUCCESS in 1h 18m 56s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/45451f9cd41f4169a81bf7e29c363b13 : FAILURE in 1h 33m 45s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6d2ba955e5c24494859cfabd8e92d0b0 : SUCCESS in 1h 30m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c8f4d28ad514bd99a83daf84c9f3305 : SUCCESS in 1h 19m 45s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6ecbf6e165024ea4999849cb0a559939 : SUCCESS in 1h 32m 45s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac2fea49dbca43babda09f7d45637d32 : SUCCESS in 1h 24m 13s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43166342eb6244e7b14bd891c0c0fa52 : SUCCESS in 1h 25m 15s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d35f7304f1864116bbd188d4de47599b : SUCCESS in 2h 39m 09s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2c93f8a27844ed0a73a4293496576f0 : SUCCESS in 2h 11m 01s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ab9cd9231f62463e9cf852e1544b80cf : SUCCESS in 2h 29m 07s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9fb37d0e433f45a881d9c8146ab2c613 : SUCCESS in 2h 21m 52s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ddf5016483b14b74b786e8639d2b377b : SUCCESS in 1h 04m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7422fa7b0a964b3faa484378af9ca2d9 : SUCCESS in 58m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a19d906ed2074a04991a3266a25ca31c : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":5},{"id":"406cb9e8e8baee0d3dacb105d1310d6a97a581f8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-12-29 22:46:26.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"c92a89573839dfb602f34f30a40889aacfc62342","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2023-12-31 01:54:38.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"126dc1dfbd1c03c717975f3bcffca7c352feea27","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-01-02 12:58:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c3e873d7d20aabf9270d64e6871e7b4992d20470","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-01-02 12:59:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6f8c91e4abdfebff9b7e207405ca76efe805c926","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2024-01-02 18:33:33.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9c911e6964d445a3f608348e2db8adeff1b84a71","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-01-03 10:24:46.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"adc0d8a8e4592aa4e0aa189d96093d24a1014fd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 10:26:53.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"bd71c2e86fb79ba82f5957f1f344f981d8bdc3c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 12:19:49.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9f42a6afbf940329d6659972406be3f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e19278a38c443a7bc1f223b10f16a06 : SUCCESS in 12m 23s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b4a8b1478337479fac3c38d9c1315846 : SUCCESS in 1h 29m 41s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b93e6dfa932a4afc876546735c4ed226 : SUCCESS in 1h 02m 53s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/55fe88b8e7e74df9b995e0b188f7f045 : SUCCESS in 1h 36m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70a258912f5c4681a1caeaf0efbba68d : SUCCESS in 1h 12m 55s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d25388abbdb24bdb99deb30b487ab935 : SUCCESS in 1h 17m 55s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/68cf9e0ff3b64a1aaa67054db24d3627 : SUCCESS in 1h 12m 37s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fcd63a6ba7124880ae3652ff012e8cc6 : SUCCESS in 1h 21m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0231244dbeb54835acaa5ff1c71353d9 : SUCCESS in 4m 13s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a7a1f6ddb3449b68759f23b47a5b894 : FAILURE in 1h 46m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c8cbbbc72fbc93a3e4df25943dc1b7bc21331e7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 12:19:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"fa5ae2c2919fc67e84874045efb58b78c0508eff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 12:21:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b29f6ba017fc4bd8a2b2ddc38012a6da\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4f235a746304b33b43dea6345c20bf4 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"34e0def6ec0ab10e5579555797ef5bf7a846d718","revisions":{"a81f10a2f89cfe2567bce223d676d941c7426363":{"kind":"REWORK","_number":1,"created":"2023-06-07 16:02:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/885519/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/885519/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/1"}}},"commit":{"parents":[{"commit":"048f9d548ee09a7c76cc5efaef9c6c56aafd9568","subject":"Merge \"Add TLS support to octavia backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/048f9d548ee09a7c76cc5efaef9c6c56aafd9568"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-06-07 16:00:40.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-06-07 16:00:50.000000000","tz":120},"subject":"Remove obsoleted provider drivers","message":"Remove obsoleted provider drivers\n\namphorav1 driver has been dropped from octavia early in 2023.2 cycle [1]\n\n[1] https://opendev.org/openstack/octavia/commit/6c0515c9880b5310dd0647a5881233f07898d812\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/884924\nChange-Id: I7afeaca12dd65e5455359e22c21a77191381bc73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a81f10a2f89cfe2567bce223d676d941c7426363"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a81f10a2f89cfe2567bce223d676d941c7426363"}]},"branch":"refs/heads/master"},"81319902df3e8fd92cf1931285582e3c304f6399":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-08-18 13:47:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/885519/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/885519/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/2"}}},"commit":{"parents":[{"commit":"138696d0f5e4ae7d0f5e448845eb026e01ac9748","subject":"Define constraints file for docs and renos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/138696d0f5e4ae7d0f5e448845eb026e01ac9748"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-06-07 16:00:40.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-08-18 13:47:47.000000000","tz":0},"subject":"Remove obsoleted provider drivers","message":"Remove obsoleted provider drivers\n\namphorav1 driver has been dropped from octavia early in 2023.2 cycle [1]\n\n[1] https://opendev.org/openstack/octavia/commit/6c0515c9880b5310dd0647a5881233f07898d812\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/884924\nChange-Id: I7afeaca12dd65e5455359e22c21a77191381bc73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/81319902df3e8fd92cf1931285582e3c304f6399"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/81319902df3e8fd92cf1931285582e3c304f6399"}]},"branch":"refs/heads/master","description":"Rebase"},"ba7a889fb9a62fefc50fed459440e8a4c3c06568":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-11-07 19:15:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/885519/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/885519/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/3"}}},"commit":{"parents":[{"commit":"8da3bddd996f4a77007fcea586d945e6e2403a47","subject":"Merge \"Add quorum queues support for service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8da3bddd996f4a77007fcea586d945e6e2403a47"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-06-07 16:00:40.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-11-07 19:15:35.000000000","tz":0},"subject":"Remove obsoleted provider drivers","message":"Remove obsoleted provider drivers\n\namphorav1 driver has been dropped from octavia early in 2023.2 cycle [1]\n\n[1] https://opendev.org/openstack/octavia/commit/6c0515c9880b5310dd0647a5881233f07898d812\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/884924\nChange-Id: I7afeaca12dd65e5455359e22c21a77191381bc73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ba7a889fb9a62fefc50fed459440e8a4c3c06568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ba7a889fb9a62fefc50fed459440e8a4c3c06568"}]},"branch":"refs/heads/master","description":"Rebase"},"203463ef157800b0ef9b7a61c4280db9da50f0b3":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-11-08 08:44:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/885519/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/885519/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/4"}}},"commit":{"parents":[{"commit":"8da3bddd996f4a77007fcea586d945e6e2403a47","subject":"Merge \"Add quorum queues support for service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8da3bddd996f4a77007fcea586d945e6e2403a47"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-06-07 16:00:40.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-11-08 08:44:09.000000000","tz":0},"subject":"Remove obsoleted provider drivers","message":"Remove obsoleted provider drivers\n\namphorav1 driver has been dropped from octavia early in 2023.2 cycle [1]\n\n[1] https://opendev.org/openstack/octavia/commit/6c0515c9880b5310dd0647a5881233f07898d812\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/900399\nChange-Id: I7afeaca12dd65e5455359e22c21a77191381bc73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/203463ef157800b0ef9b7a61c4280db9da50f0b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/203463ef157800b0ef9b7a61c4280db9da50f0b3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"34e0def6ec0ab10e5579555797ef5bf7a846d718":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-12-05 14:06:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/885519/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/885519/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/885519/5"}}},"commit":{"parents":[{"commit":"e3c74d18e3d79b512f68053120f0b97169e0f58b","subject":"Adjust condition for availability_zone definition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e3c74d18e3d79b512f68053120f0b97169e0f58b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-06-07 16:00:40.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-12-05 14:06:47.000000000","tz":0},"subject":"Remove obsoleted provider drivers","message":"Remove obsoleted provider drivers\n\namphorav1 driver has been dropped from octavia early in 2023.2 cycle [1]\n\n[1] https://opendev.org/openstack/octavia/commit/6c0515c9880b5310dd0647a5881233f07898d812\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/900399\nChange-Id: I7afeaca12dd65e5455359e22c21a77191381bc73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/34e0def6ec0ab10e5579555797ef5bf7a846d718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/34e0def6ec0ab10e5579555797ef5bf7a846d718"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
