)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~897350","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I19aa9c5e915aa7e71617de6556d1eccff1a93ebb","project":"openstack/openstack-ansible-os_octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-10-25 21:08:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I19aa9c5e915aa7e71617de6556d1eccff1a93ebb","subject":"Fix example playbook linters","status":"MERGED","created":"2023-10-04 16:28:24.000000000","updated":"2023-10-25 21:09:14.000000000","submitted":"2023-10-25 21:08:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897350","meta_rev_id":"fd9e4c22de2390992602b7796d95e6912aebecd0","_number":897350,"virtual_id_number":897350,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-25 21:08:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":2,"date":"2023-10-17 22:41:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-17 16:22:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":1,"date":"2023-10-17 22:41:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-04 20:02:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-17 16:22:22.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-10-17 22:41:15.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"}],"messages":[{"id":"dee85a0ddd05e905bd2b6298f5e9910f5e168243","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-04 16:28:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0763b0a8f92ecbcb52f0fe6c45fab6e7aeca7fc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 20:02:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bac39505820d4d03b7016be4eaf5706c\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a84fb3eb83d246e7bf5b9f0d573aa8bd : SUCCESS in 15m 02s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/490fa517f25c4e57b4ca6f1746266a54 : FAILURE in 1h 31m 47s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4fe78ed524474b5f89a50696e689cd30 : FAILURE in 1h 45m 48s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7b250e2207ad4107a5d6703763f01552 : FAILURE in 1h 54m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba38d5a1f84446eb91e5df42b9aecb47 : FAILURE in 1h 47m 19s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6dff501e371d4d0ab414322253b7002c : FAILURE in 1h 42m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/503880c9243c4fd486e9790cabdd9554 : FAILURE in 1h 46m 50s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fbd75c7630f04dd68776ab4cbae7b2ad : FAILURE in 1h 18m 17s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/670668f36368463692313c4a57b2d505 : FAILURE in 2h 40m 28s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/79c6374e36c64c87b37326e1bf7ab10e : TIMED_OUT in 3h 09m 34s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9d330390bb04303aa74f53b664fb04c : FAILURE in 26m 03s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c5f6ab8441984a2c9c712bf3a2ecbf9f : FAILURE in 1h 00m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00f4de59d7134873818da22cd7a71a7e : FAILURE in 1h 04m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/968df093cc1943a7a278a530a16f221b : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c5298c2c60b211ca50a6ca6cdfd3de05f3b44512","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-05 07:01:08.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8b1128460e7547fc16338ccf3d93a6eebccb2675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 10:12:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c94f2054b5949e1b8c336fb222491e7\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0de764b399e74f7788651e19771ef864 : SUCCESS in 12m 43s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1e203a6d28e34ec7996b524741f7857d : SUCCESS in 1h 52m 12s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8e46e4e2db74224b179f6704f099e8b : SUCCESS in 1h 26m 22s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bdffbcd14ef2463ca0ea350936daf6db : SUCCESS in 1h 56m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e232102a13d48338af5d15d13c7d2e3 : SUCCESS in 1h 59m 42s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cf05242c15d04767b8a622cdf6922d63 : SUCCESS in 1h 30m 17s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8064a8b2f9ca4853a890dab72904b1ec : SUCCESS in 1h 33m 04s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/151fbd67c6674b7cbb0cc36181669820 : SUCCESS in 2h 52m 58s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d677b71ba964ba39f824c06366da4eb : SUCCESS in 2h 02m 27s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f5fa3eeb8d9d468f8343d3ec3a565a2f : TIMED_OUT in 3h 04m 23s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a662ef596f24a91ada39675cd27f906 : SUCCESS in 2h 44m 37s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ea76de147262425b8c428b611c1fc25b : SUCCESS in 1h 11m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfd6f5ed2ebc499f960e642ccfc5b6bf : SUCCESS in 1h 11m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd5246e409f94e8d9a620918c678a60e : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":2},{"id":"6bad0617fbc3a127325cfbc03c5ab2b42d219992","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-17 12:38:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1c9b522124b6c90e266c1ed9ab74db3932c2a474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 15:48:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73619de7a6a04d20a74331525d32b4bf\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/234c53a9b2564895a199d8f93813888e : SUCCESS in 12m 35s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7a34976572bf446ca131039bab93f30b : SUCCESS in 1h 46m 21s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2574b0733cf4e10800ec5e994208104 : FAILURE in 2h 09m 39s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/74b759010a574b9787f1bacca52c73dd : SUCCESS in 1h 58m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87cb3642898746e78eb8a296c2584e35 : SUCCESS in 1h 59m 43s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2c761b95ad7645a19fab5c3861dabc3b : SUCCESS in 1h 46m 42s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7376c71a87d941189f62cb1c62434b0c : SUCCESS in 1h 46m 47s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b368ee43370242c0a6cc2f594fcce70b : SUCCESS in 3h 01m 21s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13e989e8b3df438ca183fc7fda965541 : SUCCESS in 2h 51m 00s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d1d261a62b564e4c893e9d1843e971a1 : SUCCESS in 2h 39m 22s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc1fc7d33612422e95c7c084efe1737c : SUCCESS in 3h 00m 27s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3d30979afea2415bad00cd33fdefdd20 : SUCCESS in 1h 16m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15fa11de26844decb126e1a77d2c45a8 : SUCCESS in 1h 03m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad4bb09867914294b3e68c90f7df7221 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ee9158bcb07b3c2e4f1af635d2d39ce118c87518","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-17 16:22:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2d032407c78b939b443a6a8556c37597520c7f0c","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2023-10-17 22:41:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6caf3474943313d84c136c003f307ae33c2a9cd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 22:41:26.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebdac5c3bfd54b8c9a5a8d7088dd23ca\n\n\nWarning:\n  Change 896017 in project openstack/openstack-ansible-plugins does not share a change queue with 897350 in project openstack/openstack-ansible-os_octavia","accounts_in_message":[],"_revision_number":2},{"id":"f723e3b1698c932d20efe2527b008a87735b4d1e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-19 07:09:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5dcff4c560d6a72351e07bfdba7375ab926cc379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 10:08:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/452ad488d9ec45c8bd342483aa57eddf\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b87994854b74189a4acae8487e29885 : SUCCESS in 13m 15s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/99cf7057bc314bb2a1dd298ba339eedf : SUCCESS in 1h 47m 02s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ab3bee3d9ee40a086b860f0c8f92991 : SUCCESS in 1h 50m 45s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/56dfa660a9a04deea20c5e77b5476cba : SUCCESS in 2h 00m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1726be13347d4b8fa94dabc677e322d7 : SUCCESS in 1h 48m 12s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/14ceca4e8f2948cb9a0fc89bf15d04f4 : SUCCESS in 1h 49m 53s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d7183d77d2049c8a6b59b583e50cea2 : SUCCESS in 1h 42m 43s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/02ac5fc8110f4c33afe1abecd62f4bc4 : SUCCESS in 2h 52m 25s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25ae3a7b171840ebb618dd015d6a4a74 : SUCCESS in 2h 28m 30s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c0e7c95e483143c983b5713c06f47207 : SUCCESS in 2h 50m 15s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9a7190c64fc432f8b29c86c7ae71416 : SUCCESS in 2h 38m 24s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aa0ed6f491d9459fb2700161f6b2aab3 : SUCCESS in 1h 20m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c6833b5f21a4751b614e34ebf2e6082 : SUCCESS in 1h 19m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f247ceceafad426ea9119e425a99c4c2 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"099b5cfecbaf0a46b0b6d947c3d2ae875faa4b66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 10:08:45.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/342b62e7cad24589bf8864b53b5614f5\n\n\nWarning:\n  Change 896017 in project openstack/openstack-ansible-plugins does not share a change queue with 897350 in project openstack/openstack-ansible-os_octavia","accounts_in_message":[],"_revision_number":2},{"id":"96f299dca0d2b6919e436db7e28e60678c16a127","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-25 15:34:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b95846141dae62571fa63e6887b76a48a12f2fff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 18:46:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e382ea50b45f4f8cafaf8d8ec7579f77\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/676fd515010a405797097d19c2634e83 : SUCCESS in 13m 47s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d4fe5aa6fb1b4059b8bbb99fb7e4db16 : SUCCESS in 1h 46m 09s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/887ba69cff5641379d13133fd999395b : SUCCESS in 1h 35m 15s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3a784a830cc45f8a3d875b6cb9ce5e4 : FAILURE in 1h 47m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fbf31713b7554f988273757672287070 : SUCCESS in 1h 32m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4fde3b560b34b308c8feb0ee4dc75f2 : SUCCESS in 1h 29m 39s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c88850f22bc046ff83e13bab673d0f54 : SUCCESS in 1h 58m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67b803d637d0417380513477a754544b : SUCCESS in 1h 31m 20s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ef82fb179a52484393640f5b21c44e3f : SUCCESS in 3h 03m 39s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2234ea768b5417fa854b7c9c439292f : SUCCESS in 3h 01m 59s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6123b9596d6a46beabe2c74939b997b2 : SUCCESS in 2h 58m 40s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70b90b0b3f8d4fdeb628c652442410d7 : SUCCESS in 2h 22m 08s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/51f669c266334cb28422f1c8a1657b9d : SUCCESS in 1h 12m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7a249aa7064421dbe8f51a72a70c5b0 : FAILURE in 1h 20m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d944a8bed1f741b889371115dd6a4a01 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"3fc5c8bedf05f51df9c5e4e88cff60ee857196ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 18:48:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"92eb12ad3c4e68575ec58eb9e14d578737fa9fee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 21:08:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5a08b2342da41ee965969fc130e8108\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f09b89839994123a5c2a09c80439b9d : SUCCESS in 11m 00s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5ecc99e385904dc7b7231d4124a120ad : SUCCESS in 1h 50m 55s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d8186a528d254cb3a40d486c1da48f70 : SUCCESS in 1h 43m 27s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d3323ca857d146c6aafee9f597630945 : SUCCESS in 2h 07m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d222cd27fccd4ea6bc3838b6d6c838f7 : SUCCESS in 1h 53m 32s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7cb67b0fb00b40209250c0c9f3bc67a6 : SUCCESS in 2h 02m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9273062ec9c4ea19f13b77c095123b4 : SUCCESS in 1h 47m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fa8911a1c2c442ea51121a715c1bef8 : SUCCESS in 4m 44s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/638e9be6c8644811a503e589cf4ef3c4 : FAILURE in 2h 15m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d02b11b79b8eca4c822dfa6bb0a7570ed2aab3dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 21:08:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fd9e4c22de2390992602b7796d95e6912aebecd0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 21:09:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50c27f806c3240debbc9070e5432fc0d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abc574aa31714eac819fbb0db0131a7b : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3a7c5e66340967cd1d8a2f1269e512960cb6acd8","revisions":{"69200d80347360fc8912a7ee4885f6693a5eae46":{"kind":"REWORK","_number":1,"created":"2023-10-04 16:28:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/897350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/50/897350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/50/897350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/50/897350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/50/897350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/50/897350/1"}}},"commit":{"parents":[{"commit":"1ae94217ecb1f825baf6009b2cd55c017dffff95","subject":"Use proper galera port in configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1ae94217ecb1f825baf6009b2cd55c017dffff95"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-04 16:28:16.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-04 16:28:16.000000000","tz":120},"subject":"Fix example playbook linters","message":"Fix example playbook linters\n\nChange-Id: I19aa9c5e915aa7e71617de6556d1eccff1a93ebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/69200d80347360fc8912a7ee4885f6693a5eae46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/69200d80347360fc8912a7ee4885f6693a5eae46"}]},"branch":"refs/heads/master"},"3a7c5e66340967cd1d8a2f1269e512960cb6acd8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-05 07:01:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/897350/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/50/897350/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/50/897350/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/50/897350/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/50/897350/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/50/897350/2"}}},"commit":{"parents":[{"commit":"1ae94217ecb1f825baf6009b2cd55c017dffff95","subject":"Use proper galera port in configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1ae94217ecb1f825baf6009b2cd55c017dffff95"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-04 16:28:16.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-10-05 07:01:07.000000000","tz":0},"subject":"Fix example playbook linters","message":"Fix example playbook linters\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/896017\nChange-Id: I19aa9c5e915aa7e71617de6556d1eccff1a93ebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3a7c5e66340967cd1d8a2f1269e512960cb6acd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3a7c5e66340967cd1d8a2f1269e512960cb6acd8"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
