)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~899045","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~stable%2F2023.1~Ib6b8547b69949f7af0ba0f7f436b4286d3baccb7","project":"openstack/openstack-ansible-os_octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-11-14 19:45:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib6b8547b69949f7af0ba0f7f436b4286d3baccb7","subject":"Add security rule for octavia healthmanager","status":"MERGED","created":"2023-10-23 08:01:40.000000000","updated":"2023-11-14 19:46:13.000000000","submitted":"2023-11-14 19:45:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899045","cherry_pick_of_change":897316,"cherry_pick_of_patch_set":4,"meta_rev_id":"e867fa69f2bb9965e7a8c7e7e136b82bb77facd8","_number":899045,"virtual_id_number":899045,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-14 19:45:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-10-23 08:40:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2023-10-23 08:03:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2023-10-23 10:41:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-23 08:01:40.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2023-10-23 08:03:59.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-10-23 08:40:02.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2023-10-23 10:36:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"adda56bca53d36dbf49b1f81d651d4c67d420772","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-23 08:01:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9e766dee4f7f50b7cce94781f5011f0b25487c31","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-23 08:03:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a7276a0d41a4497e1cb5da105d9c20ebb0626321","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-10-23 08:40:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"52c5d2177fb95a5bd8a9365e169cdce01667efdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 10:36:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb440585e9894dd982a5f16887e3e874\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae975cc9ae1a4398884b63a198b76f0b : SUCCESS in 11m 10s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c1bccc27b1f546f39a17d35d44058fb9 : SUCCESS in 1h 43m 38s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6cc74d1e90644243ab1a58dca427c2e6 : FAILURE in 1h 31m 04s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7e3de25627034415832f8a530c9896b7 : SUCCESS in 1h 16m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0803844c5aa7482486e40ca394d82d5f : SUCCESS in 1h 23m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfb68370f03649988bcd054a06637afc : SUCCESS in 1h 35m 46s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7b40d967d8f74956b3060cd789704049 : SUCCESS in 1h 53m 50s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb777127b2c04100834cc2701436bb95 : SUCCESS in 1h 35m 33s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/842943c487b446febac4b62ccd42c9f6 : DISK_FULL in 2h 28m 09s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee57e24d75874c59b51d43930aaf86d9 : SUCCESS in 2h 08m 36s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/208bf80588034477bdad38516655b7eb : SUCCESS in 2h 13m 20s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/59bec669cfc74649a03ec3a6fbd73749 : SUCCESS in 1h 05m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8d1c6836c2544c88b530070aa295b7a : SUCCESS in 1h 02m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91ec4690dc41488ab004d6902420ec2b : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"956f322bba2175af3e22478c5193e6904c38c863","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-23 10:41:43.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a8e1eef64e1b4523e0e4ab5e0ba4ea2d4e7b5912","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 13:17:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f99c035a5fd427389e700f7881a1ec9\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/686f7f1588d748f1aa2c33d702e2d7f4 : SUCCESS in 10m 20s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2a4465e3a2a94daeaec2d6522c59c536 : SUCCESS in 1h 33m 37s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/de692d77f998414eaa93c35e4b52e2f3 : FAILURE in 2h 02m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6196d43f2fe346638eecf4be97c8b716 : SUCCESS in 1h 52m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d53162a34543466ab794b78aa14dd195 : SUCCESS in 1h 42m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb4e18e8c24c4e0fb29ff0a9519b5525 : SUCCESS in 1h 45m 33s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/219ab722f6a14c7a8cc269a9ff55cd4f : SUCCESS in 1h 43m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16a0e292bc564e048aac535b30f358e2 : SUCCESS in 1h 45m 16s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0564d0d165c44b42891c7032ccc26bd6 : DISK_FULL in 2h 33m 18s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b9d67613111413cac8306782570d8fc : SUCCESS in 1h 46m 54s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8419fb02afdd4586937f1c4dbdc64741 : SUCCESS in 2h 08m 00s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9b0e013fe0bb455e970ea9afbead2be0 : SUCCESS in 1h 04m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82d8ecbc9505466eb5c3613f29263ad8 : SUCCESS in 1h 00m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff8e7ac3b4764870b5c3c28f3e8467b3 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"22eda37a9c68f02e5dfab9cb8166d0e00073d92e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-25 08:51:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8cdf35dcf0130a8a785afe86310116822aee72b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 11:28:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65ad4bc46e49433abcf4e8f8c294957c\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c723c6277e1a49e5a0f5e720004d02c6 : SUCCESS in 11m 41s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e4ba948ffbd54dcb9507cf8bc5f9707b : SUCCESS in 1h 28m 31s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/436aff1315734f418c7869bd29097aea : FAILURE in 1h 44m 28s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/06c415c46c4e45c4825bbacd28e94d5e : SUCCESS in 1h 43m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92bf003850694cab9d3c8e0e64360919 : SUCCESS in 1h 39m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c3282a76bd04fb7ac928ef193a50c07 : SUCCESS in 1h 35m 20s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6975e51f439b4b72aa53dddeada3efae : SUCCESS in 1h 54m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dea60a68dcbc466a80b4f74c8d03fe17 : SUCCESS in 1h 59m 01s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/719ba98ebcf341ec8b2237a18e12fbec : SUCCESS in 2h 34m 12s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4de547bbbad24345a1d895254400ce87 : SUCCESS in 2h 29m 20s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52c5f484983140a1ae1af38387ee6488 : SUCCESS in 2h 25m 43s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/46f37ad460f2464fa93c9a095c130e77 : SUCCESS in 1h 10m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7abc96abe9041b6a218f082995edab3 : SUCCESS in 1h 09m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46b2d13c5b794ccb9fe191aaa883b6a0 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e339a3940f2f8290847a743c5c7fb9bc47b6c20c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 11:28:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f4e4409d10044d4a0894f21b4beda1a\n\n\nWarning:\n  Change 896017 in project openstack/openstack-ansible-plugins does not share a change queue with 899045 in project openstack/openstack-ansible-os_octavia","accounts_in_message":[],"_revision_number":1},{"id":"10040b13fb10b82f859905e95547d283d2cefdcd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-25 16:03:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"588806d07c29240d768c0e2ee359ba56a8c0a55e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 18:58:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1926650c7b9142ee9f09e30a04b034fe\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6529aa4ca89c4cb3b2bee9c484febcc4 : SUCCESS in 12m 29s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f1a670e9a41547ffaad4c12b307c0bcc : SUCCESS in 1h 17m 21s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5829892c331d4de0bd15cd2181bd908a : FAILURE in 1h 56m 58s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c73eb2848a0e46ef8917e7a2878f32c5 : SUCCESS in 1h 48m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af9763d9cf454c42aaa12a9c5d57f920 : SUCCESS in 1h 53m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41ceb655c98f4379bb3f0599433ba460 : SUCCESS in 1h 37m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/81797494ce5448b1b7c9344a908162c9 : SUCCESS in 1h 44m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60e706f051f64f8bb07a8c3f4e7fa337 : SUCCESS in 1h 49m 36s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dcb02969767b43dab997e8512ad06aab : DISK_FULL in 2h 47m 39s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2644ac23b1944a1af519c0e98908f71 : SUCCESS in 1h 49m 32s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cecc66a70c6452fb0d1bd3de22a9387 : SUCCESS in 2h 18m 55s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/05789dad233d413097171bf1be1d6096 : SUCCESS in 59m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0fb2efb3247418fac3e9214e22200e0 : SUCCESS in 1h 01m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b92bd619dbd44643b4d09bfe1d0d4a8b : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"70ef82ec87269010880a93a8894fe3bc6b2a51be","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-26 16:12:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"399cf935e060fa12c97b115b0ea2f7c0321f8137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 18:41:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/804c64fce44047d58e675ad2245b81f5\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee3d18b5b0674e36901457599b279682 : SUCCESS in 14m 58s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2840046c72304d26ba0e6c0f1bfcfa86 : SUCCESS in 1h 27m 30s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b54115ef64d047adb5369374619c2afc : FAILURE in 1h 59m 08s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/41ab1d69a1994d28929bc7d741bb1c5d : SUCCESS in 1h 45m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be4b089654d141df9a34fbdc323af081 : SUCCESS in 1h 40m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73b254dd53574a78a023241700ee9558 : SUCCESS in 1h 40m 21s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/85ba4f51ae8c40d797dba753cc95cc23 : SUCCESS in 1h 45m 33s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c394c46c0cf4328b55522291317d5da : SUCCESS in 1h 45m 13s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2ef79972d22b4a788107bdbc5bd5daa5 : DISK_FULL in 2h 22m 49s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9dd036829209427182ed7bcb73972e52 : SUCCESS in 2h 16m 39s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93af0db9515946f29e5b909530752ece : SUCCESS in 2h 14m 34s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/71d6003c5c284d56b910eeb532604017 : SUCCESS in 1h 00m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2c572b6bec04755810db6436f369397 : SUCCESS in 1h 04m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b3671717f49431a92dec941d81093d3 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7964f337e9ea73a78df7521de7cc7580cefa03b5","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-11-02 11:50:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e8b67f51bb810340e350c51da49e27f3b8eba7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 14:26:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20b871da979641118d58764ee4c06cf5\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d247d5d20794f67bce23f79011ea7de : SUCCESS in 12m 19s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e95c8492409247f28aff07c40a96f4fb : SUCCESS in 1h 36m 09s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7e3259bcf914cf58782db067f4c11f4 : FAILURE in 1h 58m 26s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1e84a7d5bc534a0481a2aedaeb4e2853 : SUCCESS in 1h 36m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43d37b0bb5ed408e85bfa0c3c2ba0cbd : SUCCESS in 1h 49m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9979721baa34613a8aa77ef89d20a2d : SUCCESS in 1h 33m 49s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/31507435b2cd43e68ae6d85d76de30b5 : SUCCESS in 1h 50m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b61587a7a36342afb88ebde3304b449e : SUCCESS in 1h 44m 05s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b0375f4e377f4f87a8874a0da6681cc4 : DISK_FULL in 2h 29m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00ef2b45042c474d99e3d48cc83012d1 : SUCCESS in 1h 48m 17s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f02f3e8d425542c5a86c57b4dd9bea9a : SUCCESS in 2h 17m 09s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e68da0f7b18846bb82a2f63597a6f09d : SUCCESS in 51m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e267858e8784d0594d91099c4d88181 : SUCCESS in 1h 06m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e6abc31e36e46b0beeb5319c0623028 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"f34fbc892a1bfebf40008397c2569c08e3e364b3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-10 12:09:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f64a04692912e098adf7cd1c64cf6bed68c4cea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 14:43:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f0d0577447648cd8a4fd3c09e5a242a\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cd4b9fe010443fa8adb534bad5d231c : SUCCESS in 13m 59s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5595a4f60d5a44f3a70729b2d1a22510 : SUCCESS in 1h 26m 22s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1df480b0c3be419e8f025fc1dcce1171 : FAILURE in 1h 49m 38s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ca470ddfaf1746c2998903f69d5575c0 : SUCCESS in 1h 39m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa00e36b4110461cbbddb6e43d4a2ced : SUCCESS in 1h 32m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/779d20879f374e62830e8c85bcf53acd : SUCCESS in 1h 38m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c23f8475094f4226bf2e2fcd16785b63 : SUCCESS in 1h 24m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47dd241539c74491b1645b0eb800b276 : SUCCESS in 1h 44m 31s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e1997b93b9a54cb98852beadb68d7360 : DISK_FULL in 2h 27m 27s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53ad914f5e54488393e1d091c4b097ba : SUCCESS in 2h 25m 29s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee188864928d46fbbba4a79bfcac023c : SUCCESS in 2h 23m 15s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0671d74e4c2c4d1e946cc84f3949d8b7 : SUCCESS in 1h 08m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5baa5d06344f4da3afb2cf5254197c3b : SUCCESS in 55m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23cd8763a1a94b2daf199a5a94f5c993 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a0d72e82b93eeb2f441da3c50187692c7b136e00","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-10 16:54:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b10f36b58aa70cc47cbda1a52f89e33a1caf5522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 19:28:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/988fe02b508e4b6cb9b1583ac8fe29d3\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1973804090574534b44951205d5acffb : SUCCESS in 10m 55s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/132924e602a14c51804ecfd0b2bcbf2a : SUCCESS in 1h 36m 30s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e583dc3d432e4dd9966c5bd828cbb9a9 : FAILURE in 1h 54m 51s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a24af231366a4756995269246f1e5306 : SUCCESS in 1h 41m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39b4901afc084a1cbd322da811c2d4fb : SUCCESS in 1h 51m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbea076a6d774885bbf4bc704832d5e5 : SUCCESS in 1h 34m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dcce8c361d644f42aed1e942780b7fcd : SUCCESS in 1h 56m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0824c2968b92416d93a7b426512ec012 : SUCCESS in 1h 27m 19s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/22a91cdf09db4ab48c8d2bc92e1f84be : DISK_FULL in 2h 27m 34s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20330a3931db4748a3ad8a6ff6603c4e : SUCCESS in 1h 52m 06s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef1a948c37504ddb86392f5c549186f0 : SUCCESS in 2h 22m 01s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/73fdae77212542a39e75e56cacbb7238 : SUCCESS in 1h 08m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7011d041a324657a77bc81916d0494b : SUCCESS in 58m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75a0e7a73bcb4b9e8d99f545753bdd1f : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b0c140907afe2f94aba658c030d08ff13ee7863d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-13 11:26:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45e2ebc9de7872b7edd73e78e29ab02ebeb89be1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 14:21:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e1aeb77373b4c27b08a57669c80e7a6\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f79356667c5a458887d43f10c09c0aed : SUCCESS in 12m 16s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6e1e5a91059645cd98b56070918d39a6 : SUCCESS in 1h 36m 01s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b4ea28641e046969c6451894648543c : FAILURE in 1h 56m 26s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c40cfc9ab36a4f24b8a6a0ae504bbd06 : SUCCESS in 1h 46m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdda4b0ca35f4179bf92a6b6db97c71f : SUCCESS in 1h 40m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c817d3143ee34ca3bc041636c00754e3 : SUCCESS in 1h 32m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/47a40440540341fdaa6c229c6dcca6c0 : SUCCESS in 1h 46m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/820fd30682aa4ca5b361efbcd370f909 : SUCCESS in 1h 35m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b73488545fdd43e8840e8dae53699884 : DISK_FULL in 2h 46m 30s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b85e8749dde5436fae31ee467d690201 : SUCCESS in 1h 50m 27s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a7e6c15d663451789c9536f9225d766 : SUCCESS in 2h 02m 18s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aab6c58d11454d9ca5584cf740163718 : SUCCESS in 1h 05m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a561cd981de74904854b619de60b8a7d : SUCCESS in 1h 08m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/355ac2e1a7234882ad266a1cb5f375f2 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d4b8dda3b112d12fb7cd76dd5b860b5864867e64","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-14 15:09:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66fcaffe32cc2fa6c378e1f33f502f106191ccac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 17:51:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe6ff78aa26b44eaa3fa244e7ad8529e\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65e0e676796c467bbf6bad1c7e575fd6 : SUCCESS in 16m 30s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/15432362912943bbbb571e82ff2229ad : SUCCESS in 1h 38m 22s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ea64e7a092a489bba1beed675d530d6 : FAILURE in 2h 21m 25s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/426b6ad33b5f4fd0b8a442ed243a7e44 : SUCCESS in 1h 43m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6e44f386bdf448c827eba9610df4b8b : SUCCESS in 1h 42m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1dcb20c186c41faa783c011f96c8f83 : SUCCESS in 1h 26m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ba830d6e6d764bde800675d7be3dccb9 : SUCCESS in 1h 58m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/793f796105604593b40cb62f13062cc4 : SUCCESS in 1h 49m 20s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5ef71ca9ebbe498fb8fc2e32c77855f2 : SUCCESS in 2h 35m 42s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65a27f59c12845f394b075869914e4d0 : SUCCESS in 2h 20m 56s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/486baec35d4f4af6b1fb8b341241e2ed : SUCCESS in 2h 03m 24s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1832812412674eee993665048f5c3322 : SUCCESS in 1h 09m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bde40c5c0a544a2b9af57e53daee8c49 : SUCCESS in 1h 09m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c760352cc954d5ea10fcbaa60f39fa3 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"263eab01bcc11a2713efb8d1eb42b88512e7e3b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 17:53:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"541f4ca94e37758c1c219c56063461f77225c579","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 19:45:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b4147673f72478696a1ae9a9a5ea99a\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d818e9791cbc40fe9d54c350ccc7f71e : SUCCESS in 17m 37s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/18a02de9923745cd856ce68354143ff2 : SUCCESS in 1h 31m 19s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/90b2acd528664caf99c5bccec656ad11 : SUCCESS in 1h 38m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b605e34c41da4afe94772c1990a53ca9 : SUCCESS in 1h 31m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/842ecb47d6174d2f900628351a9be240 : SUCCESS in 1h 47m 15s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54e502fde6524402b4a0dfc79b0ca295 : SUCCESS in 1h 30m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9a31680860146beb097665e3320b550 : SUCCESS in 5m 52s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/11cffb5847614a81a5c4c68eb415448e : FAILURE in 1h 49m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ddae326e8899649e594c2c92e313712227e7499e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 19:45:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e867fa69f2bb9965e7a8c7e7e136b82bb77facd8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 19:46:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a12362bc8e447b59e4abc352ff4e4a9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b75f9ade7e640cfa67df88418251847 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"978a643dfd8f698baf43a5ba7a58adade4e62e3e","revisions":{"978a643dfd8f698baf43a5ba7a58adade4e62e3e":{"kind":"REWORK","_number":1,"created":"2023-10-23 08:01:40.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/45/899045/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/45/899045/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/45/899045/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/45/899045/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/45/899045/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/45/899045/1"}}},"commit":{"parents":[{"commit":"bf442c3993d907e221b4a38ced6e84bb2350822f","subject":"Use proper galera port in configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/bf442c3993d907e221b4a38ced6e84bb2350822f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-04 10:56:41.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-10-23 08:01:40.000000000","tz":0},"subject":"Add security rule for octavia healthmanager","message":"Add security rule for octavia healthmanager\n\nAmphora does report back it\u0027s status to Octavia healthmanager through\noctavia_health_manager_port. This outgoing traffic from Amphora must be\nallowed to show LB stats and operational_status.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/896017\nChange-Id: Ib6b8547b69949f7af0ba0f7f436b4286d3baccb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/978a643dfd8f698baf43a5ba7a58adade4e62e3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/978a643dfd8f698baf43a5ba7a58adade4e62e3e"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
