)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~920451","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I6ecffcaa8abea6c5fcf1822c4b3ac58c1e7a80dc","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"osa/messaging_improvements","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-28 10:36:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6ecffcaa8abea6c5fcf1822c4b3ac58c1e7a80dc","subject":"Define unique hostname for QManager","status":"MERGED","created":"2024-05-25 13:31:57.000000000","updated":"2024-05-28 10:37:13.000000000","submitted":"2024-05-28 10:36:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920451-osa/messaging_improvements","meta_rev_id":"40515cc54b296ea7cf305bca0efbd447c3a3cbf8","_number":920451,"virtual_id_number":920451,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-28 10:36:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2024-05-28 07:33:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-26 21:06:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2024-05-28 07:33:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-25 16:26:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-26 21:06:03.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-05-28 07:33:48.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"1e5553fc7468abfa563385a2eb3f454fb747be36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-25 13:31:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ae15f87ad0341cd4f10b1aa0e507533816b603f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-25 16:26:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d82f29c8725e41d2ba3111aed3ca3a89\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/197cdbf9e9be4662921e3c06672c84cb : SUCCESS in 18m 24s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/52b77ceabfea4f8b97be904c46fc078e : SUCCESS in 1h 40m 13s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/71e00bcdb55b40209fcfb5c3e56ad6f2 : SUCCESS in 1h 46m 12s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/50790374bb24434697a5ab3fde1ec7d9 : SUCCESS in 1h 43m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/da2ae5b49df04cd3bab762fd7f90b7be : SUCCESS in 1h 43m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5facc930eaf94416b27376408525e412 : SUCCESS in 1h 42m 33s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a4b6a5173e3344af8da82018cbf0831e : SUCCESS in 1h 49m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f482577ea3b4aa3a12ef192025503e7 : SUCCESS in 1h 21m 53s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b6fcde9539341e5b0699ed644452cba : SUCCESS in 1h 34m 51s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d01fbd5421164a9490e0d1fbfcc08a5f : SUCCESS in 2h 21m 06s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f7b0871478b34d7c8b9cb2d84ae61656 : SUCCESS in 2h 21m 58s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63a60b99637d47a4a81bd30c208cbe45 : SUCCESS in 2h 09m 26s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83c71eaef6b641b18a06ac917f62d7a6 : SUCCESS in 2h 33m 40s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/650899e09852424c915a4b1ab1656461 : SUCCESS in 2h 22m 35s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0667343c09d4960bb3e241ef99b2294 : SUCCESS in 1h 54m 17s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1f31ee0b6fa9474d91de8408d65fbf28 : FAILURE in 51m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75e171e8d138412db448d81ca80bbb77 : FAILURE in 1h 09m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50736b3782d543a9a48b4b4e4ba46d8b : SUCCESS in 4m 09s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da4b8a388c1849faa1b9cfc1e5e447e7 : SUCCESS in 2h 09m 57s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2ab0c7f5aaa94dd68e4656e446dcd27c : SUCCESS in 2h 47m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c90c0f60c123deecac315d4d2785961a0e712d23","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-26 21:06:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8be495cb3ce1bc35a8207f7cd235ddb86fff0332","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2024-05-28 07:33:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"24c47d0ec6740c494b2fb8b699bd72ce2a776756","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 07:35:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"364ba41ccad1508c86847f020bad575427c4da55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 10:36:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11f6220cd44141dab9ec45173d9343d0\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa2da240537b4f1f8f1ccb5413a675e0 : SUCCESS in 13m 02s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/caeedea5942e493ba05423b538aebcaa : SUCCESS in 1h 38m 58s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8e42918574a24b0a83ad70fd8196bb8f : SUCCESS in 1h 37m 17s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/857e88e8233d43cc9bb216b1e064bb34 : SUCCESS in 1h 43m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe778ef9fcb9446f89443584c8a05843 : SUCCESS in 1h 36m 56s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da3178a0397445ce8b02576e7f7ab0b1 : SUCCESS in 1h 33m 02s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/64d04f671d924dde8e4dd3859bd1477e : SUCCESS in 1h 35m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/facb235d133545c4bc157b72baff6791 : SUCCESS in 1h 27m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89918baa5a9a4e97a15ab120afa40f9d : SUCCESS in 4m 42s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9b73fbc281f4a68b81042a3f12495f3 : SUCCESS in 2h 05m 02s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aa28c35573f24f948243e143fee2e4b2 : SUCCESS in 2h 31m 04s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/fefbb3f475154ced8747d042db96ddd1 : SUCCESS in 1h 39m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"12c1356b82933db5ba5e0eb43d9b8de76440fe68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 10:36:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"40515cc54b296ea7cf305bca0efbd447c3a3cbf8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 10:37:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff04a828f94e4ac2b337cf0e45812087\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d937167788e9472694f0d9527dcb15f1 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb3a5431bbb4b66fce23148bd16ce44a79c6326e","revisions":{"eb3a5431bbb4b66fce23148bd16ce44a79c6326e":{"kind":"REWORK","_number":1,"created":"2024-05-25 13:31:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/51/920451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/51/920451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/920451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/920451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/920451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/920451/1"}}},"commit":{"parents":[{"commit":"de2740275b7687f12d39ce671c06b276726fd7a3","subject":"Merge \"Implement support for octavia-ovn-provider driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/de2740275b7687f12d39ce671c06b276726fd7a3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-25 13:31:50.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-25 13:31:50.000000000","tz":120},"subject":"Define unique hostname for QManager","message":"Define unique hostname for QManager\n\nDue to the shortcoming of QManager implementation [1], in case of uWSGI\nusage on metal hosts, the flow ends up with having the same\nhostname/processname set, making services to fight over same file\nunder SHM.\n\nIn order to avoid this, we prepend the hostname with a service_name.\nWe can not change processname instead, since it will lead to the fight\nbetween different processes of the same service.\n\n[1] https://bugs.launchpad.net/oslo.messaging/+bug/2065922\n\nChange-Id: I6ecffcaa8abea6c5fcf1822c4b3ac58c1e7a80dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/eb3a5431bbb4b66fce23148bd16ce44a79c6326e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/eb3a5431bbb4b66fce23148bd16ce44a79c6326e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
