)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~927353","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~Ia0388171b784615924f5a5dafb5e160e942123db","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"add-additional-security-group-rules-support","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-08-29 10:53:46.000000000","reason":"Change was submitted"},"37261":{"account":{"_account_id":37261,"name":"Yuki Hirano","email":"yuuki0xff@gmail.com","username":"yuuki0xff"},"last_update":"2024-08-29 06:18:26.000000000","reason":"\u003cGERRIT_ACCOUNT_37261\u003e replied on the change","reason_account":{"_account_id":37261,"name":"Yuki Hirano","email":"yuuki0xff@gmail.com","username":"yuuki0xff"}}},"hashtags":[],"change_id":"Ia0388171b784615924f5a5dafb5e160e942123db","subject":"Add additional security group rules support","status":"MERGED","created":"2024-08-28 07:44:36.000000000","updated":"2024-08-29 10:54:44.000000000","submitted":"2024-08-29 10:53:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927353-add-additional-security-group-rules-support","meta_rev_id":"ee0f48342391f3530a2f3d2a5cb19b26862e4b18","_number":927353,"virtual_id_number":927353,"owner":{"_account_id":37261,"name":"Yuki Hirano","email":"yuuki0xff@gmail.com","username":"yuuki0xff"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-29 10:53:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-08-28 14:27:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2024-08-28 09:11:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2024-08-28 14:27:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-28 07:51:11.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2024-08-28 07:52:19.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2024-08-28 12:06:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-28 14:27:16.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"c7be82f3a3cdfe0cbecf432820bc016e474bc6df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37261,"name":"Yuki Hirano","email":"yuuki0xff@gmail.com","username":"yuuki0xff"},"date":"2024-08-28 07:44:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdc17ea3574967e8bda7a81152c862f12bc29e41","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-08-28 07:51:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31db66ce6bca3eb2250bbac2d87e6bf0e490340f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-08-28 07:52:19.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"877c7800910136d01f64856ac35575950a93a967","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-08-28 08:32:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0299686307de66cb7be84fc004a605d8f8d9f94c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37261,"name":"Yuki Hirano","email":"yuuki0xff@gmail.com","username":"yuuki0xff"},"date":"2024-08-28 09:10:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"863070ed69184972ee319d384c813fdda47f4788","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-08-28 09:11:30.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"185ae510c839f3b39b46a390c59ca8a2fc03b889","author":{"_account_id":37261,"name":"Yuki Hirano","email":"yuuki0xff@gmail.com","username":"yuuki0xff"},"date":"2024-08-28 09:13:51.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"16b293d1a603d7c6c9c6271bead83a6f923c3168","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 12:06:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33bae9f7ec124b1bb8e0ed25190ec1d0\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/644ad037f5c044e5be8225236cdacd77 : SUCCESS in 6m 26s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/add9b56504e2484ab79be6dc6e74678f : SUCCESS in 1h 39m 57s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f473df8457d4450c85e0f3a4b2c16bb8 : SUCCESS in 1h 26m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cc646e0933d040a191c0e090845f701a : SUCCESS in 1h 31m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0194fa96dea24d10bc55070634d3dd4e : SUCCESS in 1h 18m 20s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2237564023c5493d89d225a5c4de6fa1 : SUCCESS in 1h 44m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f06f5ad78ed2414a880a815c18b62f6a : SUCCESS in 1h 40m 28s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba9da75f62b1468c8cb2215a12b1bc81 : SUCCESS in 1h 24m 59s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/526b568167604a2da1c043e0f3895fac : SUCCESS in 1h 59m 50s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/710835a8d1384dc696535ac64304305f : SUCCESS in 1h 57m 04s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/96a4f14a05e5405dbb1064deddb0a5aa : SUCCESS in 2h 29m 37s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9295edae4296482397a0dde0d5650936 : SUCCESS in 2h 01m 34s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/671561d331a943a2ae7e34a85ca85d78 : SUCCESS in 1h 15m 57s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90daeec5e67b4150bee70096058ee381 : FAILURE in 1h 02m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a698903c7fd45d98c93663f3b69c0bb : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06d19651c83a4c9aa2d2705f7887aa7e : SUCCESS in 3m 31s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39ad23f4238a46a3b834ad92cfd86a36 : SUCCESS in 2h 14m 16s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a752a22022534205a90a33d1fa182316 : SUCCESS in 2h 27m 22s","accounts_in_message":[],"_revision_number":2},{"id":"501240c5d1527162a886f804450c0f6428066280","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-08-28 14:27:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b2421cd4726ce5b8776a7b49cd91062cdb23323c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 14:29:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e3a71292900a5ae8bb57ee8f4eaa956b6c255bf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 17:04:36.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f695ca0341d94b729c72510ed6ba2339\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07b149fd86084c56a1103bf6bf8378a8 : SUCCESS in 6m 41s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8c6a2b4d1fcb407e8203dec3e3d2fc94 : SUCCESS in 1h 25m 53s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0c03a9d3aaa14af4a2e09713a80e31bf : SUCCESS in 1h 40m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/578d161b22544f56a26c4394a4d0c50d : FAILURE in 1h 31m 57s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d56dca70cbdb45de814acc6ff617bbf7 : SUCCESS in 1h 35m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/443cd1b856244a38af648beffec99c5f : SUCCESS in 1h 27m 42s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff0c837b035f4942b0194dbd48d5a6a6 : SUCCESS in 1h 48m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79100866410d46d5afc0df90fa2ba98c : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9944ab952a924f9db7a425446bcffb43 : SUCCESS in 2m 56s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5253bf25ded4dfc8f609eb8c973105b : SUCCESS in 1h 49m 35s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/81bfdaaf9a4c4231a90379f93f88809c : SUCCESS in 2h 20m 22s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/daf17ccba71b4d9488de52870dee4a98 : SUCCESS in 1h 44m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"703c7e4fca99ba172426bcab175f97bd3b1366ae","author":{"_account_id":37261,"name":"Yuki Hirano","email":"yuuki0xff@gmail.com","username":"yuuki0xff"},"date":"2024-08-29 06:18:26.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6932631a22c4b8c3002843f9b9ecf6d2948190af","author":{"_account_id":37261,"name":"Yuki Hirano","email":"yuuki0xff@gmail.com","username":"yuuki0xff"},"date":"2024-08-29 06:20:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6ac674989d9bb4d0bb256a33f8fd2a8350d7e65a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 08:45:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c13c4cb88db4a03ad22b4a3d80bc618\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fa5fe7026e14a7d90f95df9f394904e : SUCCESS in 6m 33s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/59b94407ca234717adec5d2573c7d174 : SUCCESS in 1h 29m 01s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3c2fc4b04484306887d5d027d8f19ae : SUCCESS in 1h 34m 55s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3cfbc9049b7d4c7a948f5b2bb866a662 : SUCCESS in 1h 27m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11e7836667bd4b39b883c85b2b958d8e : SUCCESS in 1h 34m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5544f92c30264483b9c25e5935f43dc7 : SUCCESS in 1h 37m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e5df436aa49422692fe0df6096edd47 : SUCCESS in 1h 26m 38s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b78d91813294fbf94c1b7cc7f008e60 : SUCCESS in 1h 20m 31s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a74cebf54146414ba446a4f1fc10b3d0 : SUCCESS in 2h 02m 51s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c4a5c969e474822b1aaf813d894ea07 : SUCCESS in 2h 14m 25s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6f899f741c614ec082745b9920cf7924 : SUCCESS in 2h 16m 35s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/010082a9ed2041d1b4c0d9ab402ef5b0 : SUCCESS in 2h 10m 25s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/171fd4fb9f484bab9a5da6fbfd7bfadd : SUCCESS in 1h 17m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53aa6e58940d4b408cd42560c5f543b9 : FAILURE in 1h 12m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f02bd055c274d98a7aa189653f08643 : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f778e61a95d4145b017d756f8a459af : SUCCESS in 2m 45s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e5db8b75de64c5ea55039be80ea93e3 : SUCCESS in 2h 08m 45s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/349feb5cc1a243de8da8855b712ca289 : SUCCESS in 2h 10m 46s","accounts_in_message":[],"_revision_number":2},{"id":"d4c1ea3a89ffd502148982af1db836fbbf3acb7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 08:47:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4356bf3968a7e67cbcf9e62da373a062b72c99c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 10:53:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c0d9892aebf44d7ac870bec15dabbea\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0aff9bbfb78444da955598b2be37c418 : SUCCESS in 7m 46s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/fa56c860dd354ce4b777eed218a1afae : SUCCESS in 1h 21m 41s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cd15e56dbc2744b7b86518408c0ed9bb : SUCCESS in 1h 26m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7d25116048849f2ae6f5de1c6eb2bbb : SUCCESS in 1h 20m 53s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b94893dc215449a4b850c5739c543222 : SUCCESS in 1h 29m 13s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/70e3388af1d54d0ab6472a653b61f55a : SUCCESS in 1h 31m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa3d78e663bf4e6fb436783f29814096 : SUCCESS in 1h 34m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11872d4a66ea49d09ea8ad40f87da003 : SUCCESS in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1dac96f45df84177ae6584f7dc9b27da : SUCCESS in 2m 37s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8aacfdbd39cb4bfba4b47d96bb72b2ba : SUCCESS in 2h 00m 21s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/56131ac75aeb4831bbde71182b5ef400 : SUCCESS in 2h 00m 05s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1cffd70f2324603801cd111426c7093 : SUCCESS in 1h 52m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b7c82cddbccc20fcad3a1e065ee26fdc2a191980","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 10:53:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ee0f48342391f3530a2f3d2a5cb19b26862e4b18","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 10:54:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/784ccea69290436b8f9aecd5f192f27c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3451250239e4636a03c1a3c6c9d3e0d : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1b745fea66793593cf08fe2865953ea4efa89776","revisions":{"442effa410e4c3a81dd7d0da936d2600a5f2a4ae":{"kind":"REWORK","_number":1,"created":"2024-08-28 07:44:36.000000000","uploader":{"_account_id":37261,"name":"Yuki Hirano","email":"yuuki0xff@gmail.com","username":"yuuki0xff"},"ref":"refs/changes/53/927353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/53/927353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/53/927353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/53/927353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/53/927353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/53/927353/1"}}},"commit":{"parents":[{"commit":"9efc5efcaa46d838c94b57842caa56c5c35d9f37","subject":"Ensure Octavia communicates with Neutron through internal URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9efc5efcaa46d838c94b57842caa56c5c35d9f37"}]}],"author":{"name":"Hirano Yuki","email":"yuuki0xff@gmail.com","date":"2024-08-28 07:43:24.000000000","tz":540},"committer":{"name":"Hirano Yuki","email":"yuuki0xff@gmail.com","date":"2024-08-28 07:43:24.000000000","tz":540},"subject":"Add additional security group rules support","message":"Add additional security group rules support\n\nWhen customizing Amphora, like syslog forwarding support, we need to add\nsome rules to octavia_sec_grp. This change allows us to specify\nadditional rules using variables.\n\nChange-Id: Ia0388171b784615924f5a5dafb5e160e942123db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/442effa410e4c3a81dd7d0da936d2600a5f2a4ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/442effa410e4c3a81dd7d0da936d2600a5f2a4ae"}]},"branch":"refs/heads/master"},"1b745fea66793593cf08fe2865953ea4efa89776":{"kind":"REWORK","_number":2,"created":"2024-08-28 09:10:44.000000000","uploader":{"_account_id":37261,"name":"Yuki Hirano","email":"yuuki0xff@gmail.com","username":"yuuki0xff"},"ref":"refs/changes/53/927353/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/53/927353/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/53/927353/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/53/927353/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/53/927353/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/53/927353/2"}}},"commit":{"parents":[{"commit":"9efc5efcaa46d838c94b57842caa56c5c35d9f37","subject":"Ensure Octavia communicates with Neutron through internal URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9efc5efcaa46d838c94b57842caa56c5c35d9f37"}]}],"author":{"name":"Hirano Yuki","email":"yuuki0xff@gmail.com","date":"2024-08-28 07:43:24.000000000","tz":540},"committer":{"name":"Hirano Yuki","email":"yuuki0xff@gmail.com","date":"2024-08-28 08:59:33.000000000","tz":540},"subject":"Add additional security group rules support","message":"Add additional security group rules support\n\nWhen customizing Amphora, like syslog forwarding support, we need to add\nsome rules to octavia_sec_grp. This change allows us to specify\nadditional rules using variables.\n\nChange-Id: Ia0388171b784615924f5a5dafb5e160e942123db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1b745fea66793593cf08fe2865953ea4efa89776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1b745fea66793593cf08fe2865953ea4efa89776"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
