)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~930749","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~Ia35948cd2381b13d3792fc0029d93a9e176c884c","project":"openstack/openstack-ansible-os_octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-10-05 00:46:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia35948cd2381b13d3792fc0029d93a9e176c884c","subject":"Ensure quota is set for BFV amphoras","status":"MERGED","created":"2024-09-27 16:13:03.000000000","updated":"2024-10-05 00:47:02.000000000","submitted":"2024-10-05 00:46:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930749","meta_rev_id":"01b588e871b5d9290b5894f9d499e73569341076","_number":930749,"virtual_id_number":930749,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-05 00:46:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-10-04 09:22:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-04 22:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-04 22:31:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-27 18:49:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-04 09:22:18.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-10-04 22:31:44.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"bd79e1da59dfc9a65ef5e01d6744f068c4f9fabc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-27 16:13:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2440e8b5ea7237abbe949bec0e84025cca6e4905","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 18:49:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51cea222a7bf4eb4b04368f850ef4e87\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac8b0ef314e44770b1d07e04db76a09a : SUCCESS in 5m 09s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/662f4eec1cca41d3aea296386954f282 : SUCCESS in 1h 27m 11s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/30cadcba1b2e4900ae23a0e32f163297 : FAILURE in 1h 39m 27s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f8ccf27f35f3446ba3123f03c63667b7 : FAILURE in 1h 23m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf930891f3f14d0898e5bc4fe24b8585 : SUCCESS in 1h 37m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87e99a15c80a4cc5877b8ac1ed971bdc : SUCCESS in 1h 39m 29s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e80a2c6dc9eb464bb1eb327d2675afb8 : SUCCESS in 1h 41m 04s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/903473ffba8746c6b17a5c45f99c747d : FAILURE in 1h 13m 34s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f69946e41964fafb66b46c1986d1868 : SUCCESS in 1h 24m 08s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/737f180884fe4503a7421a99a6855eae : SUCCESS in 1h 36m 51s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/962fa3055a3a47bf9658c17e22432701 : SUCCESS in 1h 34m 11s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8fb346d3ab104741815bd76ab90037fe : SUCCESS in 2h 16m 57s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28eeda190978454c97f4b1afe113def2 : SUCCESS in 2h 07m 01s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/15d8cb279530489b90c4e2ae4a71e560 : SUCCESS in 2h 27m 59s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e72b9c42fe5435591941bb5d4752417 : SUCCESS in 2h 23m 29s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6e1f535c1dfd4bae9ec91943d2c56bb4 : SUCCESS in 1h 05m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e6915d1493d4be5b784be53fbdd662b : FAILURE in 1h 06m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d1b47f9011143e7a009bae6273ab2b4 : SUCCESS in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b2dd243866a4d95b9122db4f6502019 : SUCCESS in 2m 28s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0493c302abae4769a395765fa2dbdcd9 : SUCCESS in 1h 15m 42s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e75d39eb24484a42b6cf922e9a8759b8 : SUCCESS in 2h 14m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8789f389a3d7b78af9f08b49afce462d5f57adf3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-30 07:23:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8343d7bbd3cbafebc334853901909b5314f7cc97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 10:03:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/360e08288c9d4616bc693977111b935c\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2faec6f3928404cab229548ab0f309b : SUCCESS in 6m 31s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/bc2083bc2e404c00be0fb1d94335b7df : SUCCESS in 1h 35m 56s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e0d5554b8bf47b7964d35bc5bd07d6f : SUCCESS in 1h 23m 14s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9490673c1244424997903c1cccb01d89 : SUCCESS in 1h 31m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6537b4de0fd64d4ea4023347c7149664 : SUCCESS in 1h 21m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc229aef195a4bb5aa45a626db82da86 : SUCCESS in 56m 23s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b7aec5c704f043cd8a0ef7f878115d1b : SUCCESS in 1h 41m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf78fc0d8003468c98cf1d1b367d8001 : SUCCESS in 1h 23m 16s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8d41e609ee542ae9addc7cfb60d2ea9 : SUCCESS in 1h 28m 33s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f82519917844d5d9d37d9f67ffc46eb : SUCCESS in 56m 42s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e9e270007eb4643b3d8c78d64afc02d : SUCCESS in 1h 37m 40s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f368f5d9c82945f39d38e12cb496f876 : SUCCESS in 2h 03m 34s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64f45d5b98ec4689a552a90ac829eb9d : SUCCESS in 1h 54m 56s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/25beee363b81411098213b12b5974239 : SUCCESS in 2h 29m 18s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/503a848b3ac14890ae24f24e2e481499 : SUCCESS in 2h 00m 30s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c26ba60bd8154191b91ce1b6f78553d0 : SUCCESS in 1h 17m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aaa07ea2ae2740e5ae372a2828428405 : FAILURE in 40m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02cef3bced664d3ba55300ea501faddb : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5b20b8e31b245ed89e9ee2f44e6872a : SUCCESS in 2m 22s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eff66bd3b82548549408b2bb55d3c13c : SUCCESS in 1h 51m 08s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1c5e60f54fab4014a95949659d5e6cab : SUCCESS in 2h 20m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b5789c99fbeba85707d22ea5b1013001a9380a76","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-10-04 09:22:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2200029cd077765f9073d94279bf7bb03230f5ac","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-10-04 22:31:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a7c00c8e23263ee40c1bb11dc18081f4dacef7b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 22:33:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8892a9900a8d6e5b973115631e2779d01c2a61cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 00:46:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f60894f669b472fbd59bb6cd691f76a\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/073fbc5eef014d55ba346c16bc4152c1 : SUCCESS in 6m 31s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ef819dabefcb45568092b1ca02660c5e : SUCCESS in 1h 45m 21s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a3bcde4295f444b4aaa0a637b771d35c : SUCCESS in 1h 34m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65b1b9e39a4a4fa09d6f6049f6009b5a : SUCCESS in 1h 28m 55s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68d7333cdb214e0bbd849a15a29ede40 : SUCCESS in 1h 39m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16d1ce70f9a249b4b949bb21daac6187 : SUCCESS in 2h 00m 36s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d04631c469b04b70b24b505606874668 : SUCCESS in 1h 23m 07s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9720a985cbf944c8863e8dd214459416 : SUCCESS in 1h 35m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0c3d4d8417242789e06624aaccf6e06 : SUCCESS in 54m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/540e8580f5aa42b5b4fbf48e810146df : SUCCESS in 1h 23m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/016941f9f6a14143a07e005c16530066 : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1478182c53504b4cb988587cfb0905f2 : SUCCESS in 2m 23s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de73f890e8c8445ebd109296ce19cb04 : SUCCESS in 2h 06m 42s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f0d234b53c7e4e5cb794adda023b68d5 : SUCCESS in 2h 01m 43s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c37aef935ba4bb7b4d26acc8763e379 : SUCCESS in 1h 11m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99092f294691368bcdc988476f0fb10702c64b06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 00:46:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"01b588e871b5d9290b5894f9d499e73569341076","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 00:47:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f09ac9d19d3f45e7acca4932656dd282\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eb1f220eeda434d99619b88c1d5a6d0 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b6f39ffbddd66d86e9cb125c35dcff95201022f","revisions":{"3b6f39ffbddd66d86e9cb125c35dcff95201022f":{"kind":"REWORK","_number":1,"created":"2024-09-27 16:13:03.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/930749/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/49/930749/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/930749/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/930749/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/930749/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/930749/1"}}},"commit":{"parents":[{"commit":"c091f3d2dd907c49a2c8765b90e3aac09525d341","subject":"Merge \"Provide better flexability for SSH keypair options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/c091f3d2dd907c49a2c8765b90e3aac09525d341"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-27 16:12:55.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-27 16:12:55.000000000","tz":120},"subject":"Ensure quota is set for BFV amphoras","message":"Ensure quota is set for BFV amphoras\n\nAt the moment role does not increase volumes quota for the project\nwhen BFV is enabled for Octavia. This might lead to Amphora\nfailures to spawn-up once default quota (of 10 volumes) is reached.\n\nWith that we also calculate amout of diskspace required based on the\ninstance limit.\n\nIt also changes flavor to contain 0 disk if BFV is enabled to ensure\nthat ephemeral is not gonna be used.\n\nChange-Id: Ia35948cd2381b13d3792fc0029d93a9e176c884c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3b6f39ffbddd66d86e9cb125c35dcff95201022f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3b6f39ffbddd66d86e9cb125c35dcff95201022f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
