)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~942582","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I728b1fdbd8c61853cec564203f8e58d66ea09374","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"osa/remove-cinder-v2-crud","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2025-04-02 12:01:44.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2025-04-02 12:01:44.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-04-02 10:58:13.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I728b1fdbd8c61853cec564203f8e58d66ea09374","subject":"Align on cinder service naming","status":"MERGED","created":"2025-02-24 12:59:25.000000000","updated":"2025-04-02 12:02:50.000000000","submitted":"2025-04-02 12:01:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942582-osa/remove-cinder-v2-crud","meta_rev_id":"232839d66c5fee08ca3850589e0cf15d84b49668","_number":942582,"virtual_id_number":942582,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-02 12:01:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-26 11:17:43.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-03-26 10:43:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-04-02 09:25:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-24 15:15:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-26 10:43:21.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-03-26 11:17:43.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-03-26 11:17:43.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2025-04-02 10:58:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"57e078ceebbc895f79fdec857156e8894859ee3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-24 12:59:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6e26cbd60fc81f435256ad6d806699c020f1d6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 15:15:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/931aa89e069d489597981dfb87254d14\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/404edb84393641cfb4afe7352dff6761 : SUCCESS in 6m 10s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/890ca10ab9a9464fab61d111faa12844 : SUCCESS in 1h 43m 58s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/cef091216ab24582aff4ebbae5957e63 : SUCCESS in 1h 41m 16s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/11e5e036f2b34bf28267403fa625cdc3 : SUCCESS in 1h 18m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b598b859d70046b7aa6fdfc23a12cdc3 : SUCCESS in 1h 26m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3ce047d2bd44087a5d82259c4cabf21 : SUCCESS in 1h 44m 59s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9fe3c2e477434d21853442835c308219 : SUCCESS in 1h 40m 46s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddfca6d537c7487c953b8d80fcf4b445 : SUCCESS in 1h 34m 59s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2561a218d4e74e8db6043f0b0033f0eb : SUCCESS in 1h 46m 37s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/20d8dfd414d7479789ad7f633dfb6cd0 : FAILURE in 1h 09m 08s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cf0b4548f9134af5854630ea49cca437 : FAILURE in 1h 18m 19s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0dbaf761621489f9db5b2766d674a62 : FAILURE in 1h 06m 22s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5f73aa878c246a9a2cf17a15677945b : FAILURE in 1h 14m 24s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6fa44c26e69d47d1aca71ba2f84c3388 : FAILURE in 1h 13m 30s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d77095a1a138403eb5da42334d629456 : FAILURE in 1h 07m 21s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ac09cbca934b4e4dadb498b67bf47310 : SUCCESS in 1h 17m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b3ed4583c5a45cf810dd57726985916 : SUCCESS in 46m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97827a55c6cd40809febcb1a17b3d374 : SUCCESS in 5m 01s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38f5851479ab454f90ee9625bedcf080 : SUCCESS in 2h 13m 56s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/31490f8dfb3944f5af7c62b42e321b59 : SUCCESS in 1h 15m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b0c297e6265818fb77f0b04fd6e7babc1f6fd618","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-03-25 15:47:08.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"14cabd68cd1cc4df6763e8ba628363e69924dcff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 18:55:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bd739d0dc5240a783914a7c36d6842b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cec90bf6912548aeadc3c3f3f532c3ea : SUCCESS in 5m 20s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/06d850caa882437da39ade3bdad0ef5d : SUCCESS in 1h 37m 07s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3118e9f95ce14168856c1fc611d7c9ad : SUCCESS in 2h 01m 41s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7932f6adb8274ad0be7893dbc56e98ea : SUCCESS in 1h 42m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec018c71854e4abaaf7fbea11a88c917 : SUCCESS in 1h 51m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/949bf9a18af44d6690f3a8308ed3d03b : SUCCESS in 1h 46m 19s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2cfe734d277e45bfa7e4d3f3089dfce7 : SUCCESS in 1h 36m 13s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68603b86398240f5be0034d822dd0100 : SUCCESS in 1h 45m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/054b1a95f68d4701b04524a202f76b3a : SUCCESS in 1h 41m 09s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4bcd0760a95a467f97efe79574684c12 : SUCCESS in 2h 24m 21s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ac1f1588040a4ecd83e58c7d0d39cd07 : SUCCESS in 2h 38m 48s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9abe0541563449fda83a4d2c78e485f8 : SUCCESS in 2h 30m 33s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a5394c2eb4f4c9396b10299b41769e4 : SUCCESS in 2h 42m 26s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1b488942b27c49978653e9ba6f78cf0e : SUCCESS in 2h 28m 12s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afc3d726bfbe40e887767fb6967e14d1 : SUCCESS in 2h 59m 03s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a98e5d8c6c1149178b0a3efba9632fec : SUCCESS in 1h 15m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/537d8dc797cf435caa80bce06e4a0bf0 : SUCCESS in 1h 20m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03ce3ee3bbaf407296928b0d4b1cf715 : SUCCESS in 7m 04s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b52c20127af54428935bb3b0a10568a9 : SUCCESS in 2h 20m 46s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c35cfad4eb0d43d788dd9a1c6bc209d2 : SUCCESS in 2h 40m 22s","accounts_in_message":[],"_revision_number":2},{"id":"da6a1ab13a3076afc622e13deebfe52937f9a8cf","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-03-26 10:43:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"478994e048429bb80e37f149fb72e98b3c1a74bc","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-26 11:17:43.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"30a3462a067e6516316a98721e1ad03aa540c63d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-04-02 09:25:44.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e1378c23f08cc13f6e4a5cfce6eece3a094b92b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 09:28:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d7887b45d84ab90a83184f6f80ec03bd0b0eba18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 12:01:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da0c107b16e84b06946dfdab29e3e4d4\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54ccc4ad5d9348daa4c5c6f067a3cb67 : SUCCESS in 4m 39s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7c675e8f682d425ab2bb20d47a443d43 : SUCCESS in 49m 10s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1418db8769624a70810787725b8c1d89 : SUCCESS in 53m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/438e453ca0e84500b87fb05a957d1e8d : SUCCESS in 47m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99f56248076a422ca2aceaacc082ba28 : SUCCESS in 50m 23s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4d9b5da5fd254d2c89c9a4b8ee5e2f70 : SUCCESS in 51m 49s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1fd33a7de5145b0a3e79b3d34ba782d : SUCCESS in 1h 30m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc57993db101480891553a85906a7689 : SUCCESS in 1h 36m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73aa67d5ab4941058af45ef98d943ae2 : SUCCESS in 5m 18s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0276dbd72af47588df11a63cc40f944 : SUCCESS in 1h 46m 08s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0735d07a0a2f494ca13eb2326f665371 : SUCCESS in 2h 22m 51s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/754b2a8328524792885915b6d76a9bb8 : SUCCESS in 1h 29m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9beba33a27e9407a8d6c576669f431d5f34a271d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 12:01:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"232839d66c5fee08ca3850589e0cf15d84b49668","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 12:02:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20ab002ef24d44668b06c80cfc8fd616\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee6b67555f0941ed8b996a1b72398575 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"88e7eb07fe97e617c8e2107e048e870b732e3525","revisions":{"c77aea42c52de7347f8578e45960263ab980957c":{"kind":"REWORK","_number":1,"created":"2025-02-24 12:59:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/82/942582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/82/942582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/82/942582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/82/942582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/82/942582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/82/942582/1"}}},"commit":{"parents":[{"commit":"4dfa9d782b47092c3078717cba1567a3b8e0e407","subject":"Merge \"Switch from focal to jammy based amphora image for CI testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4dfa9d782b47092c3078717cba1567a3b8e0e407"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-24 12:59:17.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-24 12:59:17.000000000","tz":60},"subject":"Align on cinder service naming","message":"Align on cinder service naming\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_cinder/+/942581\nChange-Id: I728b1fdbd8c61853cec564203f8e58d66ea09374\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/c77aea42c52de7347f8578e45960263ab980957c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/c77aea42c52de7347f8578e45960263ab980957c"}]},"branch":"refs/heads/master"},"88e7eb07fe97e617c8e2107e048e870b732e3525":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-03-25 15:47:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/82/942582/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/82/942582/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/82/942582/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/82/942582/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/82/942582/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/82/942582/2"}}},"commit":{"parents":[{"commit":"4dfa9d782b47092c3078717cba1567a3b8e0e407","subject":"Merge \"Switch from focal to jammy based amphora image for CI testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4dfa9d782b47092c3078717cba1567a3b8e0e407"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-24 12:59:17.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-03-25 15:47:07.000000000","tz":0},"subject":"Align on cinder service naming","message":"Align on cinder service naming\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_cinder/+/942581\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/942775\nChange-Id: I728b1fdbd8c61853cec564203f8e58d66ea09374\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/88e7eb07fe97e617c8e2107e048e870b732e3525"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/88e7eb07fe97e617c8e2107e048e870b732e3525"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
