)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~949419","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I107f4716d1ad165a0b3ae2cb2ee48d22c03d9aab","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"osa_hashi_vault","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-04-09 20:09:26.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-04-20 23:40:00.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-04-20 09:22:35.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I107f4716d1ad165a0b3ae2cb2ee48d22c03d9aab","subject":"Add hashi_vault pki backend support","status":"MERGED","created":"2025-05-11 20:28:46.000000000","updated":"2026-04-20 23:41:45.000000000","submitted":"2026-04-20 23:40:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949419-osa_hashi_vault","meta_rev_id":"7bcb4b8abcde0ef8a052b7103833145128c622df","_number":949419,"virtual_id_number":949419,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-20 23:40:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-04-17 11:41:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-17 11:41:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-04-17 11:41:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2026-04-20 09:22:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"CC":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-11 22:28:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-10 17:47:12.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-03-16 10:34:41.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-03-16 18:48:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-03-26 11:44:47.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"},{"updated":"2026-04-09 11:45:31.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2026-04-09 20:09:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"9105a23f13e431a40b607de75d6bed16a0222fec","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-05-11 20:28:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47321c0fafc14695e7719eaa466f2f9cd96bfb74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-11 22:28:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8ea508dc3d9467294506518ebeac295\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1af49b25bab441ab34b2a2bf0374546 : SUCCESS in 8m 42s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6083ca8424374a11acf07e5de4f4d812 : FAILURE in 1h 06m 15s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a32875d9098845d1be6673c2feb3ce76 : FAILURE in 58m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/50c5af22949049f78c68df9088f7d493 : FAILURE in 1h 07m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8a994f53891422497833582ac611c73 : FAILURE in 58m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ae1129cdd1e483bb94e4b5b0e453b5e : FAILURE in 1h 04m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/495f19d19e6d44dd9db271c2f259ba2e : FAILURE in 1h 01m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cf83ed1a58047fea976fa432f3dc74f : FAILURE in 54m 28s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15408fdae2bd4af980e860fda9c211f1 : FAILURE in 1h 10m 25s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/13d22238569b4e3bb88beec6a9debc53 : FAILURE in 1h 47m 58s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4e7d281f0dc54900935bfc4f2f1544ff : FAILURE in 1h 51m 27s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e314bd400ec4163bb202449575221f0 : FAILURE in 1h 49m 27s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc89085d0ed5439ca9f9eb89ff0e3305 : FAILURE in 1h 55m 03s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/41662cb8fc9045c580cc4c94ac24d40a : FAILURE in 1h 50m 18s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40048d938eb246c9a9980d7a65110844 : FAILURE in 1h 53m 41s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0c1d6d15e42c4e118606eeb9d1ce389c : FAILURE in 47m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9d05b1d1a49446e96e298f865feffc1 : FAILURE in 49m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00390913070c4bd58bc1d43f97253716 : SUCCESS in 6m 41s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/937d588b6e564f7aa8cc60ae4536b470 : FAILURE in 1h 39m 22s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/67f96fe945ce4d4d8ce182a0dac22ffb : FAILURE in 1h 28m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d59d4af7e10821d2af3966f864b499fc7a393bee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-05-12 16:20:39.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a893ef270107443377e614e4f7edc0cf181df0e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 19:34:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2cd39f7687644ecba5acce6ab86b38c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fa6ecf8489547399f171908ef472cb2 : SUCCESS in 6m 28s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2ef5111b95664566a8484a05649697ed : SUCCESS in 1h 36m 55s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/83a86013cf5641cfae66a7445235b2e1 : SUCCESS in 1h 37m 53s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/136505aca9ef4a7ba23805f6b6bbc977 : SUCCESS in 1h 34m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2216e97aef064c43a6d7d199aec0e7c9 : SUCCESS in 1h 34m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/569e740ee6ed41ecbf75f2cb6e09b07d : SUCCESS in 1h 40m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cd9b6d667f454a7dbdf4a9fada7ae565 : SUCCESS in 1h 35m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21b25c0cb0604e89a12ee11bb70c0ceb : SUCCESS in 1h 52m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf8537a832aa421f80e19edc176dc186 : SUCCESS in 1h 42m 14s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2c6bae7345f04f10819d94f4d34be18a : SUCCESS in 2h 24m 11s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cf66325554604903b005fbc096af7792 : SUCCESS in 2h 11m 28s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/775800a5243844098001f177ac6c6de8 : SUCCESS in 2h 45m 20s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a8a81d3c7b44085bb531502b7237004 : SUCCESS in 2h 12m 19s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a40ef015e7c3406cadfe21497de75cf9 : SUCCESS in 2h 37m 55s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37879faf303b4d269404b9cdf5deb939 : SUCCESS in 2h 31m 55s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/73a0c6a4578b49968094367a18d5b311 : SUCCESS in 1h 19m 44s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5da5dfd5ce8940ada267175c0751eaa9 : SUCCESS in 1h 20m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6822e2fa52d740acaa4acc6bd4501231 : SUCCESS in 5m 32s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29dac7553ebf4d4489922b72c3554630 : SUCCESS in 2h 11m 29s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ac60956021d04d89bd751491a7d3f005 : SUCCESS in 2h 28m 33s","accounts_in_message":[],"_revision_number":2},{"id":"39292c802bc53d3fd6d34352f4b16d37e0883d99","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-05-23 10:31:40.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"a84b888dc261a3bd8942cd6100167982d33b2109","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-10-07 19:43:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"1003cce7fceec784333f401b2bdf96c28b83a056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 19:43:49.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/ansible-role-pki for 948881,24","accounts_in_message":[],"_revision_number":3},{"id":"132694f91557606a85f4a600dadd3ec9ab61a8f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-10-09 10:20:31.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"980ef0950154e2f6dd1023aae8ccede07d59dcae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 12:38:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3650d659e8ac4025a35e9698440621b1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41493d93a6a447a6bef955a096fc9571 : SUCCESS in 6m 36s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/5d46f03ebdab4143860abf46d9686b85 : SUCCESS in 1h 24m 54s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0a7cb5bc3f0d4fadbd15745cb81011bb : SUCCESS in 55m 46s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e19381ff86354f028aa347fbc407aec5 : SUCCESS in 51m 48s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3aa1194afe4247dcade70684f33a8523 : SUCCESS in 1h 06m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/399aac1290bf41b095bd725a52fb932b : SUCCESS in 1h 26m 28s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e33c4025c7f44d3966d9aaa6f174991 : SUCCESS in 52m 36s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d5506db3ebd478fa588a6259e1c3597 : SUCCESS in 1h 29m 42s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4200ad7daf26491fa2c1d991962e95a6 : SUCCESS in 2h 06m 11s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4638119f222547208473e6df37257b79 : FAILURE in 9m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f340576ea1b1411d9a1755d52dbc1288 : FAILURE in 51m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4b048a6ca6944fe8157b28faa76c340 : SUCCESS in 3m 34s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a7c2df933364d08a9880b9486e89939 : SUCCESS in 2h 02m 30s","accounts_in_message":[],"_revision_number":4},{"id":"002b56dfb97b97942ab6fec1fe78cf51972c2419","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-03-10 17:47:12.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5b5ae40f299b77f0fc2c9cd66213f7b87b555508","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-03-16 10:34:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"844ba5fa0b8136fc3cba1eaf6a5147d21401cc0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 10:37:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8d950a03b500529e30162646f83299c8fbc7b670","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 11:57:57.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc14a18680044973b816eeaeab711b05\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fad4547ddebd41b1a30ef16a22d0b2b6 : SUCCESS in 7m 31s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/a2fb526919e04f2bbae1aeebada26e9c : FAILURE in 39m 51s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/47605d5fa9ec4f4fb13e292d4583900d : FAILURE in 41m 01s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/14599da47ab74ab6b3b666d82868631e : FAILURE in 40m 03s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ee9e8fcdca2341b89a74f08bd97bc7cf : FAILURE in 54m 19s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/25ee2f99b1f346a087f6741b43236b3c : FAILURE in 1h 06m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6242f3fef074a96b47ce0e4e0e842db : FAILURE in 1h 13m 25s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84724b3696d541d8bf5f1d5038206952 : FAILURE in 1h 00m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1798954a2ba849f399d8818561632779 : SUCCESS in 6m 04s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e317486c525c4edbae65d2c7c728ebd9 : FAILURE in 1h 20m 14s","accounts_in_message":[],"_revision_number":4},{"id":"6cb106406b136160309a9a40667c74d1d5a2821e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-16 18:48:25.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6f9649b0e27d070f56be59e181579987f02ab00d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 20:28:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d030a7feb1d493f8792499cf925f11c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84ed5c3a380d47859e81c73ff71052b0 : SUCCESS in 9m 48s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/a89fdca57ef94b6ca756b5b2bd2663a4 : FAILURE in 1h 32m 08s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b223f00e4ac749789a0a264f2c42f81c : FAILURE in 47m 37s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/160afba84d134d788a870ddb2529b454 : FAILURE in 1h 30m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/73a4b1e030394e1584ba5cc40d5bf61b : FAILURE in 1h 18m 01s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/52d029d42e45465a9aec1196164b6556 : FAILURE in 1h 13m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/824fba7191fa4619aa58802e031937b2 : FAILURE in 1h 32m 30s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d5085b5ecd2452eb2bfd9b3f772ad87 : FAILURE in 1h 24m 04s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/565223a13cea41e2a18f6708dc4b4940 : FAILURE in 57m 02s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cc6a2e431d3c4a04958dfeb2eeb00d86 : FAILURE in 1h 24m 13s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a31329f0f184982b3564ab70085ebec : FAILURE in 1h 30m 25s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c315b0972b3465f9962d90f754d6e23 : FAILURE in 1h 16m 34s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57f9c56137ad494a92ef52ed877912cd : FAILURE in 1h 06m 57s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c7812802c1f04c34bc3e726ab060aa22 : FAILURE in 10m 43s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/395658d7fdac4080a8ac7269946ba7c7 : FAILURE in 59m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9439aae2ce254265a8a1cc79c4dfe04c : SUCCESS in 5m 39s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/456819bf73794350911e1d8b40449b3b : FAILURE in 1h 35m 11s","accounts_in_message":[],"_revision_number":4},{"id":"37c41b8ed79f5ad8673e8fec06a3fdc8ffeae779","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-03-17 20:30:48.000000000","message":"Patch Set 4: Backport-Candidate+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"64ff7ac0c507e0103bee305507c0dff4e2fe0173","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-03-17 20:30:52.000000000","message":"Patch Set 4: -Backport-Candidate","accounts_in_message":[],"_revision_number":4},{"id":"b7a4bfb562ed5d86c10925f5b715b246bde3f887","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 23:13:23.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/601aa2c6dad047a7ab94e70da16aa2a9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3c9092bf1a64aedaba30fc0635f521c : SUCCESS in 7m 49s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/860ecf9a9d1b488fbaff5915876d722a : FAILURE in 1h 15m 21s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/74c993eef8eb48539349890319de4936 : FAILURE in 1h 24m 13s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/1fd5ce056a724f3e99cd1a3e3be26af0 : FAILURE in 1h 33m 28s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/13018a3c0fa54141a3ac3a61cc61cdc9 : FAILURE in 51m 34s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8391ec1f6c24463e97d419329f191fef : FAILURE in 48m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5c06212de1c4e13ad37956699c93c35 : FAILURE in 1h 26m 59s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53c10b3ad05a49c9a715fed87488a61f : FAILURE in 1h 30m 42s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4a8fd274519e4ea18c9e75f23955fb16 : FAILURE in 1h 50m 51s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/37d572011a6b473b92560cdd95348856 : FAILURE in 1h 53m 52s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61da6fe5178d46a2aaf3ad5e82e93cd8 : FAILURE in 2h 21m 49s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7f4e15697f84b70823eb3ffd3c23da7 : FAILURE in 2h 25m 47s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ae130ef77a74efca9ce85ad003f30b7 : FAILURE in 2h 35m 03s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/620e9c49755f4655a394c37d8c07d610 : FAILURE in 7m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0441efec71c34c68a9f06177f55b03e4 : FAILURE in 55m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b7dcbf69b13492a886b954c5a1fb104 : SUCCESS in 7m 03s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f8619619b8b44abb0549ede655ce0c6 : FAILURE in 2h 02m 52s","accounts_in_message":[],"_revision_number":4},{"id":"159fc115e64d7812cfc5a0b265c21420fb36b3fc","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-26 11:44:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"be4e209d00d10d55fbcde9eed4ef6dc2f129947c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-26 12:35:36.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_32666\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"_revision_number":5},{"id":"e620a18a7ad1f28a7a738afcff89e479658262fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 15:50:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a6ecde37cf64af7b4efb6de6cac54d1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75d1a739445b437db6ede41d8c4f8a26 : SUCCESS in 11m 29s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/abfcde1974e34e73a970a4ac8cfb1a20 : SUCCESS in 1h 09m 32s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c62eca52bb82411a857602d409f419f1 : SUCCESS in 1h 08m 15s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/2071ccf3f0b74d7f831bb55932894c42 : FAILURE in 2h 02m 24s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2091efc810114b49a7c37d78ac8b5020 : SUCCESS in 1h 52m 19s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c99318e49ab54b0db27c19d11e5ceba0 : SUCCESS in 1h 12m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ced83b4f49fe47fc91f41bd1864ac4f1 : SUCCESS in 2h 03m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76696c3690064cabac22de3cdfd5ef73 : SUCCESS in 1h 56m 18s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5c834c3b2fe2454db104f9554656742c : SUCCESS in 2h 29m 06s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/07f25c9f99dd4d8481806a6e1d8d5fec : SUCCESS in 2h 08m 16s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/411c858daf8d4911b1097fd24d4ac11c : SUCCESS in 2h 52m 25s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0886b6c36a4f4e19baae5af2349e0d33 : SUCCESS in 2h 40m 24s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3b3c7f029c2425dbde14c288de9bd19 : SUCCESS in 2h 58m 48s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/25c0d49fc5cf463581d1b8c816219ec2 : FAILURE in 10m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d80333ebfc64bcebefa14c2fec80a34 : FAILURE in 1h 02m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94abb860bbb0454a8bc9655133df6c8c : SUCCESS in 5m 43s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed586e3a5e9e41cba48cdeb6147b095d : TIMED_OUT in 3h 08m 06s","accounts_in_message":[],"_revision_number":5},{"id":"a03762c717a0af2835986a8c36db1343205bce0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-07 13:18:04.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_32666\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"_revision_number":6},{"id":"0609d9c81b16be54f9d5fca62095cf2a4a631820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 16:35:17.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbf05cd2819349e2b26c35fe5e161c81\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c35eadc160b44038a58828f51491d285 : RETRY_LIMIT in 15m 23s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/0ac780c3197d48858edac3ad17a1e36d : RETRY_LIMIT in 10m 16s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5c995005dc3247b789ac7a94d0d3334b : RETRY_LIMIT in 25m 29s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/aa415c8c50354ff798dd29a7879c7635 : SUCCESS in 2h 06m 36s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/28470c9bae3f43e5aae8a0ddd010b925 : RETRY_LIMIT in 13m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9d03963714ab4d1885fd167c98feafba : RETRY_LIMIT in 17m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e861bc14d2da4312bb368cfcd012fe05 : SUCCESS in 2h 02m 00s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93d8bd43919441dcbefd47b082f5dedb : RETRY_LIMIT in 12m 01s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/be966d835ae049b48a72b855ad183424 : FAILURE in 14m 10s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cdf3088af9db46ba86aab99994d4b2a2 : FAILURE in 11m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0eb562b9cae54577a056774bb8df0ee7 : FAILURE in 23m 40s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c80f51786c7f42638ce7baea1a119447 : FAILURE in 11m 32s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ec299d4627749d5bfc48517121cee47 : FAILURE in 11m 01s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5c848c15e6ed49d4bd9475fa03899ef9 : FAILURE in 14m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88a14134a9d54405a46b6d4c21379a86 : RETRY_LIMIT in 8m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9fe82a50f6242ee8284982011474f04 : SUCCESS in 6m 16s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/933ff6e845bb40869afbc9ebb00d1cdd : RETRY_LIMIT in 15m 28s","accounts_in_message":[],"_revision_number":6},{"id":"54b2edb038f209ae576908531bbc74be5ceca732","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-09 11:45:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7a470a7fa06f71a1a9004e83670b3fce0a02a494","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 12:53:33.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7880d2f9879342a082eb0a8261911f1c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/415491809b274ddfbeacbe1bcbbbc939 : SUCCESS in 10m 37s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/225d54f1783b488cb8e03b2c73d0c1a0 : RETRY_LIMIT in 6m 30s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/401fa57c921445d6bb8051b72ffed643 : RETRY_LIMIT in 10m 28s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/7f3b8db22c08470ea1ca1ef93d08e3c3 : RETRY_LIMIT in 6m 43s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a47d885ebb0842abb3489d948672e580 : SUCCESS in 1h 04m 19s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/db1396b8acb647efa5a3be7c5285aebd : RETRY_LIMIT in 9m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee76a416064a437f9fdeb1428524119c : RETRY_LIMIT in 8m 37s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d76b2dad99e540799755d4207d20e869 : RETRY_LIMIT in 10m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a34a66f7e6a244c8aa809ccbc435e27b : FAILURE in 8m 38s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d97b9dcee35a44fb89ebc761c2d611ed : FAILURE in 9m 49s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7fb295adcd76415195a3815734079a86 : FAILURE in 7m 20s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f18875e18fc54f4aa2e335366f600c8d : FAILURE in 9m 52s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d4ae347ad0949578ef7097d9092dcb9 : FAILURE in 7m 50s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dd40ab5d23bc4c568dd99c92bccfdc98 : FAILURE in 10m 25s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae2810ffa1934456be1c7d187ba84889 : FAILURE in 27m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22c83a74b5f245e686a09513c40a69aa : SUCCESS in 6m 15s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c3c13a9c5c746bdb42c6efa233423cd : RETRY_LIMIT in 9m 10s","accounts_in_message":[],"_revision_number":6},{"id":"8be483015cd3ecb18f684245564790096e27862d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-09 20:09:18.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f487c27ceb68a21441eeaedc539374c4a39fefa0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-09 21:24:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f5a732fcbf9f3ad718850a82a96039f0577e910f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 23:25:32.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9c3ae3edeba4e54aaaf15e3acbd36ff\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c79e1ae8d38e4ff692189aa14dec4543 : SUCCESS in 8m 47s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/4d2a91ab51a54a209c7e7bb563a39324 : SUCCESS in 1h 49m 17s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6259b90130b44c06ba8fdee458575c99 : SUCCESS in 1h 51m 33s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/404efb46ab3045748348a0144a76eff3 : SUCCESS in 2h 04m 22s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/82466a094b0d4fb69cd86868b313c0f7 : SUCCESS in 1h 10m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d8f069ae2c1b4c9eae11685f965110f8 : SUCCESS in 2h 01m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f26fc20bc2e49bdab7e3fe9103805d6 : SUCCESS in 1h 56m 49s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91cf8ea73aac486e8207f2155b75ca80 : SUCCESS in 2h 05m 16s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6e198f448b234cce9509b8d95b852734 : SUCCESS in 2h 50m 47s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/477e45382e114f388996ee13b7f171b6 : SUCCESS in 2h 51m 36s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/576bdb28728b4cc49b6ec2cd37e5c2c8 : SUCCESS in 1h 20m 55s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e1bf9f03d5e4b25a39388dcd4abae5e : SUCCESS in 1h 37m 26s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/236a0156b7b647b4bd7714cd906b05c8 : TIMED_OUT in 3h 07m 54s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4a02e06a1f3e406dbfe07cf705855474 : FAILURE in 8m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d4f7ccf81084849affd3a9e5cb93529 : FAILURE in 1h 12m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56cf711d4b7b4785aedbbe06d944e68d : SUCCESS in 6m 04s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f06f9e0ecb741f98a7705d6ca58a6b2 : SUCCESS in 2h 44m 25s","accounts_in_message":[],"_revision_number":6},{"id":"b6c5f87e87e6e34215b255c8b9f79bfcc72c90ec","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-10 07:20:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a5351db469294618e87d09a8f08d1079238a7e51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 10:30:57.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86fe32cd6a86428d8afa654c3600313b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5a7cd632c7c47d7b3ce86f1308ed811 : SUCCESS in 9m 15s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/21dfa3a06e184a399a110fb60f5be4fb : SUCCESS in 1h 01m 25s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/20b8ad1e82e64cecbee8f0ecdf87cd53 : FAILURE in 2h 01m 33s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/42613e901aa14585bfde9275e641c13d : SUCCESS in 1h 49m 30s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dbf1570cfdfc4a919079de1d679767c0 : SUCCESS in 1h 02m 48s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0d195b24835c47e3bef7f34160066be2 : SUCCESS in 1h 45m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd10a2f274b74461a477fe4b9e5f3ee5 : SUCCESS in 2h 08m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2218e7b4b005442ab3c0452b877ff963 : SUCCESS in 2h 04m 52s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8b2c3d50064d40bab04453a4236759b2 : SUCCESS in 2h 17m 45s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f3b363886c3f40058f48d3f11ff1f95f : SUCCESS in 1h 27m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3bcba2a6af0249749a53ffd596abb927 : SUCCESS in 2h 58m 10s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee119ee0edbc4839a6887a629bd9d1b5 : SUCCESS in 3h 01m 12s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91711caf2280448b9d08beb208d76e28 : SUCCESS in 3h 02m 29s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/09e6651d77504d7cb00d021ea121e79b : FAILURE in 11m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1cd7934e58b4abea9ee6da210945fcc : FAILURE in 1h 04m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e72b87ac7bb4e89bda9245710b406b6 : SUCCESS in 5m 44s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa74eda058204d8399f85c7d42d4fb2b : SUCCESS in 3h 01m 43s","accounts_in_message":[],"_revision_number":6},{"id":"08659b19b712555ca926a01a6fe0097d7827647d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-10 17:17:59.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f5aea08883ca8746dc9294cc242633042b8f2d2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 20:17:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a974325987d049e2a1c8981c3f18ed13\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a024f7539bca4bbf9bc09fc7fc43298e : SUCCESS in 10m 55s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/e28e2b2375c749b8b6423fb2b84a07c3 : SUCCESS in 1h 04m 16s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/159222cf9cd34dac913fed12a9056db7 : SUCCESS in 1h 50m 40s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/a135b9367e7943418b6c2ac5f26ec75f : SUCCESS in 1h 49m 38s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7081fa39d50a4802837e049cdfdefbbe : SUCCESS in 1h 59m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/810b3cf415df41d3b0d88056d86e2c0c : SUCCESS in 1h 39m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ceea11e980b146bc83d7d73ac146a50e : SUCCESS in 1h 58m 28s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/504f58ac5c554aa5aff5bf5d9f43880d : SUCCESS in 1h 14m 57s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a777429d6c644a38a4eeed9503bb1b19 : SUCCESS in 2h 18m 48s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/91e7ada7fb7041ddbc4da34328946065 : SUCCESS in 1h 31m 01s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0cc7b8251cd423f905135b79ab37d58 : SUCCESS in 1h 30m 18s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a19f98bb3b843aabcc55fb9ccdd7164 : SUCCESS in 2h 19m 32s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/085e00e945404d5f8df23d4ac0e7e1f1 : SUCCESS in 1h 32m 58s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4434443436844f68aac41d24eadae6f1 : FAILURE in 9m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66fe8d0ef7cf43a98e72e8be0ca5ba61 : FAILURE in 39m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/837320792ade44928573540c456905ef : SUCCESS in 6m 49s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf41efa6ce4640de8664e8f508d89e25 : SUCCESS in 2h 50m 12s","accounts_in_message":[],"_revision_number":6},{"id":"6150c47900b0e498917d9fe759411f844434180a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-13 07:28:14.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"aff994c636c4ee30526d187541f67ddbdb454792","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 07:31:17.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"a96e619c63e7541b930dd4ccc4e6a07aff7aef9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 10:09:59.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/85606eb225f34c70a33ad8ab8b48d294\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2ff07d7798c4519839d32ad50470ea5 : SUCCESS in 13m 34s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f903716e737c4fcd9fda24d54a536126 : SUCCESS in 2h 05m 20s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/cf78aa52f4aa4a98ac85cea83db49d02 : SUCCESS in 1h 50m 34s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/5cfab00d26c94707a70e13b0dbf2c4b8 : SUCCESS in 2h 07m 45s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1184ed723ecc44afa5e298101ef2ac83 : SUCCESS in 2h 00m 12s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/68ba3205629d412c8e76c5c5ed33933b : SUCCESS in 2h 11m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be986a42567e43cd8510326f92be71ad : SUCCESS in 2h 13m 42s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c369b509ca404a2dacffa90719ca0753 : FAILURE in 2h 20m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55687ccd6873410aba586eea3afe6cd8 : SUCCESS in 6m 59s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97454393c09045f0a2b52b7760df6ea7 : SUCCESS in 2h 34m 00s","accounts_in_message":[],"_revision_number":6},{"id":"270625714c8e3569e89b2967ed5402a2eec5fe2b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-13 10:10:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"760455c16946b78fc4071092f8aa2f870206241a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 14:11:43.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88fa8831b92a4a449a3f0aa60033dd08\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1aa07be4b64d4467a235534dbe689b28 : SUCCESS in 13m 35s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/c2fae2a482e54f25be322086a3002eb3 : SUCCESS in 1h 54m 15s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/68c1395ad79a4f5e927d0c069c00744e : FAILURE in 1h 53m 53s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/725a16776dc6429481f6d84298cbf16d : SUCCESS in 1h 54m 47s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ec3fbf286fdc4d0698a531642c8da4ad : SUCCESS in 1h 40m 00s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/303aae03711d460c9cefe2e5b4fc1847 : SUCCESS in 1h 40m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99702a7e32e34a41b2ca65698c23d340 : SUCCESS in 2h 00m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06ca6061205442f5ad49be8aba60fa03 : SUCCESS in 2h 33m 33s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ae4aded582ea422c97c998539a81ab03 : SUCCESS in 2h 42m 08s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b129a1a8ce124fd8a6878588ebc78754 : SUCCESS in 1h 30m 39s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8522c1c6f13f403fb05718cb507300dc : SUCCESS in 1h 48m 47s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/335db4fad230423cbf902d46a40b1c34 : SUCCESS in 3h 03m 42s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1bb05dc07a642cfaf8a918d86302fb2 : SUCCESS in 3h 03m 10s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/be60865324a04628acee80a2a2983d66 : FAILURE in 14m 43s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9058af0e48ec450b8f5f45e012509427 : FAILURE in 59m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1debf74bcf74a91854b53fed9d86fec : SUCCESS in 6m 47s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2bbea9b2b1e342a2a46e40b1c034dae8 : SUCCESS in 2h 41m 17s","accounts_in_message":[],"_revision_number":6},{"id":"aca709936d9def892a45ad3e4d66acf63531644b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-13 18:06:48.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_32666\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"_revision_number":7},{"id":"79437d5340f6a50fba8b7e5de0cb21409f46e9db","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-13 18:06:53.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"3b70503928a0cfb0cedffb713dd33f9c6a2543cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 21:22:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8271b177f9df4df6b9113bd268567231\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00de42ca8b614056aa1243e5b7588622 : SUCCESS in 9m 59s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/fa1879aaf93f41ff9d4c9cd8e7d66118 : SUCCESS in 1h 01m 28s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5208561d4cff4fee9645b3bda84a3eda : SUCCESS in 1h 48m 55s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/7f3b5ed6155c40f9a6e5a820d538e5a1 : SUCCESS in 1h 48m 33s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/996bd23846fe4f7f902d0321227c0033 : SUCCESS in 1h 53m 03s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8ff80572758c43efbf36ae48ed2e20fe : SUCCESS in 1h 09m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36517762c2a043dcb8191eb70842d030 : SUCCESS in 2h 14m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd673c45f22246a2b2edc4e9199097d1 : SUCCESS in 2h 00m 04s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f4d2a8bc2149428bad79cc492ee7dac4 : SUCCESS in 2h 52m 12s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e00241ced84648c4aa581a4be1458303 : SUCCESS in 1h 26m 28s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5abcad7e62ad43e7bd545a13c08a8809 : SUCCESS in 1h 26m 05s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/320ce5cc54464aa187814d6792cc6e99 : SUCCESS in 3h 05m 12s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/104329c9f8a643deb6e00724f48a2cc3 : SUCCESS in 2h 59m 51s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cf03503d41e54e7f9c86fca79f894218 : FAILURE in 13m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce794850f2f14709b7b6e0f3d775ea85 : FAILURE in 34m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58cbe775525d4c83b8e4a3b6bf142814 : SUCCESS in 6m 02s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53e52ca87fd146758bc836636c0819d5 : SUCCESS in 1h 27m 00s","accounts_in_message":[],"_revision_number":7},{"id":"2eeaa000d0d4ff0c9fec488a8478d16e47086df7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 21:25:38.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f2a1688b9ee12b04ec1e5cb8c5a4786166ac5aa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 00:02:26.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/71e8b954b6c74036aab4f710afe45681\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7d3a1a51b85462e8a89e370e60957d7 : SUCCESS in 10m 47s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/6ca335192f4940ceaf88e6e7bf564a05 : SUCCESS in 1h 36m 21s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c6f98de56ecd4f6aa04fdaa15a05fed3 : FAILURE in 2h 01m 29s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/d2fc10cb0b424e86b0c71cfdece6305d : SUCCESS in 1h 47m 03s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a791497f10f548fdac2a46865d8351d0 : SUCCESS in 1h 31m 35s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4ce1e972eb774b599744bd9d3096c948 : FAILURE in 1h 42m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56f6c21a0ee54c94abced740e5e9bda2 : SUCCESS in 1h 52m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9cd00e056d50418fa10d937781eea756 : SUCCESS in 2h 05m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d57e6dbe614cac8db48d912cfce20f : SUCCESS in 5m 44s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/500c47f950914ff884b73ba8ed8abe5a : SUCCESS in 2h 35m 33s","accounts_in_message":[],"_revision_number":7},{"id":"420e6b9ee2d3dbf7fd3cbf69cbe789087eb2239d","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-04-14 08:33:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"31b8b66ecb2f9243f94d29f81cc3b57e8f2866b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 11:53:45.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/870d188b13df4fd59ee6c66f97f0a94c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d146c5c17714b1fb2000c9f0a7568ee : SUCCESS in 12m 26s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/301dc959167544399100e1611c88a647 : SUCCESS in 1h 34m 58s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/429c7c893bb841338a6335b16c79cf51 : SUCCESS in 1h 34m 42s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/94581f65214b427583c9175608ef179a : SUCCESS in 1h 37m 01s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/718dd491e30f4786807db8f86ce6899a : SUCCESS in 1h 10m 00s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/993f553b76d642f9b03bc920d6caaa32 : SUCCESS in 1h 39m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/791c7124676942bdb75c9c928db3716f : SUCCESS in 1h 58m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/053892db1aa44fd39b9e2835fe111607 : SUCCESS in 2h 04m 29s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8ac6270b2e974400a7b3c071bbbc5434 : SUCCESS in 1h 36m 41s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/929dbfb68ae942dabc84c72afe85e635 : SUCCESS in 2h 22m 05s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6a3e2462e7e4745a2742ea9c3d77441 : SUCCESS in 3h 09m 19s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fb26e000d9145c9a7331c508a9fb02e : SUCCESS in 3h 02m 37s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1968874aa0f8422f982c18183518ef3e : SUCCESS in 2h 56m 32s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/df5ab39fdf0c4091a3746cb13ca80886 : FAILURE in 12m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb4fa0d4ea1d4caaa4940baaff7d22d8 : FAILURE in 1h 08m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df8dbb57f3c24890a6fc0b2bc6b7248b : SUCCESS in 5m 26s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca1b20a58bd245c9851513b31e80d715 : TIMED_OUT in 3h 08m 44s","accounts_in_message":[],"_revision_number":7},{"id":"ab5debf3ba9ccbd37fd8afe6d7253cb224072e0e","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-04-14 11:59:43.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e3d0b968849eb51c053a29de472d059bef025a0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 15:22:51.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddff425dbf3b4fb98eeafebab76b10b1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b436806ce0d4c40acb2b6f50c4bc480 : SUCCESS in 11m 38s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/a44e6b028fe1443da05dea8fffac4a1b : SUCCESS in 1h 13m 17s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/39e7c7b27d1e4881bfc1cee010222c51 : SUCCESS in 1h 04m 28s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/4b8e4aea37374ae8b2e3fa8eafdc4d71 : SUCCESS in 1h 01m 26s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/08d297220ab94ced9a7bdbab319dab75 : SUCCESS in 1h 45m 04s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cbdafb3e99174a94b4d6c03144d81f8b : SUCCESS in 1h 10m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28c4b6e4a583470eb5181551895da009 : SUCCESS in 2h 02m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06c48bb4b6684129be90ac1f114e7452 : SUCCESS in 1h 14m 04s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9c8c208908e64ae48593b4ce94b7ba94 : SUCCESS in 1h 29m 28s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d443aa7b563241cab4c4d32d33faf694 : SUCCESS in 2h 34m 54s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fdfc1b5b41034d95803a9a15317abb28 : SUCCESS in 2h 45m 53s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa789f4a822045edb9b890c9a29afcc9 : SUCCESS in 3h 00m 44s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d49b862bc4bc4110a8f755e8efae349f : TIMED_OUT in 3h 12m 42s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7bf654c8ce7e4097a3ca7c6274e99a2d : FAILURE in 11m 57s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d56b00da79e4d268658469775861f97 : FAILURE in 1h 09m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c89471fde3c44bda30a2769520f000e : SUCCESS in 8m 18s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc866a3680574b2b9034d4b944fe8aeb : TIMED_OUT in 3h 13m 29s","accounts_in_message":[],"_revision_number":7},{"id":"d5c3d2e81d6d8f20a597b4f3f5236f8e776a11af","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-14 15:33:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ad51d29c73fdf34ff0e5e981707c11330b1401ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 18:47:24.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af4a1e0333594864882aaa2ca09854b1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d896d46140b74ddabcd4a7da20e190f1 : SUCCESS in 14m 46s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/81de91f23c864ae98d9b6d4fce1d3b00 : SUCCESS in 1h 45m 06s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6283e2c3ae994f288f21ef65a6aadddc : FAILURE in 2h 13m 01s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/cb46837aa88d4f7fbe74893c15d5a00c : SUCCESS in 1h 06m 59s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/48766eef90b546bb86e0710c5a44ec42 : SUCCESS in 1h 50m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0a26d5c96864411fa007191012e09d1a : SUCCESS in 1h 48m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e56dd61e59544b795806ea7dbcd20a2 : SUCCESS in 1h 58m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb1a20ecfa734c8184dcc29f2b7cacef : SUCCESS in 2h 01m 50s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/153524e1bec24b5882c579f1c6ecd3d9 : SUCCESS in 1h 28m 59s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/18aa3ce9252f45d0bf1b3540452a517d : SUCCESS in 2h 29m 41s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d247172b2c7548489a54025ee71017fb : SUCCESS in 2h 39m 09s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91727d246b164af8a558e0ec66edd8ef : SUCCESS in 2h 44m 27s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f754964d34ec42639f602644d180e74a : SUCCESS in 3h 04m 59s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/49efb178ba054eac991ea1925f132f75 : FAILURE in 16m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39f712facb4b4900a5fbdd8024a005b8 : FAILURE in 36m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96b3519e243e4050bff50b4a0a8f5cec : SUCCESS in 4m 44s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a30fb3355da24a238bde9da10747f64e : SUCCESS in 2h 45m 21s","accounts_in_message":[],"_revision_number":7},{"id":"0477a8ea464182b9d7d06bdf05db1bf09a05c90f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-15 08:12:07.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"54e2d5300da9cd40fbd2f112c530269c5829d32b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 11:21:36.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd79a61a9e184f9eb79d75edd455bca0\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f528382951d84fa0a61d40876ffc737a : SUCCESS in 13m 49s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/7cc3b22d090f44df950b146b345b1460 : FAILURE in 1h 23m 09s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8c4167263e4849a09a74abf249971dca : FAILURE in 1h 20m 08s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/ef8dfa68ff32493caa7bbef17f863a12 : SUCCESS in 1h 32m 19s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0d0754aace60403b9831485468821167 : SUCCESS in 1h 08m 28s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8df581439b8547eaa02dc7afae15bfdf : FAILURE in 1h 33m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e63fe76c13564228a0299d1a245c8003 : SUCCESS in 1h 03m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2edc975ae9ec4198a11b10f8aed1dc6d : FAILURE in 1h 28m 51s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9465317fab5242b8865b6984768eae2f : SUCCESS in 2h 37m 04s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a066588f38a54978a42ec7c2916eed35 : SUCCESS in 1h 32m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8cddc4a3b0234e74a1a39c7220051140 : FAILURE in 1h 23m 39s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b083e5f25e2b474f8769dea1c7a1f76b : SUCCESS in 1h 34m 34s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81309b3a97cb4bd784aa49769330d95b : SUCCESS in 3h 01m 51s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4ea5f3e2416b474681eb6307664e82e4 : FAILURE in 15m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d04f43e6c2624b5ca861212a06a77ef4 : FAILURE in 1h 12m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/454d0e36517a4c4692c73b1e071d2e1d : SUCCESS in 4m 01s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/383c5ad8343146c181d738e27644f33f : SUCCESS in 2h 45m 28s","accounts_in_message":[],"_revision_number":7},{"id":"faca2f71f9596641059560fb4626c5fca9924d8b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-15 11:29:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"90c088f16bf024a5fbcba23391f8c088c0241332","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 14:42:43.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/08da16a3c1154e7ab2e430bd633c4106\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03efa8185e3a4c4f865a8f7f0cf6cb5c : SUCCESS in 12m 36s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/93b2bf49734c44d9b003aef541b5593c : SUCCESS in 1h 08m 19s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/1e5151b8901c4acf88409f1754ea2ebe : FAILURE in 1h 41m 05s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/f079bdbf6d4446d08f8430a0c5c41e30 : SUCCESS in 1h 32m 56s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d128f6698b654534b0ab512ee3fc7d89 : SUCCESS in 1h 40m 32s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4dbac625ce9c4a7ab9a112989ad4c317 : SUCCESS in 1h 08m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9499c1a48b704dcbb4f95791f714532b : SUCCESS in 1h 43m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efe65fc1f6d24a4abd06edf79b054c3b : SUCCESS in 1h 58m 32s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eb7169c75be24d33b92326ef97743be3 : FAILURE in 2h 15m 35s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4d3d2a5498274d7c851c8a63fa379e3f : FAILURE in 42m 40s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79b714268d3946ebb5f029034b43572f : FAILURE in 38m 13s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99e7d92c2c0d4200a848cdfcfaab6e50 : FAILURE in 2h 23m 24s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b1b7ad4163d4cce8d76b0e926ad0e86 : SUCCESS in 3h 02m 00s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cd559b4c768c42959ceef8692224b1d1 : FAILURE in 10m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d24a686afc8f4d0fa440c9bfe4d308a3 : FAILURE in 1h 05m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/840bbee349764293adab06f38fa3fed7 : SUCCESS in 6m 23s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b506bd1944c4414a5e32e9afbf391fd : SUCCESS in 3h 01m 59s","accounts_in_message":[],"_revision_number":7},{"id":"3955b7d844abe70400af6480eb7c1af938be5f86","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-15 19:55:25.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"90912296b868f820469842477059fe7200ddb92b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 23:02:29.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4736ea754234d64bebd2e7b0f1666fa\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04119e75fcca441889e89304a8521139 : SUCCESS in 9m 30s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/63704c4935c8439091df65c6b8938f6b : SUCCESS in 1h 01m 38s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/38e8b402131f43c9ae11c20f08704250 : SUCCESS in 1h 47m 56s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/e480c0928333474aab5fe156eac3a64d : SUCCESS in 1h 52m 23s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fa092cf3d8594ec58394019966d82c62 : SUCCESS in 1h 37m 18s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/42d845b14d994be99803fa47bc286f77 : SUCCESS in 1h 33m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12510da8573b4feaa1498dee62201f44 : SUCCESS in 1h 44m 17s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/876ff1fd36c943bd918a6cecc0259d00 : SUCCESS in 1h 36m 30s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/521869f9a54f463baaefedffbb41399f : SUCCESS in 2h 56m 15s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/45b31cd6eefb4ce2a4ee13e21b2bd77d : SUCCESS in 2h 27m 03s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71eb1137cb504afd83e235c252f48de5 : SUCCESS in 2h 54m 18s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a01da93aff884a60890dfacc054af451 : SUCCESS in 2h 10m 21s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31c8a230482148109326384f7222c477 : SUCCESS in 2h 53m 29s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c655920e884b4d309190646d6f6c8fe5 : FAILURE in 13m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aac7db782e7c41c4ba69b1b84311b22c : FAILURE in 1h 07m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a588c4b20c994ebd959e4d54e12757a4 : SUCCESS in 6m 04s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24f98e740bf64c608669dd1e6d5bc521 : SUCCESS in 1h 15m 32s","accounts_in_message":[],"_revision_number":7},{"id":"18dcca193ff89eb19371de0e68cb34edde67a92d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 23:05:10.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a3f5baf299a10e7451b03ec55f0caa964b09a057","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 01:40:46.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dfe0c95b5fdd450fa1d89a8a92ad55a7\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b2a85ecf30a4876a3aa546e014b861c : SUCCESS in 9m 20s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/69f622dd256f4fcba55d3be8212494a3 : SUCCESS in 1h 34m 15s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8c811ca0dd94472396352d71e1775995 : SUCCESS in 1h 55m 42s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/fdcc3f365cfe4b69849ca25ac348259e : SUCCESS in 1h 26m 22s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5b4b2f66a73640848f58552ea074adc6 : SUCCESS in 1h 53m 49s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3dec7256ffe9463d861ba878216ae7b3 : SUCCESS in 1h 05m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/573cc45080a14f04adc691fda74eb258 : FAILURE in 1h 21m 28s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32d6bfb9f013445a99c1c7df7a462b8d : SUCCESS in 1h 07m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/399ffd7dc6374934b71b9174d85056fe : SUCCESS in 6m 45s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83a3a86765bb465bb918e54eac305db6 : SUCCESS in 2h 34m 20s","accounts_in_message":[],"_revision_number":7},{"id":"555d2ddf016e5c64bf9befb7b246d70d0ee79db7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-16 08:50:20.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"198dd0cb98a68613261da170540fc0d34c7e1e2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 12:12:10.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32b633dbfa294d7e99805510647fc16a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d43b27cada447c09f5124c36d585bf9 : SUCCESS in 13m 41s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/00a1495fda5c4cc7b2b31be9e32c4853 : SUCCESS in 1h 38m 38s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/58a764dbd3854085bbc3c9ee675863ad : SUCCESS in 2h 24m 22s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/432e6db6eac445269d65c6bdda6caa6b : SUCCESS in 1h 52m 05s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/617b9884439e4c808bd98e0530006f43 : SUCCESS in 1h 04m 16s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6274f78f9c6c485abd82def87d8e8b32 : SUCCESS in 1h 39m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db34e7636663492f9e926af0ce39f871 : SUCCESS in 1h 40m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78ef8689c726450785a875fb978fda46 : SUCCESS in 2h 15m 46s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2e91c743c8a54172bb29c5944e2d9757 : FAILURE in 41m 18s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e1585fbdc0434131975ff30326c429b2 : FAILURE in 41m 21s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e5c5c5425664b5f8aea9c135a0c4831 : SUCCESS in 2h 54m 01s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b24584c7b3f41978851adbb9b46555a : SUCCESS in 2h 58m 01s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56298ff734024cbf877f7edaca8f1ad5 : TIMED_OUT in 3h 10m 10s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/304f7114a4f64151848a92af4eda7dcd : FAILURE in 9m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d18ba9346374f27846e134881622b3c : FAILURE in 1h 13m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f00f3859e38b493dbe95ea1cbf7fd2b2 : SUCCESS in 7m 00s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c987b20c01b14689bd63d34bb2d793cd : TIMED_OUT in 3h 08m 30s","accounts_in_message":[],"_revision_number":7},{"id":"567069f52c49b97b353053673755c087302d8066","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-04-16 13:52:56.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9ea73bef7d6abf336803454f2fd25b98345820c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 17:08:54.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4977eb81445c415ba8b6316392d2a842\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19e85b871c6c48168e26558f7270313b : SUCCESS in 8m 06s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f2519174987a46d4804d35e188fe025f : SUCCESS in 1h 54m 11s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/57b885d0d5074d9aafeda5853f101084 : SUCCESS in 2h 02m 04s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/46e4f76124a54225bd1297c458e48b5a : SUCCESS in 1h 50m 25s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a55e4b7fa3f647d7980097fd260259c4 : SUCCESS in 2h 01m 29s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/18bc3e0e269e456eb2e34c4fe7b60582 : SUCCESS in 2h 01m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61abe8e3defe43b79b5b470d9f447a90 : FAILURE in 2h 09m 15s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60b49af425b94151af00f107d166189a : SUCCESS in 2h 12m 10s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e486ea742fb445ef94862d51f7ffe0f2 : SUCCESS in 2h 48m 33s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b4b7d86842a943f0ba44a2f85247dffe : SUCCESS in 2h 28m 57s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63764fc7f4e54d63be49a2fab9ea504d : SUCCESS in 2h 23m 06s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70c57af290444fb29f758da9e86a00ef : SUCCESS in 2h 47m 40s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/473b85f44c144e94b5c5bc8cc756c031 : TIMED_OUT in 3h 06m 15s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3d34680e15bd42cea26d0ea4a70aabcc : FAILURE in 14m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3164f7dae93d47448dc8cc419bf9b329 : FAILURE in 1h 03m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0df20c42fc434e5b93a9b0e69451ccb9 : SUCCESS in 6m 31s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55831c5149654fe6b98429a2f45da818 : TIMED_OUT in 3h 04m 31s","accounts_in_message":[],"_revision_number":7},{"id":"79a25781daaa3e4eea30a3598a3c72968ea22971","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-17 08:12:10.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"275853ab974b9e54898b5a3cd29f92b553b1e915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 11:39:22.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0da66d61fd7d49e2af4e7bff58fb374d\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4639d957f06345919df8150ba97a2239 : SUCCESS in 9m 33s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/85d67b8871b74bbc824c8308f255db75 : SUCCESS in 1h 07m 22s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/762d8bd6b5ed4addae9155338b3cc815 : FAILURE in 1h 40m 28s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/8e71a57fc1494b0887632aba3dcf3c97 : SUCCESS in 1h 57m 55s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/432d893417ce4e34b7984e96b624d8e0 : SUCCESS in 1h 47m 10s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8340035fdf7347e581d8d4acea4efaff : SUCCESS in 1h 35m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ae9d151c2cf41c9a6f9b110418f8faa : SUCCESS in 1h 06m 53s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf0a887fc0284ad78bf6254f101fdeb6 : SUCCESS in 2h 07m 41s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a4f9932ca9894b98ae5ae6fe96dafbda : SUCCESS in 2h 59m 43s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e3638c59128a4f45945257d0773a0024 : SUCCESS in 1h 30m 17s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f430aeb93c94c4f8368b24e1087544c : SUCCESS in 2h 52m 51s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8960eb56bbb041208b5b27da51ca4d8a : SUCCESS in 3h 00m 06s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f640b2de0f474edba7643e98cba1fd47 : TIMED_OUT in 3h 12m 27s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e4b621b78570452aa174e60f465ecd9a : FAILURE in 7m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/573b107f58724423bfe98915b8eb4d02 : FAILURE in 33m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cc857fddcb549b6a041f6b2b4402308 : SUCCESS in 5m 14s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6248f334e969482aa13a289deeac1c3c : SUCCESS in 2h 54m 04s","accounts_in_message":[],"_revision_number":7},{"id":"4c35276fdcc38a9f2920fa6454243e5665acc3aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-17 11:41:06.000000000","message":"Patch Set 8: Patch Set 7 was rebased on behalf of \u003cGERRIT_ACCOUNT_32666\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"_revision_number":8},{"id":"f1e89462a6fa3b157772f8bb4b1de7a5ab702834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 14:50:24.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32160427abc543e99b8014cacd83a724\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0a707484e3e491fb5372a0f14cd3503 : SUCCESS in 8m 55s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/7d4940ae960145cbb99476f06b6cf95d : SUCCESS in 1h 26m 35s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/68d349884ff2451ab3ddd67b6c3d5e6f : FAILURE in 1h 53m 03s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/83aac7c19f2f44af8925d247be9d8ccd : SUCCESS in 1h 59m 04s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fdfe9e57447f4908906535c3045d040c : SUCCESS in 1h 30m 10s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/04bf9d502a76424baf4fca36f1da57da : SUCCESS in 1h 11m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ede9e42c6e944829b2b4a36ba16b7a0d : SUCCESS in 1h 33m 36s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d5930a6b67b4f6bb3ac1cb57be3e749 : SUCCESS in 2h 17m 34s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/73e94f8e29f64d88a2eb9ba0d24bab92 : SUCCESS in 1h 23m 58s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9a502462d0f14c0c9c14688afb4f941b : SUCCESS in 2h 54m 25s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be5d57d27b9c4044b5e01da9a729a28c : SUCCESS in 2h 15m 12s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e51692d53524cdaa0f87eb5580884e1 : SUCCESS in 2h 12m 22s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0c463643e924adba78a788d4d9ff53a : SUCCESS in 1h 33m 46s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ce2d7a264a3142dea69a8299d8d117e6 : FAILURE in 9m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15462365a8664085bff02367ffb18b03 : FAILURE in 1h 01m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8259b65be3049609705a9ad7cf2a692 : SUCCESS in 6m 57s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25310d7480934922bb2bdc288772f3a6 : SUCCESS in 3h 01m 58s","accounts_in_message":[],"_revision_number":8},{"id":"f0e9bde18c3ce098abe6cac17216fab6d07e1006","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-04-18 10:50:29.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9985a11d1b5a9b07e28e13bed1b8acf2229b9473","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-18 14:13:44.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e125026b46ce41f5a3e9db03e64bc158\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/614d6d9b9af14b14a48ed09c90765929 : POST_FAILURE in 12m 39s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/cb3f8dd9e3d7484a89e3b333b7094e1c : POST_FAILURE in 58m 11s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/82a4696b14654ec19fdd4c78dae9c8a1 : POST_FAILURE in 1h 54m 36s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/b44aee2ad03545d4a30f055ff0ddf9ef : POST_FAILURE in 1h 53m 04s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/928f20be6db1412284366ccd7bcab4ee : POST_FAILURE in 1h 37m 54s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9c4cc41959d247878d3a36c90f2d214f : POST_FAILURE in 1h 04m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a7d8101660d45b88cb150278bbb75c9 : POST_FAILURE in 1h 56m 00s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14a48ad19a804479b1609ee200aaa1bc : POST_FAILURE in 2h 09m 56s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dae103b8748b46678b00d8058dbb1c86 : POST_FAILURE in 2h 27m 10s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d3a5016f7bee4869b5df3453b243a02c : POST_FAILURE in 2h 24m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d0195981b324d4385d893073cda4918 : POST_FAILURE in 1h 23m 14s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd3102b0024b44579b3773520d2516c8 : POST_FAILURE in 2h 49m 42s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df21b8b7ad064a638d2d291dcb67dc71 : POST_FAILURE in 3h 08m 09s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bf936a9c62db41de93c7741bbe0ea7a8 : POST_FAILURE in 15m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa9165710d924837b09b9a0fb4877a92 : POST_FAILURE in 54m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8023550577804abfbddac507be9959b0 : POST_FAILURE in 6m 57s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3793985f5a9b42589f9cb0277ef2eb57 : POST_FAILURE in 2h 15m 07s","accounts_in_message":[],"_revision_number":8},{"id":"8e77d47fa4a52d8490edb3dd012f24aff7ba4535","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-20 09:22:35.000000000","message":"Patch Set 8: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f1c0fea6933f9dc643450c57ad7be42f4ba9ebfb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-20 11:20:22.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a1d50ef3ed18bb10a49021f0bc1fcfaa69b15f03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 12:40:57.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85ea42f71f054ba0834b1550badfdead\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/397d44296bed4d4aaf2ed286a547c8df : SUCCESS in 15m 52s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/7e19948c01fc47bc9ebebb2c2b33c177 : SUCCESS in 1h 06m 46s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ee3186fddcaa49d58e1d37f9bae7bf91 : SUCCESS in 1h 39m 20s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/b081d49789f047e28372688a2817d28a : SUCCESS in 1h 48m 00s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/27678d7cdf2f4e4ba8d2351a9093c521 : SUCCESS in 2h 03m 53s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b47bf48602f0405cb0d4dac566608774 : SUCCESS in 2h 03m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1744b4656ceb4c0cb97ef8dd3472a1e5 : SUCCESS in 1h 57m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aecc2bac117b4d68a6a49bcea2befc20 : FAILURE in 2h 26m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/205357fd0b534c6a9e067059961e4187 : FAILURE in 47m 43s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d6bb43aa5dea4e538c4cb8898d091934 : SUCCESS in 2h 30m 42s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/488c0cf4917246a785a74e0e8c13cd64 : SUCCESS in 1h 41m 19s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7523256c6d284a4aa0edb6e578d27830 : SUCCESS in 2h 31m 45s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ccd0b7ae057a455bae0aee562229fb98 : FAILURE in 2h 58m 53s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/581566dcef3d453093de159af8b85f8b : FAILURE in 15m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abd61667af034e78a50efa5ffbbb4ecb : FAILURE in 1h 05m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cecb45b786a476a95b95e4e37430bcb : SUCCESS in 4m 10s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69afbbf879fd4222a020d759210d6317 : SUCCESS in 2h 48m 50s","accounts_in_message":[],"_revision_number":8},{"id":"7248754a75ff09d3e49a0711cb663aba9f8202ea","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-20 17:44:50.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8c9e9d9c5429da5859c85e81b34c67d451889728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 20:52:47.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8e4eeffb2964ca69e47fcde49163999\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aeeffc5cc0b941f5b8dd68e5c88dffc1 : SUCCESS in 14m 07s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/06fd5bd3bc124121b7aebf5b13cb1686 : SUCCESS in 1h 41m 40s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/de87d713fe3245b7a0915b0da6736897 : SUCCESS in 1h 15m 26s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/9c9e6e0520714068a5565acf69da2d24 : SUCCESS in 1h 50m 46s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c5fc66c4134642efafdf93c44bc12e4e : SUCCESS in 1h 09m 13s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2f4d50db5f2d48a4919f32517767620e : SUCCESS in 1h 06m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c562e6751c6a4e4abe530342f47791fe : SUCCESS in 1h 45m 08s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/844980da2d754f13b913bce002b194af : SUCCESS in 1h 22m 39s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/69b786baacc14e45bf5369e2208befc1 : SUCCESS in 1h 44m 28s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f8843c4a83ca498d8293e9925c32dcf4 : SUCCESS in 2h 33m 24s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32a5c975003d44aa8de1e3b1d4993fa0 : SUCCESS in 1h 47m 39s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a47d5ed72c5484cb89ea6e90b564ac3 : SUCCESS in 2h 48m 20s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/298ecfe2f39f4ef58f9c5ea27c76b573 : SUCCESS in 2h 54m 12s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3bd8b14dc9824bbc9bcde0550cf707a0 : FAILURE in 14m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c2fa929840642f6b0ce9b843b7d1dda : FAILURE in 42m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8384f37a6033413aa14d7bee42047c83 : SUCCESS in 3m 57s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8d608cb4337482080ff62f23a0413a5 : SUCCESS in 2h 44m 23s","accounts_in_message":[],"_revision_number":8},{"id":"e756a501b8b91d6fa4f63b1013184f18907cdbb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 20:55:28.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"7b7962729a7ecdbf4887f5d6e26ad3c72bb50502","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 23:40:00.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d19c28207594b7e956fa863521e6053\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bae31d4300d846a7abc3630ddc4fdf50 : SUCCESS in 13m 03s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/8fd51e1f0a5242158e4d5f4926fb78b4 : SUCCESS in 1h 41m 01s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/35fc1a0c1f2e4071b2291d42d8d4acb8 : SUCCESS in 2h 09m 40s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/b73fb0f354204bbcbc3409be1531dc3a : SUCCESS in 1h 57m 07s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fecd3becd1d540559d90ab5a2a35b740 : SUCCESS in 1h 18m 08s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/67a408ba95ab480c84a6e9d7cfa17dd2 : SUCCESS in 2h 03m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4722b49120e7401aa63d3f83c02f0af6 : SUCCESS in 2h 00m 26s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8fdc7ba301c48a8a2dd6c5f0ac582d9 : SUCCESS in 2h 09m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7e8e97feada4a4889215cf452712739 : SUCCESS in 8m 02s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/713f8bb4c77b4422ab8c176f4d64cab4 : SUCCESS in 2h 39m 34s","accounts_in_message":[],"_revision_number":8},{"id":"f0ee8047b1ee9b5b7b897c20b9e08c96575ccf4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 23:40:00.000000000","message":"Change has been successfully merged\n\n4 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":8},{"id":"7bcb4b8abcde0ef8a052b7103833145128c622df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 23:41:45.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbbca828fb8d4c3aa30d85310b8924f5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d06737de67d4e9984dbf534f802564f : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"f49e21b05bdac0edf2c37efcfe110e45edb4d077","revisions":{"da7cb5cea5443ed1d6811f205fdd2f0e076ea121":{"kind":"REWORK","_number":1,"created":"2025-05-11 20:28:46.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/19/949419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/949419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/1"}}},"commit":{"parents":[{"commit":"d1b9fd84eaf60910270b0dd71545af7f023f71d5","subject":"Merge \"Align on cinder service naming\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/d1b9fd84eaf60910270b0dd71545af7f023f71d5"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-04-02 19:16:49.000000000","tz":0},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-05-11 20:28:44.000000000","tz":120},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `octavia_api_cert_san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n\nChange-Id: I107f4716d1ad165a0b3ae2cb2ee48d22c03d9aab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/da7cb5cea5443ed1d6811f205fdd2f0e076ea121"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/da7cb5cea5443ed1d6811f205fdd2f0e076ea121"}]},"branch":"refs/heads/master"},"5b8563c4e4f1762179fab53e1e08106bdcadf41b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-05-12 16:20:39.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/19/949419/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/949419/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/2"}}},"commit":{"parents":[{"commit":"d1b9fd84eaf60910270b0dd71545af7f023f71d5","subject":"Merge \"Align on cinder service naming\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/d1b9fd84eaf60910270b0dd71545af7f023f71d5"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-04-02 19:16:49.000000000","tz":0},"committer":{"name":"Damian Dąbrowski","email":"damian.dabrowski@cleura.com","date":"2025-05-12 16:20:39.000000000","tz":0},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `octavia_api_cert_san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/948881\nChange-Id: I107f4716d1ad165a0b3ae2cb2ee48d22c03d9aab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/5b8563c4e4f1762179fab53e1e08106bdcadf41b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/5b8563c4e4f1762179fab53e1e08106bdcadf41b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"61c9d60e3856432331a02dc4301d36cbe7c4d025":{"kind":"REWORK","_number":3,"created":"2025-10-07 19:43:34.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/19/949419/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/949419/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/3"}}},"commit":{"parents":[{"commit":"d1b9fd84eaf60910270b0dd71545af7f023f71d5","subject":"Merge \"Align on cinder service naming\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/d1b9fd84eaf60910270b0dd71545af7f023f71d5"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-04-02 19:16:49.000000000","tz":0},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-10-07 19:43:32.000000000","tz":120},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `octavia_api_cert_san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/948881\nChange-Id: I107f4716d1ad165a0b3ae2cb2ee48d22c03d9aab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/61c9d60e3856432331a02dc4301d36cbe7c4d025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/61c9d60e3856432331a02dc4301d36cbe7c4d025"}]},"branch":"refs/heads/master"},"75571b70dfa1038aa013889f0374bf95396f4a0e":{"kind":"REWORK","_number":4,"created":"2025-10-09 10:20:31.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/19/949419/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/949419/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/4"}}},"commit":{"parents":[{"commit":"d1b9fd84eaf60910270b0dd71545af7f023f71d5","subject":"Merge \"Align on cinder service naming\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/d1b9fd84eaf60910270b0dd71545af7f023f71d5"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-04-02 19:16:49.000000000","tz":0},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-10-09 10:20:25.000000000","tz":120},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `octavia_api_cert_san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/948881\nChange-Id: I107f4716d1ad165a0b3ae2cb2ee48d22c03d9aab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/75571b70dfa1038aa013889f0374bf95396f4a0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/75571b70dfa1038aa013889f0374bf95396f4a0e"}]},"branch":"refs/heads/master"},"6151e064a7a558bfc26619ca3ffdd696269c2c02":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-03-26 12:35:36.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/949419/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/949419/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/5"}}},"commit":{"parents":[{"commit":"fbc197f51fe28d407c9e8212657fd4580457efc1","subject":"Rename Octavia groups to remove dashes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/fbc197f51fe28d407c9e8212657fd4580457efc1"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-04-02 19:16:49.000000000","tz":0},"committer":{"name":"Damian Dąbrowski","email":"damian.dabrowski@cleura.com","date":"2026-03-26 12:35:36.000000000","tz":0},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `octavia_api_cert_san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/948881\nChange-Id: I107f4716d1ad165a0b3ae2cb2ee48d22c03d9aab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/6151e064a7a558bfc26619ca3ffdd696269c2c02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/6151e064a7a558bfc26619ca3ffdd696269c2c02"}]},"branch":"refs/heads/master","description":"Rebase"},"fdec8311faa776ffb5916c71d9524288f784e924":{"kind":"NO_CHANGE","_number":6,"created":"2026-04-07 13:18:04.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/949419/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/949419/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/6"}}},"commit":{"parents":[{"commit":"13261f6bf14f5fc9f69b31052fd290906cad1091","subject":"Rename Octavia groups to remove dashes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/13261f6bf14f5fc9f69b31052fd290906cad1091"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-04-02 19:16:49.000000000","tz":0},"committer":{"name":"Damian Dąbrowski","email":"damian.dabrowski@cleura.com","date":"2026-04-07 13:18:04.000000000","tz":0},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `octavia_api_cert_san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/948881\nChange-Id: I107f4716d1ad165a0b3ae2cb2ee48d22c03d9aab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/fdec8311faa776ffb5916c71d9524288f784e924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/fdec8311faa776ffb5916c71d9524288f784e924"}]},"branch":"refs/heads/master","description":"Rebase"},"3255255f6261538d0e0387700783e73f1000c817":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-04-13 18:06:48.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/949419/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/949419/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/7"}}},"commit":{"parents":[{"commit":"a17d6fd4a62378a93b36ce014de7e53e8fd2a632","subject":"Merge \"Remove leftover of nova-cert\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a17d6fd4a62378a93b36ce014de7e53e8fd2a632"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-04-02 19:16:49.000000000","tz":0},"committer":{"name":"Damian Dąbrowski","email":"damian.dabrowski@cleura.com","date":"2026-04-13 18:06:48.000000000","tz":0},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `octavia_api_cert_san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/948881\nChange-Id: I107f4716d1ad165a0b3ae2cb2ee48d22c03d9aab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3255255f6261538d0e0387700783e73f1000c817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3255255f6261538d0e0387700783e73f1000c817"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"fdec8311faa776ffb5916c71d9524288f784e924","theirs":"a17d6fd4a62378a93b36ce014de7e53e8fd2a632","contains_conflicts":false}},"f49e21b05bdac0edf2c37efcfe110e45edb4d077":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-04-17 11:41:06.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/19/949419/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/949419/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/949419/8"}}},"commit":{"parents":[{"commit":"c373ecbc7acaaa371bb213948f7923f9fd09038d","subject":"Merge \"Disable Octavia coordination temporarily due to ZooKeeper instability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/c373ecbc7acaaa371bb213948f7923f9fd09038d"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-04-02 19:16:49.000000000","tz":0},"committer":{"name":"Damian Dąbrowski","email":"damian.dabrowski@cleura.com","date":"2026-04-17 11:41:06.000000000","tz":0},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `octavia_api_cert_san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/948881\nChange-Id: I107f4716d1ad165a0b3ae2cb2ee48d22c03d9aab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f49e21b05bdac0edf2c37efcfe110e45edb4d077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f49e21b05bdac0edf2c37efcfe110e45edb4d077"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"3255255f6261538d0e0387700783e73f1000c817","theirs":"c373ecbc7acaaa371bb213948f7923f9fd09038d","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
