)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~958939","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~stable%2F2025.1~I358d11a7a0b82b2ed9c6aeaf4650a17e878fe25d","project":"openstack/openstack-ansible-os_octavia","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2025-09-01 06:53:23.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2025-09-03 05:44:42.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-09-03 05:44:42.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-09-03 05:44:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I358d11a7a0b82b2ed9c6aeaf4650a17e878fe25d","subject":"Add default values to environment lookups","status":"MERGED","created":"2025-08-30 06:08:42.000000000","updated":"2025-09-03 05:45:45.000000000","submitted":"2025-09-03 05:44:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958939","cherry_pick_of_change":954602,"cherry_pick_of_patch_set":2,"meta_rev_id":"c9ec02517b76b5e9b064c111e7e5d8c03244a0df","_number":958939,"virtual_id_number":958939,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-03 05:44:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2025-09-01 06:53:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2025-08-30 06:09:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-30 15:24:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2025-09-01 06:53:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-30 08:45:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-30 15:24:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-08-30 15:24:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2025-08-30 15:24:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2025-08-30 15:24:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"2e9db688f0abd9729aa5db7b668e67425aa5496d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-30 06:08:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"09535ce13cf7db9e07d3ec1c32a59440f4b357b6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-30 06:09:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dc97deaafe64fde7f40e8c7e6dabd9b89defc80b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 08:45:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fca2d4bcb39f4cbaace90fae234c75d7\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/385914ba32dd47f1a8bfa0742f512d52 : SUCCESS in 6m 09s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/fcd23bb714194b9a889a9ab2dbe22555 : SUCCESS in 1h 42m 05s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2f6466f6516432b8a3574236803a8d2 : SUCCESS in 1h 11m 47s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f993b257fa6d4d2a97aeba0ae2093aee : SUCCESS in 1h 41m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20389541a94a412788124fd8a2e61a43 : SUCCESS in 51m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6aa49bbc56240ef930c29bec0466b67 : SUCCESS in 1h 46m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/365daa63e11941929652c40fc5f5dbae : SUCCESS in 1h 49m 36s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/259b9a309ce2441ca014f496bb320570 : SUCCESS in 1h 40m 12s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/561c133066dc480fb1ecd8ccbd2a5644 : SUCCESS in 1h 47m 36s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/49eff725f3b34171bf443acd5b0d0d69 : SUCCESS in 1h 47m 25s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c678b3477cd748208c18e3a6cb5f07e5 : SUCCESS in 1h 16m 11s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc6248516fb04d72b30b8ab67e672c28 : SUCCESS in 2h 16m 03s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0dbcb90e27f540258a630d3c314d30e8 : SUCCESS in 2h 13m 18s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4b1513fda1c74b31917b2351ad74f7a6 : SUCCESS in 1h 12m 46s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4e20a21a7e241b49fdc6f56eaa0d10f : SUCCESS in 2h 27m 17s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8496343054664afebd3bdd0952ace51d : SUCCESS in 1h 20m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c8d1f5e3f2842d2ba4f5bb4a724ffad : SUCCESS in 43m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fc8e350b4b04cfb8fca30f7e3fa5166 : SUCCESS in 3m 21s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c5343ebafe04ccf8c9ff098fddb5ca3 : SUCCESS in 1h 04m 22s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1e11c4dd54ac4522b2e7c17cfe727fb2 : SUCCESS in 2h 28m 37s","accounts_in_message":[],"_revision_number":1},{"id":"62d557019312ee28185998d58354234bd86891a9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-30 15:24:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b1a4efd19ea7fc8981b0869d5a07f1c086ec8a23","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2025-09-01 06:53:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d0cf468870e0b0f3b8e379d33586453cd0c7695","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 06:55:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"60cdfd1a22899e77a8f42f553deff8e31a5e83b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 09:23:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3eec901cc444e9686e611246dc279c4\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b89da964eccc4ef186d26f7493dac5e3 : SUCCESS in 5m 38s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0e4fc96f7f1c4533aacd881a839cc308 : SUCCESS in 1h 52m 43s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a4df65c6230a4240ad9830b445cabc73 : SUCCESS in 1h 16m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/978a6f041e1649f386961536054064dc : SUCCESS in 1h 43m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/329f3d49d28c41b1b44a51bd6586a3fa : SUCCESS in 1h 58m 46s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c75ebb2ca2734689a2ece91ea9fe27c3 : SUCCESS in 55m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad15508a0a2a451bb4c98fa86e40b2f4 : SUCCESS in 57m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c3fe0de1f7042659ff0ba79c95a0c5b : FAILURE in 1h 09m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e85f6aa00b64e3f9fec7bca152e291b : SUCCESS in 6m 19s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3dfd36545aaa41ecb3769e90c4918eaa : SUCCESS in 1h 04m 35s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/991926f94da24318accfcd52020a02d9 : SUCCESS in 1h 56m 00s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0b2a82fdec34c39a10787fabb452272 : SUCCESS in 55m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb1424a42e59b13d08240c38365df680863a1f2c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-02 07:31:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c3600e9f524adcebb860ab3a26b10c852cbc291","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 10:05:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bab0432f56a4a61a6668067cd594238\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b78d8a7b815e40de9da9b0a39767e410 : SUCCESS in 4m 07s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/06ed41ad491f4dc9b6518174ba8389d0 : SUCCESS in 1h 43m 56s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/546b399283294ad990d09df4b44b8b99 : SUCCESS in 1h 29m 53s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/21c272a56bac4b619697a71be8ae8ab7 : SUCCESS in 1h 47m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/459c540859ad4ba5b65e0d3be2f8f20c : SUCCESS in 1h 22m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7432f5efd60422d824c24359bbc377e : SUCCESS in 1h 34m 39s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fb2c1bb6b57a4e8ba9c0466019b0b5ed : SUCCESS in 1h 35m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/530e35cfe8164d9bbf39cc3c9c02dba3 : SUCCESS in 1h 28m 26s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50886371196f49c491ec94d7a7a0df81 : SUCCESS in 1h 29m 46s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2711fe7970f443b6a1618f5ce3362a4d : SUCCESS in 1h 51m 32s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cc79e066798e47f79fbcf6cd3e62ffca : SUCCESS in 1h 11m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4a2a22ae5044038b7a7d7cac2919efc : SUCCESS in 2h 12m 45s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a69c4c33009d4de8aff3770d271a4a14 : SUCCESS in 2h 14m 13s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b47c1d0c3ae7479f8575717e3fcd197d : SUCCESS in 2h 26m 19s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43d08f0ba3214825860c43b6fae89791 : SUCCESS in 2h 19m 25s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b9988aaac31c4ecba73265ba95d9dd98 : SUCCESS in 50m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d3b3cc60abd4ad99d8330c87a76465e : SUCCESS in 1h 26m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce13a73cda3e4a8cb1fcd8501211add3 : SUCCESS in 4m 08s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb61c2ddd83d4d1cb6a27689ecb2e1c1 : SUCCESS in 2h 11m 00s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/76687a9264db49c198f7561d792173e8 : SUCCESS in 1h 10m 13s","accounts_in_message":[],"_revision_number":1},{"id":"af8fe8c6169d1e34939c8d1314065841f5e0319c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 10:07:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7a5ca739c4ec727c580c0c5f8c66e376417260cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 12:30:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1c880330ed94fa3890379a517681b91\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e063bbfa251b4cda82dc29d8c964f640 : SUCCESS in 4m 22s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3dcf92b5274d49cca2fcb682d2fd0869 : FAILURE in 17m 54s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/810a471d11fe4fbe97e0fd150cc8a350 : SUCCESS in 1h 15m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27ebb6bea5924744ab296b72669767b0 : FAILURE in 57m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac84ecbd77e44b94a892661d61c535c7 : SUCCESS in 1h 37m 07s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7fe9dba576c64da2870d0edcb39f9c89 : SUCCESS in 53m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e0c156e0d234c26aebbcbb749983b06 : SUCCESS in 1h 30m 23s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5745c6d4635b45f79ed7d7d3eaced82a : SUCCESS in 1h 38m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65f47f2f8e8741028a03d9592ceeacef : SUCCESS in 6m 00s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e34e95b2c6e4e9baeeb6e3b0b58cbdb : SUCCESS in 1h 04m 25s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2b62201ea15742968598bd453e065b0f : SUCCESS in 2h 17m 42s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/55ac38eed88f4efa959724e89da0d960 : SUCCESS in 1h 38m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"41a0774e941f6d6db4ce450b8b78a71a5f8a95f4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-03 00:36:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc1410a5ab4fdfd32d255a755ff9b7a1d0827593","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 03:40:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/402cd1fc5786420ab3e9521f41cf8123\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/feae0fd2964a4fdc82dc33ab478f69f6 : SUCCESS in 6m 36s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/bdd0886bbd2245e9b6942096f232ee64 : SUCCESS in 1h 31m 47s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/017a9395634a4cc680b68c0c9a78d1fb : SUCCESS in 1h 58m 32s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5a1fbc7a290448efb58381a03692fd4f : SUCCESS in 1h 48m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87809227b8ac48e1b664197ec31d1bfc : SUCCESS in 1h 17m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4fcc0446aed42f782d99eb251701131 : SUCCESS in 1h 42m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0897b1b6702241bea7b7bd99edac45e1 : SUCCESS in 1h 40m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/536bd94099594fd692f2287f2c80ee8a : SUCCESS in 1h 43m 49s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b2d2318c9724caba942aa28be77daca : SUCCESS in 1h 43m 53s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/233098aabaf044e7b86beef90c8c9afd : SUCCESS in 2h 56m 53s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7f0ecd7577a24d91bd3f329f4e5acfc9 : SUCCESS in 2h 30m 02s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/438a3e8fe72b4815bb582926f229b4e6 : SUCCESS in 2h 18m 47s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/084713e03e7f4bcfa6feaf987b4e0625 : SUCCESS in 2h 32m 22s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/81541240009a4c0a9f052e49678532ee : SUCCESS in 2h 36m 58s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8da7b9c5ba64fba8574f5c13f84c24a : SUCCESS in 2h 11m 31s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9ee1aa6ccc044c8c8346bff676481055 : SUCCESS in 1h 23m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2004a1a07144a5993c8a0a8ccc80edd : SUCCESS in 1h 03m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0322100317a04a1b8f0abf3a5fa8a659 : SUCCESS in 6m 04s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c13e1df3fdbf4728a5e93f2ce1ac2b3c : SUCCESS in 1h 43m 00s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4e99ceb382004729a164fbf8f6a34ca1 : SUCCESS in 2h 17m 31s","accounts_in_message":[],"_revision_number":1},{"id":"16e6caa394c32fa2c97737e1648335cd27862e0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 03:42:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c0f33fc5faa61719982cdd57c171344475047f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 05:44:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce9671f209ce4e5ba727e2ca4597b044\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cead827ad3e41e898dafac045c1e5b4 : SUCCESS in 4m 07s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/040d009ecf0a47fbb31bc0fc634005dd : SUCCESS in 1h 28m 07s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/acc12834a39f4f97933b16e29f9835c4 : SUCCESS in 51m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c37a4a6f1b444aa7af02f5ece37cb52c : SUCCESS in 1h 29m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/261256a529714204a554c185778a4a0e : SUCCESS in 1h 33m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/988e0719e45f4c31b292896133987236 : SUCCESS in 1h 19m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca67db3181c94c11a2e9aa3b649c713a : SUCCESS in 50m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83aad7640294403db981208f36e9bbfb : SUCCESS in 52m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9618b3de4bd24d2c920b551e66626460 : SUCCESS in 6m 11s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7cd4478c7e04cabb1b8710b0dbeb604 : SUCCESS in 1h 38m 45s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0ed46a96998f429b9c012ce4f607c316 : SUCCESS in 1h 53m 08s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/64197ac50a204ae490657ca5829320ef : SUCCESS in 1h 28m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e52d3fa346bca17488d899de5f4baf6cba33474","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 05:44:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c9ec02517b76b5e9b064c111e7e5d8c03244a0df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 05:45:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab5eb7936cab42c9bbe7f3c62ac0655c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f0b8fd4b354d13be69fb5ab95d1630 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e83d62460905ed1a7a480789087ee261dc16b423","revisions":{"e83d62460905ed1a7a480789087ee261dc16b423":{"kind":"REWORK","_number":1,"created":"2025-08-30 06:08:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/39/958939/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/39/958939/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/39/958939/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/39/958939/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/39/958939/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/39/958939/1"}}},"commit":{"parents":[{"commit":"1cf23c718015238ce25da7d57fc74be828283f0a","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1cf23c718015238ce25da7d57fc74be828283f0a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-07-10 11:08:05.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-08-30 06:08:42.000000000","tz":0},"subject":"Add default values to environment lookups","message":"Add default values to environment lookups\n\nSome deployments may not have an expected env vars defined,\nwhich may lead to unexpected failures.\n\nIn order to prevent that, let\u0027s add defaults to env lookup\n\nCloses-Bug: #2116312\nChange-Id: I358d11a7a0b82b2ed9c6aeaf4650a17e878fe25d\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n(cherry picked from commit 0bee2bbd64a15c2869dbaab5de041962fea895c5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e83d62460905ed1a7a480789087ee261dc16b423"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e83d62460905ed1a7a480789087ee261dc16b423"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
