)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~959945","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~Ie482c2f58e3f72cf10f03b0d8d27b5ebde5ab9f6","project":"openstack/openstack-ansible-os_octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-09-16 07:53:36.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-09-16 10:02:12.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-09-12 08:03:55.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"Ie482c2f58e3f72cf10f03b0d8d27b5ebde5ab9f6","subject":"Remove deprecated run_tests/vagrantfile","status":"MERGED","created":"2025-09-06 12:53:43.000000000","updated":"2025-09-16 10:02:58.000000000","submitted":"2025-09-16 10:02:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959945","meta_rev_id":"77c1da338829a91f64e87e4f674c4234d41a3aa3","_number":959945,"virtual_id_number":959945,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-16 10:02:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-09-16 07:53:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-12 08:03:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2025-09-16 07:53:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-06 12:54:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-09-06 12:54:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-09-06 16:04:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e68c04546208181ac3a314fe8dd24d6c72ad2e89","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-06 12:53:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"590c2ab161d860f84a32db27a7bf6aa84b65198d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-06 12:54:15.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b0dd2af753dff36fb3d316cfd0dd06922091511f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-06 12:54:21.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"7025f650c35f26eb1df15dadad3cdcf6be4cd1f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-06 16:04:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9a2187d606f41e49e6332b699f6c89b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56b6e92f6a1f4b89993636d115fb5a3d : SUCCESS in 8m 31s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/4cb3824a2ec54b159d95c498ec904158 : SUCCESS in 1h 48m 26s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d4a35cc187cd4dea91b26714808b82f7 : SUCCESS in 1h 52m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8708ed2b0af4b50b706526b9da8a33f : SUCCESS in 1h 23m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ee9c7210eed486dbb969edad51b785d : SUCCESS in 1h 47m 50s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be75380f657745d2acf62b8c8e67fda3 : SUCCESS in 2h 38m 12s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/994821608fda480db59fa343f1609a17 : SUCCESS in 2h 41m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1e93766dd94478f8f2ea9f54fa5d2b9 : FAILURE in 1h 00m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe68e11b9d4141648b48f6a5c80710f2 : SUCCESS in 6m 49s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03f810c5be22489ca227e21fd5b95493 : TIMED_OUT in 3h 04m 03s","accounts_in_message":[],"_revision_number":2},{"id":"4fe26976e0e34a9e5a664d83a80f0e36c4603ca8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-10 23:30:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"740152470c27012f02fdcc446cfe354b5d8d6de7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 02:11:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/354a50bf02a24640807ec87f9fa25fe5\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5fba634fc61491bb6d26b41399022a7 : SUCCESS in 9m 21s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/0680430837ae4f79bcbfa14d7b5214ae : SUCCESS in 1h 21m 36s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/018fcc665ee441319a473e6a55b102d5 : SUCCESS in 1h 42m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33cdb4750884459eaa755c8aee11cc32 : SUCCESS in 52m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a6bab7821ba486eaa54d5eee5edabe0 : SUCCESS in 1h 54m 09s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5864df80b594ab08307faba7722386e : SUCCESS in 1h 58m 24s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc22b80320374f24ba96ec95672e5f9d : SUCCESS in 2h 34m 46s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc3507cb00e84d7381c60e70347c2bb0 : FAILURE in 45m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e114347a1d5453da8252892887a9580 : SUCCESS in 4m 59s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8eadfaccc06c462286b5adacad587374 : SUCCESS in 1h 41m 19s","accounts_in_message":[],"_revision_number":2},{"id":"f09d915a2fdbfbc8f5c012bffa168a9389b30b1d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-09-12 08:03:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"87496438c3a31b127ca9fd839f96182a6993ca6d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-16 07:53:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f83d6aba24e06f2334efb882b8f583870436252c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 07:55:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"916a1b4ef7307ede1eedc941ae55ec66aa9ce58a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 10:02:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bae70733ed3a41ddaadfeb85b254acec\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/064db6b1c0904fbe991fbf7d5f1773fa : SUCCESS in 9m 02s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/25f6df0e519947d1a30fa0a87a58665c : SUCCESS in 1h 38m 27s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ecc4d6e6a9a54b8097930703742dddaa : SUCCESS in 1h 35m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5008a3396114f9ca2fa9309a4e15275 : SUCCESS in 1h 38m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a5f4c1dbcfa4cab84c1a026e091d073 : SUCCESS in 1h 39m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b5d9db4f55e41b2af620bc5485bde5f : SUCCESS in 5m 48s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d68953f303654d2e99acc090cc92f5f7 : SUCCESS in 2h 05m 03s","accounts_in_message":[],"_revision_number":2},{"id":"606a9134a0962821b5705edf35ca39eb43b4e3b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 10:02:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"77c1da338829a91f64e87e4f674c4234d41a3aa3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 10:02:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6f623aff8a648eca7095e8e52637dc2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b439f3584e44426a4c6927852cbb60e : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1a622da3efb47115b334defbbc3e6973f650a230","revisions":{"b5a2d3ffc635e1d128dcb7f3a782b8cd9bc02698":{"kind":"REWORK","_number":1,"created":"2025-09-06 12:53:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/959945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/45/959945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/45/959945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/45/959945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/45/959945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/45/959945/1"}}},"commit":{"parents":[{"commit":"2c9fd178f43907aa9c83264903df5fff80715505","subject":"Switch amphora url to noble","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2c9fd178f43907aa9c83264903df5fff80715505"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-06 12:53:43.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-06 12:53:43.000000000","tz":0},"subject":"Remove deprecated run_tests/vagrantfile","message":"Remove deprecated run_tests/vagrantfile\n\nrun_tests.sh was part of functional testing framwork which was\ndeprecated.\nVagrantfile while running integrated tests, does not contain\nany supported distro\nand effectively unmaintained.\n\nChange-Id: Ie482c2f58e3f72cf10f03b0d8d27b5ebde5ab9f6\nCo-authored-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b5a2d3ffc635e1d128dcb7f3a782b8cd9bc02698"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b5a2d3ffc635e1d128dcb7f3a782b8cd9bc02698"}]},"branch":"refs/heads/master"},"1a622da3efb47115b334defbbc3e6973f650a230":{"kind":"REWORK","_number":2,"created":"2025-09-06 12:54:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/959945/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/45/959945/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/45/959945/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/45/959945/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/45/959945/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/45/959945/2"}}},"commit":{"parents":[{"commit":"2c9fd178f43907aa9c83264903df5fff80715505","subject":"Switch amphora url to noble","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2c9fd178f43907aa9c83264903df5fff80715505"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-06 12:53:43.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-06 12:54:12.000000000","tz":0},"subject":"Remove deprecated run_tests/vagrantfile","message":"Remove deprecated run_tests/vagrantfile\n\nrun_tests.sh was part of functional testing framwork which was\ndeprecated.\nVagrantfile while running integrated tests, does not contain\nany supported distro\nand effectively unmaintained.\n\nChange-Id: Ie482c2f58e3f72cf10f03b0d8d27b5ebde5ab9f6\nCo-authored-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1a622da3efb47115b334defbbc3e6973f650a230"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1a622da3efb47115b334defbbc3e6973f650a230"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
