)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~978680","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"osa/invalid_group_chars","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-04-11 06:47:24.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-04-11 15:59:48.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-04-11 15:59:48.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-04-11 15:59:48.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-04-07 21:07:54.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-04-10 21:18:36.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0","subject":"Rename Octavia groups to remove dashes","status":"MERGED","created":"2026-03-03 09:15:50.000000000","updated":"2026-04-11 16:00:37.000000000","submitted":"2026-04-11 15:59:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978680-osa/invalid_group_chars","meta_rev_id":"167bcbac6947be8825995eeb1b75ba453288d2fa","_number":978680,"virtual_id_number":978680,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-11 15:59:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-04-10 17:23:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-09 08:13:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-04-10 21:18:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-04-10 21:18:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-03 09:47:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-05 13:08:25.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"CC"},{"updated":"2026-03-17 09:12:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-03-19 12:54:33.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-03-30 00:27:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-03-30 00:27:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-03-30 00:27:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-03-30 00:27:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-04-10 17:23:47.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"e87c1075a82fbbadfcae6d0597963e21292fff9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-03 09:15:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcb794b6b7488a4daa50b23ced88b72e058ad487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 09:47:44.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-ansible/+/978568 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"a59a4bc41b4583e6638fabe268f16ec65d53e845","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-03 09:48:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8895bc1acfb8d7d4c0de1f968dd42e19d1b7a7e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 11:03:09.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-ansible/+/978568 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"170f347ba51f49965e15322c5fdd0be5822b7438","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-03 11:09:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3239264d142634ce488c279f55e6ac819f79e7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 12:55:31.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-ansible/+/978568 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"694b17e69d21465e1d0c6ecf42b3c20a7acf40dd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-03 12:58:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29cfc8718a7ec9a8ea9bf35fe91d2920fb434ad2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 14:42:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f4796a20ee04719b7f5b5881901e54b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2959293c305249dfa9fcc9b25cc5393e : SUCCESS in 17m 07s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/117e14fb78a54e008dfc9516cb3198f6 : FAILURE in 1h 28m 20s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d6300e16cb8041ceb619eff75091c9f7 : FAILURE in 1h 19m 02s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/ae3b9a407adc406d834dc5857f49c804 : FAILURE in 1h 23m 49s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3900f43447ee4dd1a22dbfe77dc0f49a : FAILURE in 1h 13m 44s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/73eb495f1a2546d2a03d8e1fce3d3fc2 : FAILURE in 1h 27m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1d5808dd47d4fdfad526b4a5fd31455 : FAILURE in 1h 11m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13c3f6240f02487fa4d644e2d5218f0e : FAILURE in 1h 09m 19s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3a156cf422164c9499296effabe5351e : FAILURE in 1h 23m 49s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5703d06958ba4ab182863bd5f4829709 : FAILURE in 31m 05s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e8e372318cb4c0495c36f3096d8a799 : FAILURE in 28m 17s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21f1522ed1f842c5b2a02042f1065c19 : FAILURE in 27m 13s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a54108ed168497280e3005cabe31bc3 : FAILURE in 41m 18s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/aa9a6348597647aea4950fca12b06fb2 : FAILURE in 9m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f3f30ca5aed4151919414bbd3023d22 : FAILURE in 29m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ce5b47888c2473a93c56cde05445820 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d52094f3a4646708d98e6c41be2a7c2 : SUCCESS in 3m 43s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27bf0b0df564459995333a28b35d2cd4 : FAILURE in 54m 42s","accounts_in_message":[],"_revision_number":1},{"id":"cddd076f5ee8b2b4c5c75fcfa9a674a176a3c25c","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-03-05 13:08:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41fd9c9e4fc9ad7e68b6a2484a5d7c50d049a79f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 15:15:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de76c4b110474a0d9e02359c3d081726\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d12f3f3e3d3b49feacaebf11b8610568 : SUCCESS in 12m 55s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/7c2e3bed9a824b2e90281b7a9bac0559 : FAILURE in 45m 39s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/68f86cde09b14d0fb7ab72efc95c5764 : FAILURE in 1h 09m 42s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/728ec7b003f74210abc4be05570dd0d5 : FAILURE in 1h 22m 55s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d9db5b1c42dc48b9b2eec9fae264ed7a : FAILURE in 55m 02s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/743159a4f5914b9bbf7dfd864e46fd79 : FAILURE in 55m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bb315ea49ea4cccaa475a74e859c1ff : FAILURE in 1h 26m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c79ae3f56044603994783912d72d2bc : FAILURE in 1h 24m 43s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/29d469879d5b45a696b8142df0ba6cf0 : FAILURE in 1h 25m 56s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d821f849d6bc40a096c8dddd4d4a4f28 : FAILURE in 1h 51m 12s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18f03117ba794af1ab72082647eb6d4b : FAILURE in 1h 52m 21s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3591670508754a32ac378b19aa46633c : FAILURE in 2h 03m 05s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c742e32f0b004a009d2ff9afd1eee6da : FAILURE in 1h 29m 29s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0fb535478737496dbac2c5fe7c132847 : FAILURE in 14m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b5fb8d0176a4d46949f4547cecb52a3 : FAILURE in 23m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ae5cb815204439cb9f0e42171c6c2db : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ad2d3d465a2440ea99fd04666cc8afb : SUCCESS in 3m 41s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/755b170f65394982b06ac918c2c319d2 : FAILURE in 1h 38m 23s","accounts_in_message":[],"_revision_number":1},{"id":"56353e4f195412d2aaf23c3a351c8aeb81ba2429","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-06 19:20:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5424de13f161ea571a02379c5f0930ae3d44fe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 21:53:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71bb7aa532bc44dab869218e90cd5219\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4e92bec56c64874a33012c59bcd37d8 : SUCCESS in 12m 36s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/e073edeae64a44c5b2babf82d6963b58 : FAILURE in 44m 22s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ded0b404de2a4bff8709ba55821cee12 : FAILURE in 48m 53s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/9bfdcf7e819748adb9e5d20747c7acea : FAILURE in 1h 20m 47s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a869ad8995da41cfb094566f6c6a5c19 : FAILURE in 1h 10m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4913dc8a4df34a13b044a2a6508fca61 : FAILURE in 1h 13m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60e895d3914f43eab97bdb096f9f335c : FAILURE in 1h 25m 36s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb66460097fc4054abba52eeb2359bbc : FAILURE in 52m 42s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a20e6251797140f5934ba662bfa0e955 : FAILURE in 2h 26m 49s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a42e3756e33b40caa902e7c119b275d6 : FAILURE in 1h 15m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f5b95eb7a5849f7adec312040047032 : FAILURE in 2h 26m 33s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d2ec6838c15411d828c1746b409eb80 : FAILURE in 2h 20m 27s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/307afcd315c54cc6a95e26f52672cc5c : FAILURE in 1h 45m 04s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d4d47a70665548fba296b3967b26db9f : FAILURE in 10m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e910e37c234c4201a5a98156281310e9 : FAILURE in 28m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab22b161f3734aab963435564f73f49a : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/591e13057c9d400084f76c03933445ab : SUCCESS in 3m 20s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc3386cd54134f6186b9d7f5e2a6ff8e : FAILURE in 2h 12m 20s","accounts_in_message":[],"_revision_number":1},{"id":"50d285a5ddf7bf58561efcf9f7de1e04fcbe8a6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-11 09:38:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0463f8f07c46eed9e869f965241ea6745a528a0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 12:18:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/acab031c094c4e23ba7378fd367cda4b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da9ad00dd8b04c5cab5f402fe3a12f30 : SUCCESS in 14m 18s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/30773938a993482cbb186b8617c617ef : FAILURE in 1h 28m 56s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/bc90396f80404e6fb6d1a65a80ae1490 : FAILURE in 1h 19m 02s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/05a3eadede084920bacf16c1937e89dd : FAILURE in 1h 21m 17s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0633c6d82db34caf8936e13223029a3f : FAILURE in 47m 42s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/76d6757657114c0fb2cc3217a53d0474 : FAILURE in 1h 32m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/886932b17c0d4b39abf46b38ae332d98 : FAILURE in 44m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23117d3039084881843ac3ea712bd4e3 : FAILURE in 46m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/851c770c8a2a4b6b9a0c3f3eca0f6a85 : FAILURE in 2h 26m 47s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6c45f38af5214e91ac88896c3eacd2b0 : FAILURE in 1h 10m 30s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9c322f17bae4300ab2a6773e2f2eeeb : FAILURE in 1h 49m 17s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebbf89faed2f4222ae06b8fdbd8d2199 : FAILURE in 1h 15m 26s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/812566b2f2804caf8f219bc1b7cc1799 : FAILURE in 2h 29m 12s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dbb87befa9cb4017abf214fb5683ffb8 : FAILURE in 15m 56s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b2cb1ce317048a4baa4522559d4f6e3 : FAILURE in 1h 13m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e1399f8d36541b2ab36e89d475fd361 : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90656ce073c646b5a4c2d766d5651b88 : SUCCESS in 4m 28s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f668bb8447444e5adbd7541bc9ea3b0 : FAILURE in 1h 40m 08s","accounts_in_message":[],"_revision_number":2},{"id":"d2ce90d3d075b66f306ded7fbc716a639a1f5593","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-11 22:33:24.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6d545ce9376a5007710e13721d934447f04da608","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-11 22:33:27.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ae9b83091ddb18a5555820656be35e9351640989","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-11 22:35:25.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"0c122e0b284bb2a800f613a7945e2a9c9afe604b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 01:01:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb941a364d1440a09524704b28c22ae5\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33ffd70f96da4cc8877f373024a00dea : SUCCESS in 12m 02s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/1d07a9fd3bba4eb1a4de56299b8ff9de : FAILURE in 44m 54s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c3b2b403f13e4b4bbbd75063c494bafa : FAILURE in 1h 01m 28s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/78abcd6f21bf492590ad2f2ed01da8f1 : FAILURE in 49m 15s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/623efd2073bb4ef98c97eef4f1179f8e : FAILURE in 1h 27m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1979f30dceb14357b40d4b3adba64e5d : FAILURE in 1h 18m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/232a0140f4db4434ad11cbab7ab12c32 : FAILURE in 1h 19m 50s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7e0ddaf3c3842aa81bc0eb5a527e0cc : FAILURE in 1h 23m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/42f5c48ab7a94248afe41969291da943 : FAILURE in 45m 09s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1aabf298d22e478bb166635059c95961 : FAILURE in 1h 10m 35s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb392adcfee44bdcb03359be1f713555 : FAILURE in 2h 18m 03s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4022bc22faa4de8b37ee9a531bfa93c : FAILURE in 49m 42s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26fd4b42bd5446b2b1de53ddd8b1d829 : FAILURE in 1h 13m 48s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d6910e7771eb4ba391fa56f91143d315 : FAILURE in 14m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a727dec0ee24dda98d3bf66ab3c6ecb : FAILURE in 40m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95785023e06d42b7b69dd643ea47bdb1 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/437c085d9f1e48e7a13ab1d67ba5f18f : SUCCESS in 4m 01s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17d587a6dfd14291aaefbd08f28ac0ee : FAILURE in 1h 49m 21s","accounts_in_message":[],"_revision_number":5},{"id":"2175e6554fda8fc0a1a90c216848cec6053f1f28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 01:01:55.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb941a364d1440a09524704b28c22ae5\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33ffd70f96da4cc8877f373024a00dea : SUCCESS in 12m 02s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/1d07a9fd3bba4eb1a4de56299b8ff9de : FAILURE in 44m 54s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c3b2b403f13e4b4bbbd75063c494bafa : FAILURE in 1h 01m 28s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/78abcd6f21bf492590ad2f2ed01da8f1 : FAILURE in 49m 15s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/623efd2073bb4ef98c97eef4f1179f8e : FAILURE in 1h 27m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1979f30dceb14357b40d4b3adba64e5d : FAILURE in 1h 18m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/232a0140f4db4434ad11cbab7ab12c32 : FAILURE in 1h 19m 50s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7e0ddaf3c3842aa81bc0eb5a527e0cc : FAILURE in 1h 23m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/42f5c48ab7a94248afe41969291da943 : FAILURE in 45m 09s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1aabf298d22e478bb166635059c95961 : FAILURE in 1h 10m 35s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb392adcfee44bdcb03359be1f713555 : FAILURE in 2h 18m 03s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4022bc22faa4de8b37ee9a531bfa93c : FAILURE in 49m 42s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26fd4b42bd5446b2b1de53ddd8b1d829 : FAILURE in 1h 13m 48s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d6910e7771eb4ba391fa56f91143d315 : FAILURE in 14m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a727dec0ee24dda98d3bf66ab3c6ecb : FAILURE in 40m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95785023e06d42b7b69dd643ea47bdb1 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/437c085d9f1e48e7a13ab1d67ba5f18f : SUCCESS in 4m 01s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17d587a6dfd14291aaefbd08f28ac0ee : FAILURE in 1h 49m 21s","accounts_in_message":[],"_revision_number":5},{"id":"40f2ed8370572df70401e65944f2907cb7d5daf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-16 13:28:45.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"2a89c1053d06a83d5b129cb01251f8eaf8d94bd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 16:18:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df79b61f29304275b935c0a442542aaa\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee6ae821a64e40ba86e66926074dc56e : SUCCESS in 16m 09s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/41ee9520d4cf4332a9d5d3daac36cfb4 : FAILURE in 52m 57s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/360dceb34669426db0572c2d756f2bb0 : FAILURE in 1h 25m 15s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/096a6824be4b40709ebafe773ea3a55b : FAILURE in 1h 17m 43s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9b0564dcbf114cd2b1c6a7138b3ec5b7 : FAILURE in 58m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7cf7b60ee6df4d108ee43914cf004ea1 : FAILURE in 1h 25m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99479b14b3dc4d9f8e7a689941b41668 : FAILURE in 1h 25m 23s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69e74293f8c14a008000f4d8aa28ba3c : FAILURE in 1h 24m 49s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2e49f5634e5341d59e1433abc8ceda5f : FAILURE in 52m 47s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f88f241c94ac4a7ca2fd796d7da32988 : FAILURE in 1h 22m 21s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d41fdee744644447a61e30a1dc32eb85 : FAILURE in 1h 19m 55s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88b4044cef4f467ca0fd3477dc10cfa7 : FAILURE in 1h 11m 02s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4ddfc9471d0462fb4d58756dd8513d8 : FAILURE in 1h 31m 31s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9c76f3fbb6f24329a6b8a6aa4fc0b272 : FAILURE in 15m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b60ac1e97e6e4421996f0bf9c5a943a0 : FAILURE in 1h 09m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0559f9db0f034c11817542fa358ff4ca : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1621144292ac42a9abfd76e0b2513bc5 : SUCCESS in 3m 56s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8f636acfa5b476a84ddb67d51004e6d : FAILURE in 2h 24m 48s","accounts_in_message":[],"_revision_number":6},{"id":"c7567afca476f8e635c4790405b7cd8e45c6ece6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-16 16:42:21.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"90829b6239bcfd741808d78ed7d7a933dfd9c2d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 19:38:03.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/03f0972550e54494a2003d3aed0d90ea\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc899c633d574e0fa07ffcbf37393c8a : SUCCESS in 13m 21s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f4260179a22e4e3d96fd06a5175a438d : FAILURE in 1h 29m 46s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7226ad71bfa541329635bd359ed957c2 : SUCCESS in 1h 11m 17s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/7002f523b89b4c6286e5e424c47324d5 : FAILURE in 2h 08m 39s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e0fdf178b4d449eba7765a053fc7fb11 : SUCCESS in 1h 19m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c29b9db48c4d46d5a9d261b65a4fb7f9 : SUCCESS in 1h 12m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68247c4ddad6493d833fdead12fbe82f : SUCCESS in 1h 45m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/884e15da1f314b5697c4d171823faa57 : SUCCESS in 1h 22m 36s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8e5371f85cad400eb23d3041d5529e66 : SUCCESS in 2h 48m 32s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5a8ba907f1a24a72abd4d86360c82a19 : SUCCESS in 1h 29m 31s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b78337d1fb04e5c9045a6d2fda6011a : SUCCESS in 1h 49m 33s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7373c32f2eff4bd8ad96ab4033ff24f2 : SUCCESS in 2h 47m 33s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49bb34cb1ef74d2f8cb46654a7e00f8d : FAILURE in 1h 33m 13s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7de8f08564ff4144b6dfefa14a4da320 : FAILURE in 10m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36a00203c45e4d28840c762b84f4b69f : FAILURE in 1h 07m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a52b9a9410e74d1e9dd229f9afefe441 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7517c76a3dec4658899632ed043540b3 : SUCCESS in 2m 38s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3d8e9dbf0f74b788ff1a7802e3982e9 : FAILURE in 2h 39m 55s","accounts_in_message":[],"_revision_number":7},{"id":"112fb4d7d77c288b5630d6e83ca456857ad43423","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-16 21:07:13.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"e34e61cfc6933877f5bc85380b4a1611fc27a7ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-16 21:07:40.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"692f6af3b8180058d49112e986bc7a80643867ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 23:31:19.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1a027d912224519bad03625c0569c3a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6220fc15f3764803806d37ec1b9fa1a1 : SUCCESS in 7m 22s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/15ba0e858d184204974a468f51607766 : SUCCESS in 1h 04m 40s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/34a6af59eecb4601a53b7e038da96439 : FAILURE in 1h 27m 10s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/85f77073c947433f957eb77d3388601b : FAILURE in 2h 15m 41s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d9683dc8e79c43bb9995a5c4eae7081d : SUCCESS in 1h 05m 01s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e5031ee545264a0ebecf7c8601784f8a : SUCCESS in 1h 42m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3277f53323b4c1db2ea924091057fd1 : FAILURE in 1h 30m 08s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a39633b48856416290ed61e39c21b329 : FAILURE in 1h 36m 03s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5c0aa37c118b436a8d651e3c57c849cb : FAILURE in 1h 14m 12s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/418c0d8b2f20421fa0e33ce2d41be35f : FAILURE in 38m 37s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bd79b54db6a469f91e1850fba82bd1f : FAILURE in 1h 12m 44s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71d0303b12364bada0752b01f78438bc : FAILURE in 38m 40s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2b1095787a342a6b6321b110df2ee02 : FAILURE in 39m 05s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/116d3a2995804436bfc951636ad4a60a : FAILURE in 13m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f45a44873ff4ddd8d7ed3e8ceb5300b : FAILURE in 51m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59bc6a9b0e6745e386933e54f0a6a242 : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7e75c51fec8422e94d84b84339290de : SUCCESS in 2m 38s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6654db6882e143f3acde9e13a49c38a9 : FAILURE in 1h 13m 30s","accounts_in_message":[],"_revision_number":9},{"id":"a3def4a9d01f0ec0d7f3e11a921d6b4a72cba984","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-17 09:12:24.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5b04b059f79c2e81c8414ce0d412bbe527b78eda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 12:28:12.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e1d8fcfadb34e77935a27c3d0ef7fb1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecadd77fa6874fe9928d6f47adb6ca55 : SUCCESS in 9m 27s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/ef8f707fe08544608b751b4f12a5abf2 : SUCCESS in 1h 04m 46s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5f5ccc105699455a88d142aee617b66f : SUCCESS in 2h 05m 06s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/d5b0f7bd3f994bb081c26a196cae6255 : FAILURE in 2h 10m 58s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7642bec47408458d8bb69eb7faf3db3f : SUCCESS in 1h 20m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8ce80eb4b23644fba8a962dbe82365a5 : SUCCESS in 1h 17m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/529b289b83684deeb104579ffe0def01 : SUCCESS in 2h 07m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffaf4b9840cf407fbe1442daf2b0ead7 : SUCCESS in 2h 06m 38s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9886a170185d4d35a169b918c0100682 : SUCCESS in 1h 42m 01s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b3d8864d166f4e089acbab30b6c9884c : SUCCESS in 1h 59m 22s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20d44f5f52a44f6bb258d4d7e0c71d20 : SUCCESS in 3h 06m 44s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6cfd6ad02d894fedacf21971706457be : FAILURE in 2h 42m 39s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd6ff1c2a80b4e18809cb2041cff9082 : FAILURE in 2h 38m 18s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f457498a7b714993b30c0f66a74abe11 : FAILURE in 8m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88c01d70d8ea44b08088acfd8688c01a : FAILURE in 1h 03m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5405d9a28a924cb2a18f59990e5c7053 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44936d6ab00d4847906eae80b6066bf3 : SUCCESS in 2m 27s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc17fe1db72a4cd2a2d1c211c4f1dc87 : FAILURE in 2h 24m 56s","accounts_in_message":[],"_revision_number":9},{"id":"97fb1c67545a05118d667b89336c5e2b8880ae8a","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-19 12:54:33.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"c58a090fa6e5ae90b82e8e8259079a13b1748bd8","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-20 19:24:53.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7487cf79b4018cbda62ee930014845017ee0a1e2","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-20 19:25:01.000000000","message":"Removed Code-Review+1 by \u003cGERRIT_ACCOUNT_37632\u003e\n","accounts_in_message":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"_revision_number":9},{"id":"a66b5cae9955e8510ff4e0458819ba0b43cc7a04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-20 19:33:38.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"8675cac3930460c3cab52ac9b9b2b07d5ad059b3","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-20 19:33:42.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"694d7432ba2568f3ce909d030f81366ff3330d05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 22:38:45.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4e30f8c5a1945e69bca654882d3d139\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c5c11de58494e978fd7b640aef6fa60 : SUCCESS in 12m 38s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/76578c3280b34d80985ac827d8077f2a : SUCCESS in 1h 42m 21s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ac47b025bbf64e8ebff1b0dbf5d50432 : SUCCESS in 1h 48m 27s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/b0059ed8833e44849555b40c59b5428e : FAILURE in 2h 09m 33s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1bdf0e550aaa42c5a9e60b12ce93719c : SUCCESS in 1h 48m 17s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dbab4f382c9f4b2c9c3a6ebcd57eb875 : SUCCESS in 2h 14m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/124626d289324349b759dc3a17b0328f : SUCCESS in 1h 58m 37s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3c9884a62bd4c9e9367f5a810e40ffc : SUCCESS in 1h 54m 56s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d51a55b225164afcb678f025e9f8ac0f : SUCCESS in 1h 26m 03s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1a04c46d022545bca58446d0be10c456 : SUCCESS in 2h 58m 18s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c27137d83d840cd84e9ec889e67bef5 : SUCCESS in 2h 58m 12s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b0a24bb8ac24add86956a1b456e8313 : SUCCESS in 2h 35m 02s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a328785bfd1e4d7fbaa79fc423bf8131 : SUCCESS in 1h 29m 02s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b19410cdf0db4513838adf91eafc43f0 : FAILURE in 8m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/710a2088ebba465485f3844b98db3fec : FAILURE in 1h 02m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e98480218bd4792a06e4e5a0ef4085d : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d88b016015504b9f894d711b8cd0054b : SUCCESS in 3m 39s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3386ec75572c4d0986a573c290a900db : SUCCESS in 2h 28m 51s","accounts_in_message":[],"_revision_number":10},{"id":"e0525d2c550383c1273d7f7c151eaf07c59a5841","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-23 10:13:35.000000000","message":"Patch Set 11: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"3980bcd5f15300c01a8ebf18d073c2b7bf7a6849","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 13:10:55.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aeafe6d1a6744c19980049d96294b5a0\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b782e205be9a45c085f10f230b1d8fe9 : SUCCESS in 13m 43s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/586dce63f8a240f1b463cb2610ce0e0c : FAILURE in 1h 29m 06s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/111cc1c387c14a68b0ddc560d14bbbd0 : FAILURE in 1h 35m 54s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/e1fed015541d4ad0b98f22ea93511e2c : FAILURE in 1h 00m 14s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/566c04430c6d48cf8b3291a4847598cf : FAILURE in 1h 32m 08s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/66927452b117442ba814153cacb4fdac : FAILURE in 1h 38m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/324b340d5d8f4807a03ff5efac8ec24b : FAILURE in 1h 48m 51s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfc3cadfeb0c46a0aaa78c85ee96132c : FAILURE in 1h 55m 53s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bb924b29615f41b39cfe58af0725537b : FAILURE in 1h 24m 16s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/39c14ddecc924c45a03b60863e4fdc7f : FAILURE in 2h 41m 36s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/004c3999584245739c61ce34a049a43a : FAILURE in 2h 37m 34s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f60d8cbb9b7426999e07d029c40ad02 : FAILURE in 2h 39m 52s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4a9ffbcadaf4ee48d6f287ada3d1006 : FAILURE in 1h 19m 07s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ffc71aea7fbd4a00a0da719710b5f2f1 : FAILURE in 11m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a89ac1f5a7854b3799b39e4420385840 : FAILURE in 37m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f8f26e963194b7490d83f921097f7db : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86764db8248848948da3b20754b4b2af : SUCCESS in 3m 11s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23e986853f364c64946a0365c7b8f9a3 : SUCCESS in 2h 39m 30s","accounts_in_message":[],"_revision_number":11},{"id":"e57b997f23528d3f3bfb40bca056c142c8c45f2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-23 13:57:38.000000000","message":"Patch Set 12: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"cc7a5185364ee203cbf915148595b386f05b6fde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 17:14:14.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1acc8b7b40cb48088754bdac9191b057\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd46c02c05e44af19b4c7948f3fe25e6 : SUCCESS in 10m 49s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/5e2b99a2ed07427a8ddcb741223655ef : SUCCESS in 1h 04m 16s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/45ad1530084f466ab39ebdb1c1905fc2 : SUCCESS in 2h 10m 07s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/8f26d531f01e4f15918c5f5344bf1f8a : FAILURE in 2h 11m 49s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6adc2f2f20b34a0999289d59113c0d58 : SUCCESS in 1h 44m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6ce51f6a13bc4b91abdb5f0993d58091 : SUCCESS in 2h 03m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26ffcfe6995b4156b61458b02b5e5368 : SUCCESS in 1h 45m 16s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91f0b325997b44eb97522071f53b2469 : FAILURE in 1h 22m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5b31ecf90b8f40d69041ef26ed9f1701 : SUCCESS in 2h 54m 38s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/757ee4d5ac7148e1bc5c9b29a763155e : SUCCESS in 3h 02m 34s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb0aae93ff61491b87bb2b1954432e59 : SUCCESS in 2h 52m 49s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fae545c8c10490a815fdb3ed2df6dc1 : SUCCESS in 2h 56m 41s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ca3d866966f43b29a4d501e48e72ec9 : TIMED_OUT in 3h 08m 51s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cf86dc9f2a0642b4812e6ff3944f1a69 : FAILURE in 15m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef916bfdfbfd4b0b9ab1ddfd10068b17 : FAILURE in 1h 05m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd793eb78ffa4f5e822f786038e153ea : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f97fad6f5abe42f9a8c3448498bf755d : SUCCESS in 2m 49s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ecbf230984345129d5faaffbf874bed : SUCCESS in 2h 56m 16s","accounts_in_message":[],"_revision_number":12},{"id":"68ac2ddee2b9204c9bd05170164eeb4a22f6eef9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-23 19:51:51.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"995fb8cf5410a22f9ceb721e7d60e6b24a30c5a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 23:00:00.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4960c0e73cad498794b0a6919a4e40e5\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b70d128fac9248df9e2f7d696700049a : SUCCESS in 7m 36s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/d904e7a5373c4b51a3865b29dd8fc546 : SUCCESS in 1h 41m 29s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/fd3f69ac6a314c19bf7d9b9aac0c05aa : SUCCESS in 2h 00m 39s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/8fd6a09d31ba4dbbb00727e3d8eccced : FAILURE in 1h 53m 01s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9fe493ca28904e1db42816b2f88a1221 : SUCCESS in 1h 33m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e0677ed128774af28e6d668c2d583410 : SUCCESS in 1h 41m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63d4da3a91e044f08fd9b5e59077b842 : SUCCESS in 1h 05m 59s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9d73f0d715444769d55224ecef8410e : SUCCESS in 1h 38m 02s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1b331c05e2f2443f8ebd6cddaa38455d : SUCCESS in 1h 31m 55s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/42961db795e449b5bc3d75b7de018439 : SUCCESS in 1h 27m 56s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3f6875260ec4187b6aca474953d5047 : SUCCESS in 1h 40m 41s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a44dab50a5b748e9b51e133ec6ec6c4c : SUCCESS in 2h 55m 57s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0941a486140548dabf60adc9eb0259bd : SUCCESS in 3h 01m 08s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/af5c4f0a39a54fa8a43fba18cc6ae174 : FAILURE in 10m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b68684bc2b864e15a072f647eb3ff88b : FAILURE in 33m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2b91eadf9f8411fb2358b1017ead3a0 : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb91580c136d4c789ce685f5f7c4e669 : SUCCESS in 2m 47s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d44d0746c6514b35958c5516eec39058 : SUCCESS in 2h 41m 50s","accounts_in_message":[],"_revision_number":12},{"id":"febf2e639a6d8b1789eae9495976360fc194af74","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-24 07:44:35.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"cf5a4ff60751a9af065d2579382f51e5e7d9f901","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 10:53:38.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/185579b57c3c4dc298a24a8752e902ca\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7927cd6b76442ee88f23bc6be13da74 : SUCCESS in 8m 35s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/0379bfeb851a4907937f764bc4abf717 : SUCCESS in 1h 39m 04s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ef8ec4a1c47f45c18cf89a57fc9a1232 : SUCCESS in 2h 03m 57s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/adf036018f6f416296d0328158597b47 : FAILURE in 2h 03m 13s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/22f1df2c30a24938ba83b4874a395ca4 : SUCCESS in 1h 39m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/773aed00c3624eed8be86d0dbc29e9f7 : SUCCESS in 1h 44m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2e023eebe964388b55110ee1272b736 : SUCCESS in 2h 07m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53cf188fbc5b47ba93a1334ef82d6cbe : SUCCESS in 2h 20m 09s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e8605e36849344c08911cc6e81febe6d : SUCCESS in 2h 50m 21s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0ea50d6c5a7646f4887d8d5cff1872ba : SUCCESS in 2h 28m 19s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a56396d2ca446ec9890cc88c50789de : SUCCESS in 2h 32m 04s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e980e53600b47e0a4866e85690fe04b : SUCCESS in 2h 26m 49s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/feda1c1aa6f64c9e806c3712f9bb8ce8 : SUCCESS in 3h 01m 02s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eb58d4e66dd84704ae8e2448af8e720c : FAILURE in 9m 37s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c4cd832b5f940bf8d913ebb630bcc9d : FAILURE in 1h 01m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4881f51d3cf47dbba4baf7bb79c7f35 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59bcf036ea6c45f2bda4aaeb9af010d5 : SUCCESS in 2m 59s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8be22635396d4049af50e61378616936 : SUCCESS in 2h 06m 17s","accounts_in_message":[],"_revision_number":12},{"id":"a238f57b05f0f1eaf402bf2bdf997904d8780ce4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-24 20:31:39.000000000","message":"Patch Set 13: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"8315df263ea6bc3d2fe53af1ca4845a5fba53b97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 23:27:42.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d33a50ec62f5459aa7973efa31dbd608\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/925a2fd8516145859c3b359eb213ec6d : SUCCESS in 9m 51s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3bab157654af41fd876155433f4fbd73 : FAILURE in 1h 19m 41s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/aefd24e61925432eab17042ba21d3198 : SUCCESS in 1h 56m 29s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/ee7e57eaf8464c87a02786a8bbe825eb : FAILURE in 2h 04m 27s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bf5daa4ca70046a58a10758a058232cc : SUCCESS in 1h 29m 20s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d7efed852abb46feb5e464fbef80b57f : SUCCESS in 1h 15m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ac78e1f78b6479bbf38eada414d97a2 : SUCCESS in 2h 00m 11s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bcbb49a10b942a1ba2db62ba9be62c9 : SUCCESS in 1h 08m 51s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/efdcfcb686654b59b564ef9d83bd44e7 : SUCCESS in 1h 24m 38s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/91a4e4a255d947ce884ef145e1a63e50 : SUCCESS in 2h 52m 09s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d1c9c01e69943acbe6660123d5de678 : FAILURE in 1h 46m 29s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f54ec4b66d6e418abaf28c003eb3ca43 : SUCCESS in 1h 29m 23s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12c27d24d6e0435a8ac72ac3c264544b : SUCCESS in 1h 34m 45s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d819d472fa184a3f8a5302fa48ff7f14 : FAILURE in 10m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/905fcf8c0ebd44d8aeec98846961bc55 : FAILURE in 1h 04m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80776aace602493e971233f16f522b04 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fca82498c259465caf37056281a4ae15 : SUCCESS in 5m 26s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18f57eba22314493bea9d172989e9a91 : SUCCESS in 2h 29m 37s","accounts_in_message":[],"_revision_number":13},{"id":"b0fa3f54cbb4425892997c727fa34267653d4ed5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-25 21:22:11.000000000","message":"Patch Set 14: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"d46f7cb3205c54e6f9ff67213d0b27ba10dd83fc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-25 22:33:19.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"dc33a8415ec0f14a34b03f07af7d206cedbb1ac0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-25 22:33:43.000000000","message":"Patch Set 15: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":15},{"id":"70028f2c44db640e6182358b896cfa2cb292c9c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 00:55:52.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bf92907aad747fc8740f861472fd5a3\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1faa11de61dd4c45bdb5c01fd835aa61 : SUCCESS in 9m 41s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/27e89d5780a74f18a3bcd2eee23980fb : SUCCESS in 1h 25m 27s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e3efd680a1f6499bae72498befa7990c : SUCCESS in 1h 07m 38s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/fbea36748fd04404892c31b4a0d2a665 : FAILURE in 2h 04m 55s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0c7f917b28c143939177d7312a54b0b8 : SUCCESS in 1h 05m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c812c8bac2204a99b091084ca84b4075 : SUCCESS in 1h 08m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7143260002545728500a3badff235ff : SUCCESS in 1h 40m 45s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7347fc5e491b4ed0b71f7872fcc45fee : SUCCESS in 2h 06m 01s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a803551bbb9e47d88fd932bdd9589f25 : SUCCESS in 2h 18m 47s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/090cca1a2d2b495ea3bb87384edf59e8 : SUCCESS in 1h 28m 24s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f330319e76174bf5b3e3683b5efd8317 : SUCCESS in 1h 22m 34s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60598af14c944128880fa2fa0aa82822 : SUCCESS in 2h 14m 25s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b589309e73f4f938484f56fbf74ceff : SUCCESS in 2h 08m 50s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a35ddf7bed3649409c150ea7b3eb90e2 : FAILURE in 11m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e27fff7ebcdc41559eb788abc8a46ed1 : FAILURE in 1h 07m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cb352f890a44bea8fe77b1562b383c2 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f952791c8e5445fbc4d66951d46e4c0 : SUCCESS in 2m 40s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df9a5695817d4699b453f89315567b9b : SUCCESS in 1h 14m 57s","accounts_in_message":[],"_revision_number":15},{"id":"65e737dadd50d9638ede9b9cfe65c97fa92e7726","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-26 11:37:14.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"767243ec62dd85e58fdf667676000bbb8e2dbc15","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-26 11:37:21.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a2f34eea6c610d8c4cd2d2ae08ae062f24909ea2","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-26 11:43:00.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"049d1bc12e7e13dcdb9cf00ae4c088f8ef759054","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-29 21:01:58.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"50e25e647af2d3de4645ded524138875ed9864a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-29 21:03:21.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"3c11ec8b1d5bae84eda65f21d1dd38f7857e03cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 00:16:39.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8be6f6e144df43dbbb437badbb251f49\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24fcd3e147ec4e2f94711612c16f35b8 : SUCCESS in 10m 21s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/dfe1d29b299c4b478c7163321f1a0408 : SUCCESS in 1h 01m 01s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/73677c8be5ab43359ed85f9a09b26a6f : SUCCESS in 1h 02m 43s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/a66cfcd6b548441482aa6dcd31a4c98a : FAILURE in 1h 45m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7b87840f372e46a2844374ffc314faef : SUCCESS in 1h 04m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/31b39591198140b886ba8dd2a17bd954 : SUCCESS in 1h 37m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e7eacab25604177a377b2a53f95eb6d : SUCCESS in 1h 08m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77f0ff567ec14698870f8eac8d3db7a1 : SUCCESS in 2h 07m 26s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1735f0295aa0483baa924f843d3f7eb5 : SUCCESS in 2h 16m 58s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/686be53f66de447695d1f472209f8b31 : SUCCESS in 2h 04m 35s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c5f7813b78141c5830e1fb63f2465bc : SUCCESS in 3h 05m 31s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79cd06d2166546f7bdbba1f01ce4c4dc : SUCCESS in 1h 24m 32s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b36abe640cd40309483c823f5e07096 : SUCCESS in 2h 16m 54s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8ceb89bfdc624be09915ae2418e9b148 : FAILURE in 13m 41s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/005d240cac35481db165ae5eb1baa5ec : FAILURE in 1h 02m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a989000c4fcb42ebb3abd75074e5553d : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cb3cbe1b41244f0b5d2211ef28eb24c : SUCCESS in 2m 33s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a880db37c34a424983929868917966dc : SUCCESS in 1h 49m 12s","accounts_in_message":[],"_revision_number":17},{"id":"4e83a6f060de2a47d17a929861aed13811dea862","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-30 00:27:50.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"91bc75d1b44a1224fa6f87481d21ffda11949061","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-30 07:29:47.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"68af221ffdfac99d1204fe38660bf592b4700dc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-30 08:03:31.000000000","message":"Patch Set 18: Published edit on patch set 17.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"bbe9c2d5795298265018063e783127ed73e39f36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 11:13:37.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd382342f9384f8aa9a179fb7764e82c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31001a9394e148739b1093750c421791 : SUCCESS in 11m 31s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/60413ff5a85f4c44a22a447ccaea9f63 : SUCCESS in 1h 00m 19s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7d37e6c4de5e483391ecfcd368134cd5 : SUCCESS in 59m 23s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/8c3930f594fb426597ebc831c3b6ccca : FAILURE in 2h 11m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4b3997f648f04364b219b37e14810c72 : SUCCESS in 1h 53m 28s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/11a952f27d7c49f2a56bce6ecb5fc2d1 : FAILURE in 1h 28m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/448f7d96eccf4ea6b9938e4815e8cced : SUCCESS in 2h 11m 24s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3db7dbeacfbd4a2ca839e217405aee87 : SUCCESS in 2h 24m 12s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/41b1d9c8b73e426a938a115feb004efc : SUCCESS in 2h 36m 36s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cea9cb0a320c461e8ec40799d3e05115 : FAILURE in 1h 39m 17s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da01052211cd4a1b8f6c2229d1d540ea : SUCCESS in 2h 52m 52s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e67cb7edf9554e4f8b16d5557af4ac14 : SUCCESS in 2h 58m 37s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41d3cf3df9434b63a4a12139d20f4a90 : SUCCESS in 3h 02m 20s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/239c314d0d7d44e491adebe811bf0e43 : FAILURE in 9m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e015d23cb5d4c5689778fb9941c3ad3 : FAILURE in 1h 07m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c915485a83af48baa0f56e7371438c01 : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cb3c27ea4774b8791a8835e2377e50b : SUCCESS in 2m 43s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be9c300eaccd497f8dceca9a56b149f9 : SUCCESS in 2h 18m 44s","accounts_in_message":[],"_revision_number":18},{"id":"65b5ffc7ed13c415904291da1623e5d943b20b05","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-07 10:40:17.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"8b8862896df02f316852e14c1b67aefdc3d591a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 12:48:21.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b78caa0efce49988627cfd2bbdd4023\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9964e3bf95d14f50b792ba8835467334 : SUCCESS in 12m 59s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/55fbec4cf8f445c695bf6b36002c7845 : RETRY_LIMIT in 11m 27s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/974fb5b9b5ec4b68ba4bf48b3320832b : SUCCESS in 1h 54m 17s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/508cada9709943a69e8839875e548cd8 : RETRY_LIMIT in 14m 05s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3fcb3d0161a84b1d8eba4acf1c9040a7 : RETRY_LIMIT in 12m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c0ee378c586345d0be9b248a780dec3b : RETRY_LIMIT in 15m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/141ede61494f4648989b7e6d87cd42f2 : RETRY_LIMIT in 10m 47s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55a6478ca39445beb6def2d4225ce1c2 : RETRY_LIMIT in 10m 50s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/43f3e03cbe29471691cd2a76a5832e4c : FAILURE in 10m 56s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/53778e61130642de86962090731d9047 : FAILURE in 6m 05s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49659f009fed4588a19f0b77c8a42999 : FAILURE in 11m 44s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5984c348d91d4875825b42bd43118c99 : FAILURE in 12m 41s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b59af5a2cc5f4735ab9225201b587f99 : FAILURE in 7m 00s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/630514dd2a1d4c3e8cba6b5042fd12e8 : RETRY_LIMIT in 15m 44s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1316a66264f2452f8d8493b086c50a27 : FAILURE in 54m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ab1313108ca4dcca8bf18b5fe125e56 : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c47786bb5c1c4dc7b34c5f408a041c60 : SUCCESS in 1m 37s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4623239a9034d53a999321611048fb2 : SUCCESS in 2h 05m 00s","accounts_in_message":[],"_revision_number":18},{"id":"babe2da9cca8f59ca489f2b522943584bd19d9d8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-07 21:07:54.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"bb07c3a4e147f76ffbe57b4e6699495ab33b0ca4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 00:22:28.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/597a43fa27e842ecbc6aa946c1c91a2d\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30be24c1497a49ad973932d67882fb9f : SUCCESS in 7m 45s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/a8ebc33270494d4bbc74597249cc1a1b : SUCCESS in 1h 50m 56s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0ae7ba28ce94487ca4ae30a90700e0e6 : FAILURE in 2h 06m 11s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/96d46144d99e414f8154dac96c34200d : SUCCESS in 1h 25m 34s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1dd59bb7c76a435e9f63d15d47beda68 : SUCCESS in 1h 55m 56s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b0cc9cd3df574eda93a33c73ab199676 : SUCCESS in 1h 34m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a407e351cea4dee9ef14aa5e68b2057 : SUCCESS in 2h 10m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aeb97b127c8b4ec08fe324758c19ab0d : SUCCESS in 2h 00m 05s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c42652368050454ab165a305c90d0317 : SUCCESS in 1h 24m 59s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dced6ff46d384d6cae5f9bfa84891d98 : FAILURE in 1h 29m 48s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3757d1fdf6074f719405a2382ec9f20e : SUCCESS in 2h 30m 12s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd9c7fee82b84298b873f03a96fbfdb7 : SUCCESS in 2h 58m 12s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ccfd4968c114d6b974e32f83e26705d : SUCCESS in 3h 05m 06s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f9e29b3001b64e4f804cb0e3ca9e8cc2 : FAILURE in 10m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9365f5501aab49178b3bbf06146d6ac9 : FAILURE in 1h 07m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0a96b4d75af4be685048b6ee1da4f23 : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ed5ff9247ec442384631e3a50e1ae97 : SUCCESS in 2m 45s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66f44c94464d4fcca41b7b68099410a3 : SUCCESS in 2h 01m 36s","accounts_in_message":[],"_revision_number":18},{"id":"759c3a3532378a5828f988a8123811aac3dac69e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-08 06:50:18.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"02747fb9f99f8ebe5988a908950868973b550e80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 10:07:30.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32b59bb715374b57b383e637c91b552b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6be3e114f074e15b74d78169ca7b4b2 : SUCCESS in 6m 41s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/41637707bf9f44b492406446e796a4f6 : SUCCESS in 1h 08m 48s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/4249e65ef9dd4246a5c8204cad8f9744 : SUCCESS in 1h 55m 44s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/67f5d514c82e4a0cb5339753d9a4b5da : SUCCESS in 56m 58s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9b4aa74fc98146ae972c7497d7cfa322 : SUCCESS in 1h 04m 28s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/13a440b35dae49ac8c67329c926521a6 : SUCCESS in 1h 06m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6db1c58924ff4a439fb565b3029faaf0 : SUCCESS in 2h 02m 59s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71c3c14f3fde4200ba4ff8bbe39de7cb : SUCCESS in 2h 15m 40s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b76c63c33e284c5f86ec42475a82caff : SUCCESS in 2h 56m 42s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dc370f756cea415588f068f5f19f0127 : SUCCESS in 1h 29m 24s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7316de347ec948ff919d6c10e7038a68 : TIMED_OUT in 3h 07m 51s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8426e9af89d049c99118b84f5282cef9 : SUCCESS in 2h 56m 20s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fd30ba709ed459c820f9452139c29e3 : TIMED_OUT in 3h 09m 28s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/adb0c31e78844212bcd525319d2ad4c9 : FAILURE in 6m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fbf66e778f440079117a15a6bd6117f : FAILURE in 38m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac91ba15dbf54bde94ec29c58a145027 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd0f4116a7224178b878b775f8f53c66 : SUCCESS in 2m 40s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eba700a265a04796bf34a409d0fc2550 : SUCCESS in 1h 24m 50s","accounts_in_message":[],"_revision_number":18},{"id":"832fccdb8a3c7989f9030390274330f3e1e5937d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-08 12:00:26.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"04c95eefc408d28667e20fb3a9afd08abbed3241","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 15:10:08.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb045fcf9ef74d1ea71a678cd37add80\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14f6b2a78341476bbece432910b5643b : SUCCESS in 14m 03s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/bf9a283310824990b94e889fafce901e : SUCCESS in 1h 35m 30s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c6ae1ea1aead46129df741c97c518192 : SUCCESS in 1h 49m 20s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/ad6588bd242a4abbb0d59807d40daa48 : SUCCESS in 1h 06m 35s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0cf5bc2daac44e6e87e77e0b965e1c4c : SUCCESS in 1h 47m 28s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/14ed8fae12ff48b5954871941a38a458 : SUCCESS in 1h 42m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e22a9fc37cd14821a1a1de2cd13be469 : SUCCESS in 1h 57m 12s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/946d2c4183be49e0b9002ef6b86d536a : SUCCESS in 2h 28m 06s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/70b26eaa4369469699a1da94b8edb1e9 : FAILURE in 7m 54s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/28cb0582d5834605b7cde667aae12cee : FAILURE in 10m 33s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5b19c98b82d497aa4623c21f7a9df1a : FAILURE in 1h 10m 55s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e7b8ece53cd4dc5881ad3c528bd8c2e : FAILURE in 1h 06m 57s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3e1b0bd61254d6986902ec2d4aa5724 : FAILURE in 1h 14m 09s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d2289ff35a864546b9f519302a1c997c : FAILURE in 17m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1179f1cc31f649ee92a228c9077b1dcc : FAILURE in 1h 08m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee1f2e6168014584819e27ef7fc40548 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a69d606b4ea0459cb991df1fbd6789e4 : SUCCESS in 3m 02s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a2b69977c0548ee8827392360d63d4b : SUCCESS in 2h 42m 42s","accounts_in_message":[],"_revision_number":18},{"id":"81b30066524258ea9cfb2f66e3242ea655b6439f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-08 15:40:59.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"9070ab5dfd8da0a299d61d01f7246c8e6cfb757b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 18:59:56.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0866daf9e6ac40b7b502f47fb8c968e8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/339426d24a2c4202a7c15209f2bb9b9f : SUCCESS in 15m 07s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/6f657e8fa21b4e948f5e39fadb5373aa : SUCCESS in 1h 55m 40s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/313c26e8685b4f21a8d1e56fdd7101df : FAILURE in 1h 56m 56s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/e6097bed71b44856b9055873071945fe : SUCCESS in 59m 27s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b24f950b1a7849949aeb60a46dc72022 : SUCCESS in 2h 05m 57s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5e65102160744bf28064b2c9b6bdd9b6 : SUCCESS in 1h 02m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fd25714710f4ee1b8ba876c1d4cb4f9 : SUCCESS in 1h 57m 10s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2e9b05d083b46a0abc96eea72ad0e20 : SUCCESS in 2h 08m 52s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0688026acb4c487f8f773e770e7415e6 : FAILURE in 1h 35m 53s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8e7cde6e201746c9a718e95fc62a079e : FAILURE in 1h 08m 37s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9b85400245e43d8ac68c397641470d3 : TIMED_OUT in 3h 11m 50s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af707be4a5414727aa8e10e81c4f1bc5 : FAILURE in 1h 12m 18s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef7742afb0d84199b5acc0ef5f67076c : FAILURE in 1h 16m 21s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/68eb4a4c807c4f8ebac4f39ba299fa75 : FAILURE in 13m 25s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e18cb503a31c4bb4a436bf183df6f729 : FAILURE in 1h 07m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be228bafb2934a71af75298f2dbfa14b : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7b0cc412fd64e40823f35cb1a5547b9 : SUCCESS in 2m 44s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c750728668e4b45aa8b71fc21236403 : SUCCESS in 1h 18m 27s","accounts_in_message":[],"_revision_number":18},{"id":"134a16b973fbae63041ba06e2af8bec47900e9dd","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-09 08:13:53.000000000","message":"Patch Set 18: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"a4b7ecf84eed1f96d10aa95d044c1e6450811302","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 11:32:12.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b41ea2af8efa499495cc6bc405fad0cc\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/311480722e274db3b347b6a5c519580a : SUCCESS in 14m 16s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/4c627de7f08e4350bc64ee7f17fb52c7 : SUCCESS in 1h 19m 32s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/cd89f2575fcb40f5828e8dfeea9cdc30 : SUCCESS in 1h 51m 06s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/ac4790913bcf4bacad168fef1202ec47 : SUCCESS in 1h 53m 32s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/98e95274b1ad43a98034c8b0d8a552ab : SUCCESS in 1h 24m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1100993fd0b54737b8f54c91c5aa6939 : SUCCESS in 1h 43m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/066bf6ee196c4a7eb04f1d8832a2cf4a : SUCCESS in 2h 08m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edd75282cc1b4ad4ba011ed3a7ebeb99 : SUCCESS in 2h 07m 43s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/20c6ff463d0c4c4781a08dae3a2ae907 : SUCCESS in 2h 32m 48s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bebe527388d14213902991e33d2d0339 : SUCCESS in 2h 32m 18s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b822967b9cb045ab965f33055cecba44 : SUCCESS in 3h 03m 11s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b16155c2eb944fab5deeea1079fcae8 : SUCCESS in 2h 49m 33s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca29175d5c99402fbc7e3dfc84eaa65f : SUCCESS in 2h 54m 27s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/175e9babde944c0d83cc48a966d7cef2 : FAILURE in 11m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2cca9be4578f4334b00959120bbba7e5 : FAILURE in 55m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81415b9cfeda4f2ca1a58cc9f4ba9625 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51acba474d774cb6b5e2ddd302bb12af : SUCCESS in 3m 59s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca996fee05634848948bfca785e41f42 : SUCCESS in 2h 38m 07s","accounts_in_message":[],"_revision_number":18},{"id":"37b206abfabd7cfa056e45c663dbe9c664d7809c","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2026-04-10 17:23:47.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"19975babc23f4f8606c78fecdf35d5c1977ec5dc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-04-10 21:18:36.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"3aa78bc61606d9a9c8cf2ab7cd729421dcc7677f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 21:20:45.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"ab8660583b78379fc6e7ede65784406582cb886b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 00:08:58.000000000","message":"Patch Set 18: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a5617cbec964146a303d3e8da5a2d15\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9f7d8c55332427682c73fc0cb6eae34 : SUCCESS in 11m 40s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/5aaa2d9f0a7444c48796bc9cd72e51d9 : SUCCESS in 1h 57m 06s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/fcf1c1d8e9c4442395df6ab2870c5648 : SUCCESS in 1h 29m 48s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/3397e53ee40c416dbb0598f1a74d8b01 : SUCCESS in 59m 24s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5ad53b24b7e1461db9ff259e4e116ef1 : SUCCESS in 1h 57m 59s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0ad40d529a2c4caa816191d03c834ee4 : SUCCESS in 1h 33m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1d0d605f72b48b78dc872333fe617c9 : FAILURE in 2h 20m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6677af93136c4962a7b669bc3cc8d73f : SUCCESS in 2h 08m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c10f0d5555d4ac4b3a5acc0ef7a39ba : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e749bbdae4254f0abb9f18bcc7256811 : SUCCESS in 3m 04s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/366ba2cf60fa49c4a70242095441f400 : SUCCESS in 2h 43m 10s","accounts_in_message":[],"_revision_number":18},{"id":"5256ca16402a396ccbcca50fc1507860b6a58dcb","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-04-11 06:47:24.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"f0fec551196d313ee4c260b13a551f92b7530de5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 10:01:44.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f040b3ba8178430887841d653b8c65e5\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b291679a5ab24d59894b4a284398634b : SUCCESS in 8m 10s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f9d338befc8345a8ac745cb03212f581 : SUCCESS in 1h 54m 13s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9044085844b447228398245d7410d25b : SUCCESS in 1h 00m 18s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/3779919731164034b3bf673c381758b7 : SUCCESS in 1h 45m 15s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ede4e173530e4708908053724e886519 : SUCCESS in 1h 30m 50s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0b9e26eb446a424f9a0b8689a8d5abd9 : SUCCESS in 2h 02m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3828150a4d8b42b7ba0baef8935adf58 : SUCCESS in 1h 01m 52s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3160885f7a64884ba5a30ce5d84d365 : SUCCESS in 2h 07m 05s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/30ab4b5a298a412a90b8d0df74bceed3 : SUCCESS in 2h 10m 58s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0cf9ed9c2ff4471bb134816d49d89756 : SUCCESS in 1h 29m 42s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8af53a97cbfd4997aadd9effe4f8246a : FAILURE in 3h 03m 47s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b664e1d50664055b4bcccaf8cc60fa3 : SUCCESS in 2h 09m 53s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5c4a753e5f2455b807cbac28fc72db7 : SUCCESS in 1h 26m 27s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/27e331e98c6d43148b44b02006207f7d : FAILURE in 10m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1947f9b9fe0f4c68ad4dbd0ec2f7e991 : FAILURE in 50m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d08338f0295142c5868ec501ae3ce46f : SUCCESS in 3m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c9d36460ef74c36bcf49dfbb335c38e : SUCCESS in 1m 40s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c603a185082b475493d9453a9730ef7f : SUCCESS in 2h 58m 08s","accounts_in_message":[],"_revision_number":18},{"id":"b67b0019c21c9574f77dbe954f672471c2150eb7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-11 10:02:15.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"231ccd9e0ed676111981b8ee192e1f5e402c66b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 13:13:15.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82fcf0aa86be459991d6a71dc884e7af\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7b6401446984a1bb9b8c8ef7187b694 : SUCCESS in 8m 40s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/34206e51680248a88aa3cefa52eb8726 : SUCCESS in 1h 04m 02s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/cb7fcb69f9954714808d5a023fde85d4 : SUCCESS in 1h 52m 23s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/85c4c70baaf44c5bb4f3f188feda979f : SUCCESS in 1h 29m 54s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/06c0e17874814c70b3b172c777a359fd : SUCCESS in 58m 50s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5a56c601c5cb42fa842ec4163fafca84 : SUCCESS in 1h 38m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9cedbbf0c0747d492a70b688ad0615d : SUCCESS in 1h 01m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d2756b915334d72864129363b2cf86a : SUCCESS in 2h 00m 09s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/16b7c010803e46788e854f541dda2ebe : SUCCESS in 2h 45m 52s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d96bc06b021d491fb7e8efec77a47833 : SUCCESS in 1h 22m 08s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0dd052f48ce4690a1188ed3e4b16657 : SUCCESS in 2h 47m 38s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95ecbd6dc074408790d62ab5eb2c9caa : SUCCESS in 3h 01m 52s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2852a7a21b6489db3499564bedf0053 : SUCCESS in 3h 03m 47s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/00812438080f488caf1bccfb4f475cf5 : FAILURE in 6m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fd3733ed9d74b819f47febbbe67b138 : FAILURE in 1h 02m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63ed0fabc08a45d488aab609ebc05db2 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef510e6a8d1b4a44b8a67f130027e829 : SUCCESS in 2m 17s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2f5015e17a34df6988ffb775e6bb218 : SUCCESS in 2h 31m 08s","accounts_in_message":[],"_revision_number":18},{"id":"45fdeff5ffcfc837fe85cd0262d0ef0fb2bb9aef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 13:17:33.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"2ba08916937f397eeb5fca71fd64865ecb689182","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 15:59:47.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0637c4fcd44942bca7b735b70a3113e6\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d085d5c9de3a41cab318aedb8d3424dd : SUCCESS in 8m 49s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/30a52480bd9046d1a718d731127aa71b : SUCCESS in 1h 03m 44s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/958235e098b54eeeab8775889b27d327 : SUCCESS in 1h 01m 06s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/6c0c4ddaf2924721ba2ca44da418f36f : SUCCESS in 1h 25m 11s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/854812a8e8fc43b9816672d9151cf29b : SUCCESS in 1h 01m 54s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8c4022f4d55a4ef392c21e080ca9c223 : SUCCESS in 1h 28m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd28ce5fc3b2465bbea70382f15a45d8 : SUCCESS in 1h 01m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81bee3ca2f474af19f10d0c2aca065d9 : SUCCESS in 2h 05m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71f097141942481184ed7989a6e153bd : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6efef5c1da240c8a3f3ec7cc0cc4e2f : SUCCESS in 3m 13s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/875fb821b4d241288e3ad5e706ea8383 : SUCCESS in 2h 37m 03s","accounts_in_message":[],"_revision_number":18},{"id":"24920d21575de289127d3130b547b0311000b83d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 15:59:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":18},{"id":"167bcbac6947be8825995eeb1b75ba453288d2fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 16:00:37.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/250ef722201a401fb855a034520b1253\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbc595b2c21e430cbd8b3b458dd2b10e : SUCCESS in 41s","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"13261f6bf14f5fc9f69b31052fd290906cad1091","revisions":{"4f7620bcba03c1f73f25c80efd7aed1dd34ec901":{"kind":"REWORK","_number":1,"created":"2026-03-03 09:15:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/80/978680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/80/978680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/1"}}},"commit":{"parents":[{"commit":"304e85fe8e3e4b8658433d9221514e811662a9c6","subject":"tox: remove linters and functional from envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/304e85fe8e3e4b8658433d9221514e811662a9c6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"subject":"Rename Octavia groups to remove dashes","message":"Rename Octavia groups to remove dashes\n\nIn order to align with other service group naming and avoid using\ninvalid chars in ansible group names, we replace dashes with underscores\nfor Octavia groups.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/978568\nChange-Id: Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4f7620bcba03c1f73f25c80efd7aed1dd34ec901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4f7620bcba03c1f73f25c80efd7aed1dd34ec901"}]},"branch":"refs/heads/master"},"465288d071d93b2ba9196c9c181fe4ab9a806c7c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-11 09:38:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/80/978680/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/80/978680/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/2"}}},"commit":{"parents":[{"commit":"fb6b794ebbb4e46e14e035a32675d33592ca3f7d","subject":"Remove unused Makefile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/fb6b794ebbb4e46e14e035a32675d33592ca3f7d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-11 09:38:30.000000000","tz":0},"subject":"Rename Octavia groups to remove dashes","message":"Rename Octavia groups to remove dashes\n\nIn order to align with other service group naming and avoid using\ninvalid chars in ansible group names, we replace dashes with underscores\nfor Octavia groups.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/978568\nChange-Id: Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/465288d071d93b2ba9196c9c181fe4ab9a806c7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/465288d071d93b2ba9196c9c181fe4ab9a806c7c"}]},"branch":"refs/heads/master","description":"Rebase"},"1a161c0509f4b0e538465cebf82df67c70147c2f":{"kind":"REWORK","_number":3,"created":"2026-03-11 22:33:24.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/978680/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/80/978680/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/3"}}},"commit":{"parents":[{"commit":"fb6b794ebbb4e46e14e035a32675d33592ca3f7d","subject":"Remove unused Makefile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/fb6b794ebbb4e46e14e035a32675d33592ca3f7d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-11 22:33:22.000000000","tz":0},"subject":"Rename Octavia groups to remove dashes","message":"Rename Octavia groups to remove dashes\n\nIn order to align with other service group naming and avoid using\ninvalid chars in ansible group names, we replace dashes with underscores\nfor Octavia groups.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/978568\nChange-Id: Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1a161c0509f4b0e538465cebf82df67c70147c2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1a161c0509f4b0e538465cebf82df67c70147c2f"}]},"branch":"refs/heads/master"},"bce6bb1418a26feffa61f310915b19ac43bdfc62":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-03-11 22:33:27.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/978680/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/80/978680/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/4"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-11 22:33:27.000000000","tz":0},"subject":"Rename Octavia groups to remove dashes","message":"Rename Octavia groups to remove dashes\n\nIn order to align with other service group naming and avoid using\ninvalid chars in ansible group names, we replace dashes with underscores\nfor Octavia groups.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/978568\nChange-Id: Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/bce6bb1418a26feffa61f310915b19ac43bdfc62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/bce6bb1418a26feffa61f310915b19ac43bdfc62"}]},"branch":"refs/heads/master","description":"Rebase"},"742be6aee35071b178a98973f65935e9c039b09c":{"kind":"REWORK","_number":5,"created":"2026-03-11 22:35:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/978680/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/80/978680/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/5"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-11 22:35:25.000000000","tz":0},"subject":"Rename Octavia groups to remove dashes","message":"Rename Octavia groups to remove dashes\n\nIn order to align with other service group naming and avoid using\ninvalid chars in ansible group names, we replace dashes with underscores\nfor Octavia groups.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/978568\nChange-Id: Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/742be6aee35071b178a98973f65935e9c039b09c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/742be6aee35071b178a98973f65935e9c039b09c"}]},"branch":"refs/heads/master"},"51fcf508c904c37bc9b9a2f6ac27ca96a1f35ab3":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-03-16 13:28:45.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/80/978680/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/80/978680/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/6"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-16 13:28:45.000000000","tz":0},"subject":"Rename Octavia groups to remove dashes","message":"Rename Octavia groups to remove dashes\n\nIn order to align with other service group naming and avoid using\ninvalid chars in ansible group names, we replace dashes with underscores\nfor Octavia groups.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/978568\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_octavia/+/978864\n\nChange-Id: Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/51fcf508c904c37bc9b9a2f6ac27ca96a1f35ab3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/51fcf508c904c37bc9b9a2f6ac27ca96a1f35ab3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"332d571348af37d80761f22f62c3e8ccc7423c9a":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-03-16 16:42:21.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/978680/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/80/978680/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/7"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-16 16:42:21.000000000","tz":0},"subject":"Rename Octavia groups to remove dashes","message":"Rename Octavia groups to remove dashes\n\nIn order to align with other service group naming and avoid using\ninvalid chars in ansible group names, we replace dashes with underscores\nfor Octavia groups.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_octavia/+/978864\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_octavia/+/979028\nChange-Id: Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/332d571348af37d80761f22f62c3e8ccc7423c9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/332d571348af37d80761f22f62c3e8ccc7423c9a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5dffe06df4843ca6a2904b18f2923333f3a577f7":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2026-03-16 21:07:13.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/978680/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/80/978680/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/8"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-16 21:07:13.000000000","tz":0},"subject":"Rename Octavia groups to remove dashes","message":"Rename Octavia groups to remove dashes\n\nIn order to align with other service group naming and avoid using\ninvalid chars in ansible group names, we replace dashes with underscores\nfor Octavia groups.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_octavia/+/979028\nChange-Id: Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/5dffe06df4843ca6a2904b18f2923333f3a577f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/5dffe06df4843ca6a2904b18f2923333f3a577f7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2283f2f64d6c4077eae2217f58a9a887d6e29386":{"kind":"REWORK","_number":9,"created":"2026-03-16 21:07:40.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/978680/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/80/978680/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/9"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-16 21:07:40.000000000","tz":0},"subject":"Rename Octavia groups to remove dashes","message":"Rename Octavia groups to remove dashes\n\nIn order to align with other service group naming and avoid using\ninvalid chars in ansible group names, we replace dashes with underscores\nfor Octavia groups.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_octavia/+/979028\nChange-Id: Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2283f2f64d6c4077eae2217f58a9a887d6e29386"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2283f2f64d6c4077eae2217f58a9a887d6e29386"}]},"branch":"refs/heads/master"},"ffd9062a01def8b216f2b15b1097d6b6fc1a2c6d":{"kind":"REWORK","_number":10,"created":"2026-03-20 19:33:38.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/80/978680/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/80/978680/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/10"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-20 19:33:04.000000000","tz":0},"subject":"Rename Octavia groups to remove dashes","message":"Rename Octavia groups to remove dashes\n\nIn order to align with other service group naming and avoid using\ninvalid chars in ansible group names, we replace dashes with underscores\nfor Octavia groups.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_octavia/+/979028\nChange-Id: Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ffd9062a01def8b216f2b15b1097d6b6fc1a2c6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ffd9062a01def8b216f2b15b1097d6b6fc1a2c6d"}]},"branch":"refs/heads/master"},"19e763c67a9221eb0316c5c5e3486a1486aff3a6":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2026-03-23 10:13:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/80/978680/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/80/978680/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/11"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-03-23 10:13:35.000000000","tz":0},"subject":"Rename Octavia groups to remove dashes","message":"Rename Octavia groups to remove dashes\n\nIn order to align with other service group naming and avoid using\ninvalid chars in ansible group names, we replace dashes with underscores\nfor Octavia groups.\n\nChange-Id: Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/19e763c67a9221eb0316c5c5e3486a1486aff3a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/19e763c67a9221eb0316c5c5e3486a1486aff3a6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f5d84df0ed4fded19edfb78e4b1ee406ef1a2494":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2026-03-23 13:57:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/80/978680/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/80/978680/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/12"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-03-23 13:57:38.000000000","tz":0},"subject":"Rename Octavia groups to remove dashes","message":"Rename Octavia groups to remove dashes\n\nIn order to align with other service group naming and avoid using\ninvalid chars in ansible group names, we replace dashes with underscores\nfor Octavia groups.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-python_venv_build/+/981776\nChange-Id: Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f5d84df0ed4fded19edfb78e4b1ee406ef1a2494"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f5d84df0ed4fded19edfb78e4b1ee406ef1a2494"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3a2ac653bac3890ee2130da0c0a76cc15455d9db":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2026-03-24 20:31:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/978680/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/80/978680/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/13"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-24 20:31:39.000000000","tz":0},"subject":"Rename Octavia groups to remove dashes","message":"Rename Octavia groups to remove dashes\n\nIn order to align with other service group naming and avoid using\ninvalid chars in ansible group names, we replace dashes with underscores\nfor Octavia groups.\n\nChange-Id: Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3a2ac653bac3890ee2130da0c0a76cc15455d9db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3a2ac653bac3890ee2130da0c0a76cc15455d9db"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3fc008262d3ff73ba10442c6fe8d6334e3b075ee":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2026-03-25 21:22:11.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/978680/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/80/978680/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/14"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-25 21:22:11.000000000","tz":0},"subject":"Rename Octavia groups to remove dashes","message":"Rename Octavia groups to remove dashes\n\nIn order to align with other service group naming and avoid using\ninvalid chars in ansible group names, we replace dashes with underscores\nfor Octavia groups.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/964549\nChange-Id: Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3fc008262d3ff73ba10442c6fe8d6334e3b075ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3fc008262d3ff73ba10442c6fe8d6334e3b075ee"}]},"branch":"refs/heads/master","description":"Edit commit message"},"fbc197f51fe28d407c9e8212657fd4580457efc1":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2026-03-25 22:33:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/80/978680/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/80/978680/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/15"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-03-25 22:33:43.000000000","tz":0},"subject":"Rename Octavia groups to remove dashes","message":"Rename Octavia groups to remove dashes\n\nIn order to align with other service group naming and avoid using\ninvalid chars in ansible group names, we replace dashes with underscores\nfor Octavia groups.\n\nChange-Id: Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/fbc197f51fe28d407c9e8212657fd4580457efc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/fbc197f51fe28d407c9e8212657fd4580457efc1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3f5052d6eced241a912ccc39df2206f8203fe218":{"kind":"REWORK","_number":16,"created":"2026-03-29 21:01:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/978680/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/80/978680/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/16"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-29 21:01:58.000000000","tz":0},"subject":"Rename Octavia groups to remove dashes","message":"Rename Octavia groups to remove dashes\n\nIn order to align with other service group naming and avoid using\ninvalid chars in ansible group names, we replace dashes with underscores\nfor Octavia groups.\n\nChange-Id: Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3f5052d6eced241a912ccc39df2206f8203fe218"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3f5052d6eced241a912ccc39df2206f8203fe218"}]},"branch":"refs/heads/master"},"0e30457732ac981aaa2a6d1a8c0665c0502280f1":{"kind":"REWORK","_number":17,"created":"2026-03-29 21:03:21.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/978680/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/80/978680/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/17"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-29 21:03:21.000000000","tz":0},"subject":"Rename Octavia groups to remove dashes","message":"Rename Octavia groups to remove dashes\n\nIn order to align with other service group naming and avoid using\ninvalid chars in ansible group names, we replace dashes with underscores\nfor Octavia groups.\n\nChange-Id: Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/0e30457732ac981aaa2a6d1a8c0665c0502280f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/0e30457732ac981aaa2a6d1a8c0665c0502280f1"}]},"branch":"refs/heads/master"},"13261f6bf14f5fc9f69b31052fd290906cad1091":{"kind":"REWORK","_number":18,"created":"2026-03-30 08:03:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/80/978680/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/80/978680/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/80/978680/18"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-03 09:15:38.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-03-30 08:03:26.000000000","tz":0},"subject":"Rename Octavia groups to remove dashes","message":"Rename Octavia groups to remove dashes\n\nIn order to align with other service group naming and avoid using\ninvalid chars in ansible group names, we replace dashes with underscores\nfor Octavia groups.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/982618\nChange-Id: Ic6264f4eb4d441b3f12f4ca194d3adfd3a9c7ad0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/13261f6bf14f5fc9f69b31052fd290906cad1091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/13261f6bf14f5fc9f69b31052fd290906cad1091"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
