)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~979638","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I40ffc50e09c6c40128c90a8f0c0ac69ee6a4609f","project":"openstack/openstack-ansible-os_octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-23 22:30:36.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-03-31 17:31:51.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I40ffc50e09c6c40128c90a8f0c0ac69ee6a4609f","subject":"Add keystoneauth options to [neutron]","status":"MERGED","created":"2026-03-09 13:04:26.000000000","updated":"2026-04-23 22:31:38.000000000","submitted":"2026-04-23 22:30:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"979638","meta_rev_id":"4969df0560bd5646f2587481aaa5c18eb75e20dc","_number":979638,"virtual_id_number":979638,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-23 22:30:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-04-23 10:20:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-23 19:41:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-04-23 19:42:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-04-23 19:42:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 13:15:50.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-03-09 16:24:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-23 19:41:38.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-04-23 19:42:16.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"d5ea4cc928150283476ee4fc01183a30d34959cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-09 13:04:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1906ad7f419fd29d7a26034f8f426b6093f5423f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-09 13:15:50.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4b6221f6ff25c9732c1f7dc02536ef1a6e47a3bf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-09 13:15:57.000000000","message":"Patch Set 1: -Backport-Candidate","accounts_in_message":[],"_revision_number":1},{"id":"3687cdb15a5377a70047efc8f5580d82783abd06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 16:24:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d0bcea76b3246e2bd3792e07aabf1f1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea39618d88b346329f1eb77568c30a2b : SUCCESS in 12m 09s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/5fa53fadf1fa4596a4194328e2fd15e6 : FAILURE in 58m 57s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b93ee744bd0e4dc1aac3657c49a5d5e8 : FAILURE in 54m 32s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/f765bfcea10140da92bb8178de47e42b : FAILURE in 1h 27m 42s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1d5a7a08e7484bbd8ccd66e99dd295a2 : FAILURE in 1h 27m 07s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/614af5c72be14ca4922e5860cd938ab9 : FAILURE in 1h 01m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ce51470e2794885afb464d38c7d3692 : FAILURE in 52m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f696e7d269eb462bac8e9e51c90cb23f : FAILURE in 53m 02s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b710879baf14495299d20b1d23dbfcaa : FAILURE in 2h 08m 18s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e237f3ac946243a5b0b38d317cc82ed4 : FAILURE in 1h 33m 08s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d7e2554718c44b0b41c360baec11d1e : FAILURE in 2h 10m 46s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/696388b2ca81468f86404b4baf5af89f : FAILURE in 2h 56m 55s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54429c5833d94f68869d320da5cb9cd3 : FAILURE in 3h 03m 40s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eab40028a20e4391a39354d361a031c2 : FAILURE in 15m 43s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2401eb6c71c44062a7d0ff52d22eaa5d : FAILURE in 32m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3014f333b9f0433f9182e883803f34f4 : SUCCESS in 6m 29s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7dd35acd5864457a9be2187ab77f345 : SUCCESS in 1h 21m 15s","accounts_in_message":[],"_revision_number":1},{"id":"fcc8e084907b4e3908bd0d829fc0121b9d88ae95","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 14:05:42.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ef58832604ba1f1c6de8c1e3b8e59ef744c6ae8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-31 15:35:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41b49d466728be56689b42434780392cb0867757","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 17:31:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a996a83b357c385a557752f84e73653310f898a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-23 06:26:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a90dbb43f843169d1195b7b089244112038f7649","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-23 06:30:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c1bf4212d9e08fbf4e0d39279955baf7618757c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 09:53:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/717b1642ffa545e7b6a386cd922612bc\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af6603868bc74ef1bdacd0dd008a25f9 : SUCCESS in 14m 20s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/c1cdf396f9be4e5aa371cba37bab7f26 : SUCCESS in 1h 27m 50s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/282ca66d8ac448028bd2112bd79ae43b : SUCCESS in 1h 53m 32s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/5ffad02facd74ac8964023c8b720f75b : SUCCESS in 1h 52m 32s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b059f5a8db854dd0a375db8b1227b360 : SUCCESS in 1h 04m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a91d347d9df84fbaa9280a39590091d1 : SUCCESS in 1h 05m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/030a39d2b8e14482a5547877a9224827 : SUCCESS in 2h 00m 53s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad44421effe1457a8ee598eeb1fa7a2b : SUCCESS in 2h 09m 15s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4ffba68914df44f6874229f686ce727b : SUCCESS in 2h 18m 47s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/09edf41dfa374b9aa5c59793fca7d077 : SUCCESS in 2h 04m 17s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c7c1f02ba0c4b83977e58e6123e78c5 : SUCCESS in 2h 53m 35s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0819a7a61f744365806087c8cd2e4be6 : SUCCESS in 2h 51m 11s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5af26a8aa3b413f98154dedfaec7e16 : TIMED_OUT in 3h 08m 27s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/239172e0b8394b52bfd2024b79463d10 : FAILURE in 10m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfc3198c1b254423a26aab38f6dcaee7 : FAILURE in 1h 09m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebf7ac1898b94e2f974660632a02bb78 : SUCCESS in 5m 41s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e0b6e5bbcdd43988bd1ebcbf5b76543 : SUCCESS in 2h 58m 32s","accounts_in_message":[],"_revision_number":2},{"id":"d2bc2ed423813db65e251f40ea303cf10c2c66e4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-23 10:20:04.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f8137860af681c9575d959e55f02e161e6e04838","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-23 10:20:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5128c4b38edc9fd7a11b934900160f7c08482e63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 13:31:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7ab859b4758426c96b56191bef2a36d\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd215d37dce1479cbac9ba6fb92670d5 : SUCCESS in 8m 52s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/ce5495ba8c0d4c73bf85578a47ec71c0 : SUCCESS in 1h 34m 29s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/1a9626e05f504785b269aea743fb3673 : SUCCESS in 1h 00m 53s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/1e587d15083d4a149e129f1efd49df78 : SUCCESS in 2h 05m 39s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/65b7364527ce455c935274e5598a24c5 : SUCCESS in 1h 42m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9e2ac5b60ace46bc8873266664c8802f : SUCCESS in 1h 07m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/875efa2c4cfa484bb150ccf06635ff00 : SUCCESS in 2h 02m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea5ed268bbc74ef99713f1beafda9e61 : SUCCESS in 1h 00m 52s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/15d079e53748456c8638c85919fb5112 : SUCCESS in 2h 10m 17s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d3c46414820a40ce99bc8912bf4609ba : SUCCESS in 1h 29m 50s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bdaeb25febc4681a4afa34b415e2110 : SUCCESS in 2h 59m 20s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3657fb8db8e468ba8f97475677df962 : SUCCESS in 3h 04m 29s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83c3f845b27a4ceabb7f09cafefc5e13 : SUCCESS in 2h 59m 04s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8674a88f0b3f4ec3ae0054a9d8117858 : FAILURE in 9m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce20e01e68ed4ee58bc735268894f25c : FAILURE in 39m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c392b4dd7f024c7f9121ccbaa96d9c3d : SUCCESS in 6m 12s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ba2b91053c5434690fa3824b4df84a3 : SUCCESS in 2h 57m 44s","accounts_in_message":[],"_revision_number":2},{"id":"1f23f425ee71b55818d6a217876abcf30e598190","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-23 19:41:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"44db973ec4ccc09a8c81cb3fc7f3cab749bb8363","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-04-23 19:42:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bb3dbf5ee7b45e1f60090697d017e649acd48eaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 19:44:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"095af9bcbc863f52fed58f449b74327a60c2c2ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 22:30:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa109516f31245d7a4226d30a6b0558f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a19145f528f046ae8d456454d728ba70 : SUCCESS in 15m 14s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/646076d8fa5248ef9956816431d89c41 : SUCCESS in 1h 51m 26s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/aff941671f1844a49e122937a7ab2bb1 : SUCCESS in 1h 50m 09s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/33dce7c3fff1495d8097eb14b7bc0b69 : SUCCESS in 1h 53m 46s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6516dea2da0542fb89a7d4724f92df11 : SUCCESS in 1h 11m 18s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/346a725d95a9446eacd2dd1802db88f8 : SUCCESS in 1h 08m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf06afc14a944e6b8f512028ecaedb7e : SUCCESS in 2h 04m 15s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5eae7fd12084c699d8c7289f94cd02c : SUCCESS in 1h 03m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a00839ac00c445bf9876d9230d471278 : SUCCESS in 3m 26s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d662862bcc9434ab187619cbc02fdde : SUCCESS in 2h 44m 28s","accounts_in_message":[],"_revision_number":2},{"id":"d6d8e10b9f19c3ef08237d5cdd0513bbc7d0076d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 22:30:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4969df0560bd5646f2587481aaa5c18eb75e20dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 22:31:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/738ef46fb0a04abf9cbb07956453db02\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c4e54a8953d4a658e2017fe2f1830c5 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b91cb7ae44fc91b6f0da56c831215777df3c4bfc","revisions":{"b00af2098235a430f734d1654196a0378efab2f6":{"kind":"REWORK","_number":1,"created":"2026-03-09 13:04:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/979638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/38/979638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/979638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/979638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/979638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/979638/1"}}},"commit":{"parents":[{"commit":"a08be912741392f892b47190d5aab10ac1b5c777","subject":"Drop non-existent options from service_auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a08be912741392f892b47190d5aab10ac1b5c777"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-09 13:02:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-09 13:03:57.000000000","tz":540},"subject":"Add keystoneauth options to [neutron]","message":"Add keystoneauth options to [neutron]\n\nOctavia introduced the separate keystoneauth options for the neutron\nsection during 2023.2 cycle due to replacement of neutronclient by\nopenstacksdk. Add these options before these are actually required.\n\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/866327\nChange-Id: I40ffc50e09c6c40128c90a8f0c0ac69ee6a4609f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b00af2098235a430f734d1654196a0378efab2f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b00af2098235a430f734d1654196a0378efab2f6"}]},"branch":"refs/heads/master"},"b91cb7ae44fc91b6f0da56c831215777df3c4bfc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-23 06:30:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/979638/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/38/979638/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/979638/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/979638/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/979638/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/979638/2"}}},"commit":{"parents":[{"commit":"f7bb8686dd9e90b7d35bf909a119da58cd6034ad","subject":"Merge \"Add hashi_vault pki backend support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f7bb8686dd9e90b7d35bf909a119da58cd6034ad"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-09 13:02:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-23 06:30:06.000000000","tz":0},"subject":"Add keystoneauth options to [neutron]","message":"Add keystoneauth options to [neutron]\n\nOctavia introduced the separate keystoneauth options for the neutron\nsection during 2023.2 cycle due to replacement of neutronclient by\nopenstacksdk. Add these options before these are actually required.\n\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/866327\nChange-Id: I40ffc50e09c6c40128c90a8f0c0ac69ee6a4609f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b91cb7ae44fc91b6f0da56c831215777df3c4bfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b91cb7ae44fc91b6f0da56c831215777df3c4bfc"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"b00af2098235a430f734d1654196a0378efab2f6","theirs":"f7bb8686dd9e90b7d35bf909a119da58cd6034ad","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
