)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~982412","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~stable%2F2025.2~I0148eda33790eb89502cea72be60a6ce82cdeff4","project":"openstack/openstack-ansible-os_octavia","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-03-30 19:52:00.000000000","reason":"removed on reply"},"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-04-02 03:41:36.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-04-02 03:41:36.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-03-31 14:02:05.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-04-01 21:59:39.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-04-01 08:46:06.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I0148eda33790eb89502cea72be60a6ce82cdeff4","subject":"Disable Octavia coordination temporarily due to ZooKeeper instability","status":"MERGED","created":"2026-03-27 13:00:48.000000000","updated":"2026-04-02 03:42:35.000000000","submitted":"2026-04-02 03:41:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982412","cherry_pick_of_change":981135,"cherry_pick_of_patch_set":6,"meta_rev_id":"fda52c0fa0e7370f7caf2af30262e669535fbdc5","_number":982412,"virtual_id_number":982412,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-02 03:41:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2026-03-30 19:52:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-03-31 14:02:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-01 08:46:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-01 08:46:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-27 13:00:48.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"},{"updated":"2026-03-27 15:02:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-30 19:47:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-03-30 19:47:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-03-30 19:47:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-03-30 19:47:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-03-30 19:47:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"28f4a0b98e3f11e5060b81e8ee5fabcf184f36ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-27 13:00:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"93d10da44a84cb7b7fb7bcde5922ac976857a757","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-27 13:01:24.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f2d1601c6cb98a9ad7bf4df9ab37c6a49dacc140","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 15:02:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0fd7165a80543cb990d7edda427a969\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6de18a3189d4f9da238c1cb98a26357 : SUCCESS in 12m 35s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/573a1b7286f14cef96b9a9e745866d0d : SUCCESS in 1h 05m 43s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b03b0c97e98b4cd99c3002d038d1f3d0 : SUCCESS in 1h 53m 31s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/c93425287cf649ddb5087eae9864170d : FAILURE in 1h 24m 13s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/587a98c8cc464e08a0c869ecaa2f4d52 : SUCCESS in 1h 01m 50s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6ffab99c958f494ea3c6d0b1c041ed93 : SUCCESS in 1h 02m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1f171993e0149c1a948e33eb51c4e19 : FAILURE in 1h 07m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e7bb41bc5554887b210fa247f1d8f51 : FAILURE in 1h 15m 11s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d76f6a7999b45afb2f3b11c951a49e8 : SUCCESS in 1h 28m 19s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a40c376dc1b4eaa8dda2a8d8e487bfb : SUCCESS in 1h 25m 04s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/aeee82b8d6ab4904b9211edd26d794ad : FAILURE in 17m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0918d86bf2c34f86812e4176b4556af7 : FAILURE in 41m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/571cac2383394544b207217b344c409f : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/572640e2c573451690022cf2b4f062a2 : SUCCESS in 2m 45s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6374089c2db543f19c53b5030d29ff85 : FAILURE in 1h 01m 13s","accounts_in_message":[],"_revision_number":2},{"id":"02da4fd4ca2e0a1356d249e235f4f5ce5f8d1a0e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-29 21:05:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5d6c8bd350460a73a1344a923228e919d93aeeca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 00:03:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97eab8da8621415e8a23c6b25d6ab9f9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56149eff55894e4083c3bd13da025dcd : SUCCESS in 11m 10s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/21f6a70ded8246bc9b09662de263d88e : SUCCESS in 1h 19m 34s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/02133bd79dc544c4a8a1d61b25bb78cc : SUCCESS in 1h 51m 13s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/19f157ea978b41fc9d70ccb10c49fe00 : SUCCESS in 1h 49m 58s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3481e037bc364f02a7827b6c078c25fb : SUCCESS in 1h 52m 45s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c2cd76df488c47c88197f7b67ff8dc77 : SUCCESS in 1h 33m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40903ff4493442e3bbf8e65fc764092e : FAILURE in 1h 04m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dbe44dc517f54c25bd0ebf704839c297 : FAILURE in 1h 05m 18s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5132a8c2bcf94268b2d2c27fab9aef0e : SUCCESS in 2h 37m 16s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3561fa78685e4d16b2709c5887960063 : SUCCESS in 2h 51m 55s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/59c78d3ebaae47819077d2b8261e2131 : FAILURE in 10m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60834aa54b3c4be8ad15ef3131df742d : FAILURE in 46m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8980b4a09c40464dbe095138401158ab : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53e45763af0d4d789b34a074d5820607 : SUCCESS in 2m 36s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2091fded1f9045e18f2e6c99997276b7 : FAILURE in 1h 47m 21s","accounts_in_message":[],"_revision_number":2},{"id":"2278d5ff9e24a155b5bf42355928f0d666bc233c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-30 16:54:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"16a1d38ff2638262764ff38a352970490ccc57e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 19:46:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eab4c335050d4fc89cfd1fb2eae3818f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/273b2250d8764d9586aa7367e29b0635 : SUCCESS in 10m 11s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/7f032ca6588e430fa7a325968cb20ad6 : SUCCESS in 1h 33m 12s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ab3c35f360a74e3aace083debe0ef2b0 : SUCCESS in 1h 47m 56s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/f72b6d5037234cbc90f927c11ece60f9 : SUCCESS in 1h 55m 00s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3536c0bc5c9645858cb0f055bef7ce24 : SUCCESS in 2h 05m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6531f3550cd74d809e96ee2155f37019 : SUCCESS in 1h 44m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1ee9129c43f421eac77cc05ba4471b2 : SUCCESS in 1h 55m 17s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ccdc8bc34a8474b8f2b209375ab2f78 : SUCCESS in 2h 08m 28s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ce58c3549914d21876e1098107a315e : SUCCESS in 2h 45m 18s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9efa37d52d1b4bad9305411432cd0c69 : SUCCESS in 2h 44m 10s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/37f279ad2d7f4cf582ab6f8f15c7ebfe : FAILURE in 13m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff0210f07e224876997158bddc7172ff : FAILURE in 1h 07m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/831c35a2c7d54e6497906f4d4ecd7183 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af4045c70ff449f6a36dbf6a6f5c666a : SUCCESS in 3m 21s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/473136158fc8474eb322e58048f2b853 : SUCCESS in 2h 43m 41s","accounts_in_message":[],"_revision_number":2},{"id":"6de284f96f7360e22651a8151775881dd260ea3e","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-30 19:52:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0f420b291f89bed27bc36b0e0e32d69d288c044e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 14:02:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"db663590f34a8e9f904893f971c7b82ff52c7c71","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-04-01 08:46:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0e1fd4c791c3cc27a43ff956dd39be5c55a9d419","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 08:48:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0fc071aaf28fb1540a3fad77f4ebd40bc940cfb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 11:50:04.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc96604fa1ea4c1fb19f5eea8440fa3a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c274915f6654987979c633cf8abab21 : SUCCESS in 11m 13s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/d70f584f0c68417abc25c668baeb54f9 : SUCCESS in 1h 52m 50s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/dcaf0793ccf542b6912b39987ec99fab : SUCCESS in 1h 00m 40s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/34d3b140fbfe40358b0bff035d9b3e1c : SUCCESS in 1h 54m 11s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/917adba93fce444193c2d0449a0f680c : SUCCESS in 1h 45m 53s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f51f24cdeb86463992534b2f5b87aad9 : SUCCESS in 1h 03m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58a4dd1535bf4d448d75ed990e22a796 : SUCCESS in 2h 01m 08s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f946f768db694ac6af160503209fa171 : FAILURE in 2h 21m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4964b3398a4b43d2a918d7f03073ddd0 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/617c3e719a0d4f248bd3cb66359437f7 : SUCCESS in 2m 53s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d43070de173345b78e4395450194c32e : SUCCESS in 2h 56m 36s","accounts_in_message":[],"_revision_number":2},{"id":"33143b1095cb82a97c558eb10f972905d0012b9a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-01 12:00:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"544ac7d73e26f1e960786c568bd2bd8a75a8b07e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 14:07:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50fc43c64cb4443a8ad9e771aef73dbf\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f759c3de710045419372ea3e79c71b49 : SUCCESS in 9m 58s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/03082c390e1e4c8590f2a217d12d19a6 : SUCCESS in 1h 00m 34s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/350ab86f9ffd44d8a7409d587989b2df : SUCCESS in 2h 00m 36s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/667933cea4124e0c90945932e76c49b6 : SUCCESS in 1h 50m 36s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c66774be15f1435a8b4d34f752d511e1 : SUCCESS in 58m 32s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4a6a4b6fa4a24df5a9fdbf4f18a0531a : SUCCESS in 2h 04m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/247789c7f02749d7834c2038175eb486 : SUCCESS in 1h 00m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c42e07ab9c7b456c8f09924fee74c396 : SUCCESS in 1h 02m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da2561171d4e43fd97a51ab1bbab3752 : FAILURE in 10m 24s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6513bf6ef394af799f54b0b6f8d0d4e : FAILURE in 12m 04s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/26afa037df374e8996e9d5c31fb1a86b : FAILURE in 10m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a339240f8be4870a865d12d9d0417f3 : FAILURE in 26m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32d9dc5ce45a4e66ad36bb10eb957f65 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0e92f7de697414f8c566c46ec18def8 : SUCCESS in 2m 42s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec25463137984d4da9f8ed51a7a57e78 : SUCCESS in 1h 38m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e83b118ef856f3ddcdbc460a72a7a8fbd7034d95","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-04-01 14:12:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a617025fac6c9295274a27c913fd6fa9609e5295","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 16:34:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8f5f59cebb64e8781dc6c20e2067f8b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1d280be8e8d4100892255c3bba65bcb : SUCCESS in 12m 21s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/6de5f12f007841b6b18ad5f764ea52e2 : SUCCESS in 1h 00m 00s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/09a0d7093eaa455ca784a524112a1c21 : SUCCESS in 59m 52s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/e4ed0d52be844066a6997899f18fd99e : SUCCESS in 1h 46m 43s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/44b1e2c515814d49bb3f7f891567a920 : RETRY_LIMIT in 18m 49s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f244af5c0eeb428ca9233ac451c8ca35 : SUCCESS in 1h 43m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/166f4e87ae314583a074dd536fc58a8c : SUCCESS in 1h 36m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5126ee84eed0475fbc7cc51dbff80178 : SUCCESS in 1h 45m 50s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57ca8871d17a43989401e7d0e7600a3f : FAILURE in 14m 04s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7371f6479b264e008881cd0a882c085c : FAILURE in 8m 39s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ecf8985efb5047b5a377e367c0b70a91 : FAILURE in 11m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b6c323073764c6c88a6e5baba7767fe : FAILURE in 36m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdaf473117ed40b0a0b80ad1f0c28781 : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fa564c650834852ac37e4cc9462b9a5 : SUCCESS in 2m 35s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/039d144df2544594a2c748d2868ffb4f : SUCCESS in 1h 15m 24s","accounts_in_message":[],"_revision_number":2},{"id":"d059aadddbc08cf35abe17a04b6c652deff86b6f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-01 21:59:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b041c55cf0847f6f7a9fbe35809f8f91d6796887","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 00:52:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82f6d809ea9a4b2f93aba5b926e7f213\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b15a48ab35c48f0b179b621ba376739 : SUCCESS in 8m 19s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/89b4c1d42d24486e99d708a8ea063aa7 : SUCCESS in 57m 41s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/411e9fae4e354cea925e1f1147b428e8 : SUCCESS in 1h 45m 10s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/07245c143d6243268d954400694f3131 : SUCCESS in 1h 42m 08s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1d4f90c3722e474dbfea1e064a6c8bef : SUCCESS in 1h 01m 03s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f847b25fdd1f42a38e014e4cf4da21b3 : SUCCESS in 1h 56m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c6ae6ab04e14c17828df593df5a390b : SUCCESS in 1h 53m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/080f8e08d8e641d683d5eda2a7977a24 : SUCCESS in 59m 19s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d750076bc1a64aabb95a8aab54660721 : SUCCESS in 2h 37m 29s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34d878adbaa14c699209b252d4e12cee : SUCCESS in 2h 41m 15s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7b6e9d45714e4dd598c51a427d560896 : FAILURE in 10m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a5e233741cf4e1789acfbc1bceb0304 : FAILURE in 1h 05m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ffce2345031477f995af6eb8369684e : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1eb3a3a01b464a9d9524c58aff444257 : SUCCESS in 3m 23s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c49e2dd4c4f4917a193a29f455f5423 : SUCCESS in 1h 17m 24s","accounts_in_message":[],"_revision_number":2},{"id":"31b4c315aacb0294ba55368b815e05063f33f293","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 00:54:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d0ab6862c83c900a900df0b1464eb469e2162a16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 03:41:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d87b3d11063f46d08466b31e8fabda89\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14cd2a5cc9d7444cb5149320bfcafb42 : SUCCESS in 6m 20s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f0fe72234a51444584d9992b6b321478 : SUCCESS in 1h 22m 57s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/312598845b554a6e9194be42a3f20cc9 : SUCCESS in 1h 22m 28s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/7782327bf98f4282affb38801520b7d7 : SUCCESS in 1h 00m 04s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3288203af7564bdf89ee7b1ab01c819f : SUCCESS in 57m 35s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f76d97e631ff4613903c5685c022822d : SUCCESS in 2h 00m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07501dd6338f458da6911c4da18df1da : SUCCESS in 1h 56m 13s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1950181a49d42aeb220809c78f01970 : SUCCESS in 55m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95419f7da66648a1af44a37c2955b5a5 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2282530ee8c54e6db094f38cbe93779f : SUCCESS in 1m 44s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac0f7128d5f643e0ab19140c5720a81e : SUCCESS in 2h 42m 35s","accounts_in_message":[],"_revision_number":2},{"id":"32b20b80b1aaeedb4821046b680230ef76a4dc1a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 03:41:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fda52c0fa0e7370f7caf2af30262e669535fbdc5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 03:42:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35a26cb56fa6496692b9ebde013c4753\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdcaf52527e440529a7c874e91ebb5bb : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d7820cab217952fe8798b3ea04fe4bbe872dbdf6","revisions":{"8c59288deb292e03b6906aba7294b927713f9e67":{"kind":"REWORK","_number":1,"created":"2026-03-27 13:00:48.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/12/982412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/12/982412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/12/982412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/12/982412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/12/982412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/12/982412/1"}}},"commit":{"parents":[{"commit":"f34847d6054efd8a30298d6ab2f916d06e751ad0","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f34847d6054efd8a30298d6ab2f916d06e751ad0"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-18 14:59:59.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-27 13:00:48.000000000","tz":0},"subject":"Disable Octavia coordination temporarily due to ZooKeeper instability","message":"Disable Octavia coordination temporarily due to ZooKeeper instability\n\nDisable coordination and jobboard in Octavia due to recurring ZooKeeper\nconnection issues, including session expirations and ConnectionLoss errors observed from the Kazoo client.\nKeep coordination disabled unless explicitly required (redis can be used also) or until  issues with Zookeeper will be fixed from Octavia side.\n\nBug: https://bugs.launchpad.net/octavia/+bug/2144800\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_octavia/+/978864\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_octavia/+/978680\n\nChange-Id: I0148eda33790eb89502cea72be60a6ce82cdeff4\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8c59288deb292e03b6906aba7294b927713f9e67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8c59288deb292e03b6906aba7294b927713f9e67"}]},"branch":"refs/heads/stable/2025.2"},"d7820cab217952fe8798b3ea04fe4bbe872dbdf6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-03-27 13:01:24.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/12/982412/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/12/982412/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/12/982412/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/12/982412/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/12/982412/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/12/982412/2"}}},"commit":{"parents":[{"commit":"f34847d6054efd8a30298d6ab2f916d06e751ad0","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f34847d6054efd8a30298d6ab2f916d06e751ad0"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-18 14:59:59.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-27 13:01:24.000000000","tz":0},"subject":"Disable Octavia coordination temporarily due to ZooKeeper instability","message":"Disable Octavia coordination temporarily due to ZooKeeper instability\n\nDisable coordination and jobboard in Octavia due to recurring ZooKeeper\nconnection issues, including session expirations and ConnectionLoss errors observed from the Kazoo client.\nKeep coordination disabled unless explicitly required (redis can be used also) or until  issues with Zookeeper will be fixed from Octavia side.\n\nBug: https://bugs.launchpad.net/octavia/+bug/2144800\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_octavia/+/982410\nChange-Id: I0148eda33790eb89502cea72be60a6ce82cdeff4\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/d7820cab217952fe8798b3ea04fe4bbe872dbdf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/d7820cab217952fe8798b3ea04fe4bbe872dbdf6"}]},"branch":"refs/heads/stable/2025.2","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
