)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~982763","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I50b4d3257f624572f5a20e795cad57f4083b4062","project":"openstack/openstack-ansible-os_octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-04-11 00:06:23.000000000","reason":"Change was submitted"},"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-04-11 00:06:23.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-04-11 00:06:23.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-04-08 14:38:31.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-04-01 09:36:28.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-04-10 21:21:44.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I50b4d3257f624572f5a20e795cad57f4083b4062","subject":"docs: Add basic Octavia load balancer troubleshooting guide","status":"MERGED","created":"2026-03-31 10:30:37.000000000","updated":"2026-04-11 00:07:14.000000000","submitted":"2026-04-11 00:06:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982763","meta_rev_id":"b0d9b6aa534271dbc74b28a52bb39e4dfe83e2c8","_number":982763,"virtual_id_number":982763,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-11 00:06:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-04-07 10:42:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-01 09:36:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-04-10 21:21:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-04-10 21:21:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"CC":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-31 13:15:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-31 19:24:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"CC"},{"updated":"2026-03-31 19:24:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2026-03-31 19:24:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-03-31 19:24:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-03-31 19:24:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-03-31 19:24:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-03-31 19:24:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-04-01 09:49:28.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REMOVED"}],"messages":[{"id":"e1be609a3e2449362b1a79a16ceb0e19dfe5de50","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-31 10:30:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb729f9a281118ece30e40095cf3eb772c28a2de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 13:15:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4a3ee6229104edaa6228a8abceacf49\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6bd40e296b2c432090f732692f34e8b5 : SUCCESS in 1m 56s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05e765953e944817890232a3ef2e0986 : SUCCESS in 6m 51s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/317fe98021524f219925c6fbc8692f6e : FAILURE in 1h 24m 21s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/cef3583141624d9c80b2aa13bd614746 : FAILURE in 1h 16m 53s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/bbf9dd0e962e46adbc2f36a789a818d9 : FAILURE in 1h 25m 29s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4cab397fb1cf406f90485ce96aaf3e46 : FAILURE in 50m 45s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/89f9f1834c234d6aa3c177aeca2c5784 : FAILURE in 52m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bc79e06716d4dc18702ac036a5eb45c : FAILURE in 1h 20m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24b72b5fc15a47369b93bd36e855f126 : FAILURE in 1h 28m 53s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/39d37e0076b048889383c93c57b84642 : FAILURE in 1h 23m 49s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cc0e2efbf916406dacf0d377bdbf0e25 : FAILURE in 2h 33m 34s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b09c32c72004cdd96f556d6f3b0db2f : FAILURE in 1h 17m 18s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1afcfc9f6ee44b58a5fa22fc9c7416c7 : FAILURE in 2h 29m 08s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd74d493ec2743db9000e8f8f9f04d51 : FAILURE in 2h 21m 12s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6377109e651c4739b8e48a57c92e3d00 : FAILURE in 13m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad272fafca2241aa97eee0fc1bff09d1 : FAILURE in 1h 04m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47492e2eee2d4cdface02c03991087f8 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1010397107343a9bdf4fb20a0343c4b : SUCCESS in 4m 26s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1703de795654bcd834a49c1e16b4775 : FAILURE in 1h 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6245fbff1229112965b7285276b7914a7e40e3d2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-31 15:37:08.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b48849c24bbf3645eaef2d17b7df35af3dea2e87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 15:54:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a36a2c2bf0044788365993d68a8d342\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cf02170c0244834a8d716fa8fbc546b : SUCCESS in 10m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2365dd5d60244232b69df984c5d43e98 : FAILURE in 5m 37s","accounts_in_message":[],"_revision_number":2},{"id":"7a16ff92d74ee9a384a20eb524c707a955cca1b4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-31 16:00:32.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7c8215a2ab152289b8eca66affec39d26c0ce2f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 16:11:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ab4986ad5b743e2b94c2cacf20d67e4\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/587448fa11ae452cb11555623ff137a3 : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45a8ea3064d645419848bf43b8bf106a : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":3},{"id":"a92faf956b525751d71c23eb7158a9cf70649a71","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-31 16:43:42.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f47c8d963ce5160faa6daed0dcc84f3d003e7b23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 16:57:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b674a458f0ce4e7ca47ed95585c30697\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94400524d8ce48849c3cf88d31c1554e : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03c4feec72ea4cc293af72a140d78e99 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":4},{"id":"d4795b8d3670315ed6a0e9690cd1415dda1f549f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-31 17:19:11.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"84e7282e99b619f6afed6fbc99bbea1a8ac14b56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 17:32:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92770aac130a4414b14850f536bce291\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8bab9a3459d49d891fbb306a17e0639 : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/466ff70d460542229913eb975635e114 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":5},{"id":"378e4f57d139502b367f6967faf81e46eeadc475","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-31 17:52:44.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"1f7db7012ceb819433510285381b8e721c4e1c47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 18:03:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07ad59c8f3524638bc2b0fa1fd2cf4d6\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99d9773f4c2b4105bf56fc8bf3a56704 : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4d98555418b44d3943b42b3ce05c6fb : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":6},{"id":"f1ecf3050a216ef3515509346629bc43c9147535","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-31 18:19:07.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"0e6002da04827899c24d0fc8f27a913380e4a3c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 18:28:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d32469bd4c674fd59cf4eebd95e6709a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/109a1cf251e445fea829fa59d9b70874 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56492ee4dba54112b08ef1243ea91d26 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":7},{"id":"9559aa790b4324063a860d32fcb4cc53d401f942","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-31 19:11:40.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":7},{"id":"c0bd42e9bebc9916e32a7100b5003a692aa10ccf","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-31 19:24:02.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"088b4192f16eb0bd1e68bd2be485e60478c4e5b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-04-01 09:31:30.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"db1015813644728cc37457965c7b60dd6dae7908","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-01 09:36:28.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"4d4afbb31e579e942e5c91929bb6d2e8809b0fa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 09:46:35.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/148ac77cd39b4d4596f1a22e90df5695\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4dbf8aa0e8446e6bcfa83904921e9f8 : SUCCESS in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1576e72261124571a3dee80942cc9eb9 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":8},{"id":"074b70cfacb40b9cb8073fa0aef4b6032cc7c3e2","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-04-01 09:49:28.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_29244\u003e.","accounts_in_message":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"_revision_number":8},{"id":"22a0511621fc2a3d4997084c3453bcad0cf67025","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-07 10:42:23.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7c8804451698fffa29684fa5bd6566415f96e397","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-04-10 21:21:44.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"92191b286e8c20d4a4f8b36df1420f89deb72986","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 21:23:47.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"a6c7ac54346e062ea425837f0ab4c9ac3f31b063","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 00:06:22.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5a22c9bef8b4ccfb5b3018895386615\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e40fcf5c21a4f869308dcd3d29425f2 : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/739cb427bfea4b6bb16d01b63f1a2510 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":8},{"id":"65fa0c79e65dadb762b8050ebb3e529c1df54288","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 00:06:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"b0d9b6aa534271dbc74b28a52bb39e4dfe83e2c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 00:07:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fb8ae179eef4b26ac726a9b9332b3bf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3c209d39ced4b4ca97ea6cce98977bc : SUCCESS in 42s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"9984ba065927e898f342ee69fa9d0b37e34e2f9f","revisions":{"207ecf03a86cb343a4137adf9257a00b63890dae":{"kind":"REWORK","_number":1,"created":"2026-03-31 10:30:37.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/63/982763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/63/982763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/1"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-31 10:30:37.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-31 10:30:37.000000000","tz":0},"subject":"docs:","message":"docs:\n\nChange-Id: I50b4d3257f624572f5a20e795cad57f4083b4062\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/207ecf03a86cb343a4137adf9257a00b63890dae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/207ecf03a86cb343a4137adf9257a00b63890dae"}]},"branch":"refs/heads/master"},"3758780e55b687d61bcac8996eae5c94997311db":{"kind":"REWORK","_number":2,"created":"2026-03-31 15:37:08.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/63/982763/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/63/982763/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/2"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-31 10:30:37.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-31 15:36:40.000000000","tz":0},"subject":"docs: Add basic Octavia load balancer troubleshooting guide","message":"docs: Add basic Octavia load balancer troubleshooting guide\n\nIntroduce a quick troubleshooting guide for OpenStack Octavia\nload balancer.\n\nChange-Id: I50b4d3257f624572f5a20e795cad57f4083b4062\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3758780e55b687d61bcac8996eae5c94997311db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3758780e55b687d61bcac8996eae5c94997311db"}]},"branch":"refs/heads/master"},"0788705e269ff298bf270b550fc1a96f3fc2c444":{"kind":"REWORK","_number":3,"created":"2026-03-31 16:00:32.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/63/982763/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/63/982763/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/3"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-31 10:30:37.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-31 16:00:28.000000000","tz":0},"subject":"docs: Add basic Octavia load balancer troubleshooting guide","message":"docs: Add basic Octavia load balancer troubleshooting guide\n\nIntroduce a quick troubleshooting guide for OpenStack Octavia\nload balancer.\n\nChange-Id: I50b4d3257f624572f5a20e795cad57f4083b4062\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/0788705e269ff298bf270b550fc1a96f3fc2c444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/0788705e269ff298bf270b550fc1a96f3fc2c444"}]},"branch":"refs/heads/master"},"e9d5dc0e117eac57f207372496a09b7d0492e376":{"kind":"REWORK","_number":4,"created":"2026-03-31 16:43:42.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/63/982763/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/63/982763/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/4"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-31 10:30:37.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-31 16:43:38.000000000","tz":0},"subject":"docs: Add basic Octavia load balancer troubleshooting guide","message":"docs: Add basic Octavia load balancer troubleshooting guide\n\nIntroduce a quick troubleshooting guide for OpenStack Octavia\nload balancer.\n\nChange-Id: I50b4d3257f624572f5a20e795cad57f4083b4062\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e9d5dc0e117eac57f207372496a09b7d0492e376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e9d5dc0e117eac57f207372496a09b7d0492e376"}]},"branch":"refs/heads/master"},"e8dcc9dc564278bd9f5da96eb486e46f82b9352d":{"kind":"REWORK","_number":5,"created":"2026-03-31 17:19:11.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/63/982763/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/63/982763/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/5"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-31 10:30:37.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-31 17:17:59.000000000","tz":0},"subject":"docs: Add basic Octavia load balancer troubleshooting guide","message":"docs: Add basic Octavia load balancer troubleshooting guide\n\nIntroduce a quick troubleshooting guide for OpenStack Octavia\nload balancer.\n\nChange-Id: I50b4d3257f624572f5a20e795cad57f4083b4062\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e8dcc9dc564278bd9f5da96eb486e46f82b9352d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e8dcc9dc564278bd9f5da96eb486e46f82b9352d"}]},"branch":"refs/heads/master"},"2aadc68ec36ff7d44c85adc36e1b6e924539d298":{"kind":"REWORK","_number":6,"created":"2026-03-31 17:52:44.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/63/982763/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/63/982763/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/6"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-31 10:30:37.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-31 17:52:43.000000000","tz":0},"subject":"docs: Add basic Octavia load balancer troubleshooting guide","message":"docs: Add basic Octavia load balancer troubleshooting guide\n\nIntroduce a quick troubleshooting guide for OpenStack Octavia\nload balancer.\n\nChange-Id: I50b4d3257f624572f5a20e795cad57f4083b4062\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2aadc68ec36ff7d44c85adc36e1b6e924539d298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2aadc68ec36ff7d44c85adc36e1b6e924539d298"}]},"branch":"refs/heads/master"},"6d6731135541fe956e7788eefb87cddaebf90ce8":{"kind":"REWORK","_number":7,"created":"2026-03-31 18:19:07.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/63/982763/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/63/982763/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/7"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-31 10:30:37.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-31 18:18:59.000000000","tz":0},"subject":"docs: Add basic Octavia load balancer troubleshooting guide","message":"docs: Add basic Octavia load balancer troubleshooting guide\n\nIntroduce a quick troubleshooting guide for OpenStack Octavia\nload balancer.\n\nChange-Id: I50b4d3257f624572f5a20e795cad57f4083b4062\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/6d6731135541fe956e7788eefb87cddaebf90ce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/6d6731135541fe956e7788eefb87cddaebf90ce8"}]},"branch":"refs/heads/master"},"9984ba065927e898f342ee69fa9d0b37e34e2f9f":{"kind":"REWORK","_number":8,"created":"2026-04-01 09:31:30.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/63/982763/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/63/982763/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/63/982763/8"}}},"commit":{"parents":[{"commit":"754e68ca4d84f5bc8330d3e9960ec1b77d587164","subject":"docs: warn about Octavia deployment on single compute host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/754e68ca4d84f5bc8330d3e9960ec1b77d587164"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-31 10:30:37.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-04-01 09:31:24.000000000","tz":0},"subject":"docs: Add basic Octavia load balancer troubleshooting guide","message":"docs: Add basic Octavia load balancer troubleshooting guide\n\nIntroduce a quick troubleshooting guide for OpenStack Octavia\nload balancer.\n\nChange-Id: I50b4d3257f624572f5a20e795cad57f4083b4062\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9984ba065927e898f342ee69fa9d0b37e34e2f9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9984ba065927e898f342ee69fa9d0b37e34e2f9f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
