)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~986789","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~stable%2F2025.1~I2bf72f8ed15efc9750321876ba8dd02589481073","project":"openstack/openstack-ansible-os_octavia","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-05-05 13:17:53.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-05-04 07:40:07.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-05-05 13:17:53.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-05-05 13:17:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2bf72f8ed15efc9750321876ba8dd02589481073","subject":"Switch to auth_section","status":"MERGED","created":"2026-04-30 12:27:31.000000000","updated":"2026-05-05 13:18:45.000000000","submitted":"2026-05-05 13:17:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986789","cherry_pick_of_change":986788,"cherry_pick_of_patch_set":1,"meta_rev_id":"607a404cb7541a454f1ccafe2946a0b9d044a993","_number":986789,"virtual_id_number":986789,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-05 13:17:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-05-04 07:40:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-30 17:24:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-05-04 07:40:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-30 12:27:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2026-04-30 15:39:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-30 17:24:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-04-30 17:24:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-04-30 17:24:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"4b8a639d719613798f55e3b50c84e8137e32819a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-30 12:27:31.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"ede7d55fdb24bc08104685f9c9d2d00390d7e1d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 15:39:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d130e0c71fe4da4b3d75e5c08bc4bff\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe09c35e3bdd44729694b91e0f51ad63 : SUCCESS in 11m 49s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/6cc4d14494cc4a8ab91f6791d606ed4b : SUCCESS in 1h 56m 04s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/4f749aac31fc4a3492440d002be3c22f : SUCCESS in 1h 52m 19s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed9b62550ead4339bd53c34ccc2743ee : SUCCESS in 1h 51m 08s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0fea7d5a5c7c49dd99d77de6031dead0 : SUCCESS in 1h 50m 53s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ffcd40d3c7e743d4983064cccc49a858 : SUCCESS in 1h 04m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/24342b10f40c4e048a2ae834463a15de : SUCCESS in 2h 07m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3c18f3cb08248ad891d14db08a3e57a : SUCCESS in 2h 03m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff40516dd0f5476f945631d3ed70aee2 : SUCCESS in 1h 30m 10s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7058f37b0ae84b91988493be9e9f6405 : SUCCESS in 1h 59m 16s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9491051a38be4ebeb8bc34a836a576b3 : SUCCESS in 1h 32m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7449963a76ec438a9e7bc64bcf4de46f : SUCCESS in 1h 58m 47s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a3a3a2dd9dda4cc3b160186398484bd8 : SUCCESS in 2h 59m 30s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5b11ce58f2884c0abd1a4f8624992a79 : SUCCESS in 2h 42m 53s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c078645719544faaf7dcfcb3ab7952d : SUCCESS in 2h 04m 46s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a754513d84ac4a11a2b9765abc08908c : SUCCESS in 2h 29m 21s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bbf63265bed14102948a718d7d64c406 : SUCCESS in 2h 41m 17s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd2dd0f04593429daae770ad682512a9 : SUCCESS in 2h 23m 25s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1ef5828b004443528947d6be3b501b6b : SUCCESS in 1h 31m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3f883d2d29124198ba6c8b51bef66ffc : FAILURE in 14m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef1c43ca28404fb9be54968f47b726c9 : SUCCESS in 1h 40m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95deb1e8531f4220b8a6db12a03404a1 : SUCCESS in 6m 55s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c716f76929b41d787b7fd6533c026f2 : SUCCESS in 2h 37m 32s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/660165051de84b36b79f803a9d10a400 : SUCCESS in 2h 56m 48s","accounts_in_message":[],"_revision_number":1},{"id":"275ee20b1930ad70ddc8e8190a3e9bf96e980fa9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-30 17:24:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c8b59277b8c07b442d7f3001310391e78b72f94","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-04 07:40:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bf618a2742e3431c07ccead7001d8f6bdc65cbc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 07:42:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"399dc23f89fd31a240ccfc909389df9c16d7097b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 10:14:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf871ce682de43fc86009e20361b8f70\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a945481132b543eb894c70353dc7db9f : SUCCESS in 11m 09s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/8a317274bf174a8aaae7e86b264bc96b : SUCCESS in 1h 25m 40s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ad750facf22e481f9df3147262b1b727 : SUCCESS in 1h 56m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e06658f7bd624bf5ad4feba860c549fe : SUCCESS in 57m 06s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/037e375200a4477380d1146e11d27d5e : SUCCESS in 1h 57m 11s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1f59aa3e4f714082ac61e322119ff0c4 : SUCCESS in 1h 24m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34211625755c43629ce4657b85c32329 : SUCCESS in 1h 48m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/993f637c95064eb2b7d7a23563189bd6 : SUCCESS in 2h 06m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e047049cee0f4c5883173f18e552054a : SUCCESS in 1h 24m 59s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba5946524ff44026bac84bd931aa7b67 : FAILURE in 2h 25m 42s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/586111ddf6474b269dcbb2fd7ab6e2bd : SUCCESS in 1h 35m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f79ceec27114623ae87f09db6a21abf : SUCCESS in 4m 47s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ee731987fe543789ee6d367f8b625f7 : SUCCESS in 1h 41m 32s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c8d4e619391b48f6839fee3cf3390e9b : SUCCESS in 1h 51m 09s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/30f99162fe66432994c495dfa0544e19 : SUCCESS in 1h 46m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b5e39457bcff2e10e6598927e54842246c4a92b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 10:18:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"60121f60c0104da143e94b65acf1a23d723f0320","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-04 12:46:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b598ffc625c72adc918052e766f2df04fe01bdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 13:15:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18d9ef8652ce4baba10c57e3fc9e8de9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65f15427da9246a59611879dde4eef06 : SUCCESS in 12m 13s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f6084eac7bca4a848480bf9df1140fbe : SUCCESS in 1h 06m 21s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3532a04516f747aea7766b8f6680b980 : SUCCESS in 1h 47m 20s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6291d17bf8994b60a548c4503023faf1 : SUCCESS in 1h 53m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/06570c9625694c6bbb6aa68e80d169c7 : SUCCESS in 1h 59m 58s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0fa6998a87874e5abf0237cab3f49dc2 : SUCCESS in 1h 51m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/44f2da50ec7740389a86600b5a8c92ea : SUCCESS in 1h 30m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e01702feb9e431ba591861a40ba330d : SUCCESS in 1h 45m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45a11c35746b4d48bf3adb4914ed77da : SUCCESS in 2h 06m 39s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eb8e211117e049efa34c0e4c27633856 : SUCCESS in 1h 56m 33s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1df891b829b24b3d8aed7e1f5bc83449 : SUCCESS in 1h 50m 45s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7551de9328e946b5bc0cb93ed509948d : SUCCESS in 1h 58m 48s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/04dab527fce74204af95122d2fa4b868 : SUCCESS in 2h 33m 53s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fc190d7ed3844065889fc44820ec6b5a : SUCCESS in 2h 50m 11s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7c600574635404eb704fc70d69d484c : SUCCESS in 2h 24m 20s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/268c7ec04fca4248be65176d1170a627 : SUCCESS in 2h 32m 34s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/de2dbb9c40b648b19b3c539efd279373 : SUCCESS in 2h 06m 28s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32aa47a6fb9549d9a34a8fb6edf40da2 : SUCCESS in 2h 34m 36s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/399b46404e774821bf399e5e13681203 : SUCCESS in 1h 19m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bdd3c65db8db41e291bb10be99d2873d : FAILURE in 15m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/885cfaefe6c34effa6725b329f03efe1 : SUCCESS in 1h 39m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/521d35be5fee4625a97f3db123cce547 : SUCCESS in 3m 30s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/776d5506add2404596177d9cb1d7b8e8 : SUCCESS in 2h 20m 17s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8508c2123fe1486db242c7e55ba58c02 : SUCCESS in 2h 25m 04s","accounts_in_message":[],"_revision_number":1},{"id":"207e529e741119f5f85fba85eb94f3115e7a639f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 13:18:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"126379dffdb3b354d0a95f0a30ca1c53fb528820","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 16:29:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9207efa8345d441fa11ccc8a85743567\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dba3337a229f44ef9b48848db3be8290 : SUCCESS in 14m 03s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/d996e5c8341744afbb4cafb314c951e9 : SUCCESS in 1h 29m 10s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2e68b3ec5450477b9aec8ab66007c937 : SUCCESS in 1h 15m 37s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/443fa773e88e4d3ba194ef2d247e9b5f : SUCCESS in 1h 44m 06s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/86a3a272767f4539bbddb5c078d44856 : SUCCESS in 1h 38m 16s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/06ddb62319d4493eaaad7a7908123fd1 : SUCCESS in 1h 22m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8bca037bce043ecb8a32c041f303509 : FAILURE in 1h 22m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b28eeb70a3b4134833efad98a327687 : SUCCESS in 2h 02m 25s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/64b8019b71e849309943750ec05cd091 : SUCCESS in 2h 08m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0dfdd634596444efa28268a78865f64d : SUCCESS in 1h 44m 11s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47c4bb51f8a2427caaabc1c852e536fa : SUCCESS in 1h 59m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f880626f2ae4f3098da5b53b39ae255 : SUCCESS in 4m 13s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/842e11f3f13540b986130403c60585d4 : TIMED_OUT in 3h 06m 19s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c3fbb3176ada4607a2754114637c7676 : SUCCESS in 2h 31m 23s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a69a76f7ec3d4b54a03cb68624ec188a : SUCCESS in 2h 01m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"904e34d570a8b0f1e4d45ca59783b33da8235c01","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 16:31:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a52f4a4b5375819bdcd169157bea0a3ffc3c340f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 19:48:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/625f42a915a942f8be08092eb70a6a84\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e04bd5ae163a4ee7aaec5b4d7ed6288c : SUCCESS in 9m 04s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/42b1f7aff82e44b69e0705c4fdf18f5a : SUCCESS in 1h 44m 32s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/acc6765eb8704f47898e466f795b012a : FAILURE in 1h 54m 45s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/90dea368cac44f8080438469113ee724 : SUCCESS in 1h 53m 15s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a88922ee70824345a8d0174cd6e52c9d : SUCCESS in 1h 22m 36s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3c97a480ea2d456daae4122e49735807 : SUCCESS in 1h 24m 43s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2552e762fde0478b8502575433379eff : SUCCESS in 1h 25m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ffe5c3fb22c45818d35c595c54967cb : SUCCESS in 2h 05m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cdc41f96ec04181b5c8546da7a3faed : SUCCESS in 1h 52m 31s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b654a191ce3649e7800226750eaa6f7f : SUCCESS in 1h 56m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfdd6e9b92ce4eee91e86812dfc1fed6 : SUCCESS in 1h 57m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7b92da3fbd54fd19b48163c19b98ab8 : SUCCESS in 2h 08m 30s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bed9f0935ce54ccba7beb873543792dc : SUCCESS in 2h 38m 35s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/898c224cad4d49788b20da93b11064a5 : SUCCESS in 2h 08m 58s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f49fa64a4d24fdb815311a0da7beedb : SUCCESS in 2h 30m 49s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc5a06615afe4aee9e94d25aa9e87562 : SUCCESS in 2h 26m 40s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/af3b01d3313c4cc5970a76bace154289 : SUCCESS in 1h 56m 45s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdccd3be866f4202aa764b205aacb2d8 : SUCCESS in 2h 38m 22s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e67832703c384a97befc820d908f5b4d : SUCCESS in 1h 43m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ee3558d0c0f34060986dff1f86bcf982 : FAILURE in 7m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aef319c3331a41c2a0a90751df5024e8 : FAILURE in 1h 45m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bc471be182042a3b094b2eb03defc4b : SUCCESS in 7m 45s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/564cd33fa21c4c7c985b47b1b724e35c : TIMED_OUT in 3h 08m 10s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ec0e08ef759b4cd9a6b42cb1758d044a : SUCCESS in 2h 33m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d55b2b22bc331e720836f4de07844e8cfbd11b8e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-05 07:52:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1190be4147de1a5b27b5dfee499d83b1e27a9fb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 10:39:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/008446ed41d0484ea9e0cd001b2f343a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/837bf8922386432e8eb5f68ecdab6f10 : SUCCESS in 13m 02s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/1089a61810e84ccba9ebfffb06da3ba3 : SUCCESS in 1h 44m 40s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2f7c32b510c94dc2ab1410b9f523cda1 : SUCCESS in 56m 14s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/751abda0982b46ef8aaf0e81b365755e : SUCCESS in 1h 26m 49s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5d33ba2c08e44b5ebb22d63e397ac5f6 : SUCCESS in 1h 56m 35s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2f8c74be6b394d468e6ddd67553826dd : SUCCESS in 1h 56m 56s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/489a878b8a3a430aaf9b652ba9c86250 : SUCCESS in 1h 24m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08c36c5c4c4e4deda803dab85359e932 : SUCCESS in 1h 23m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1a3612a75c641998eabd3c8d99d68ac : SUCCESS in 1h 49m 43s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/840cb583ba3a49fb9efe397146e7712b : SUCCESS in 1h 03m 17s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6dd0a115b2684856bd5803f1d5c52d46 : SUCCESS in 1h 00m 28s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51945b5862f84fa2925cc3bd6bdfff61 : SUCCESS in 1h 30m 54s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/032cacb6b27a4cd389cc2002cd0195d5 : SUCCESS in 2h 01m 22s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7abd5446f5d341a4bd5309efa2b1a9d4 : SUCCESS in 2h 02m 21s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54535043bcf446dcb22cb75e7f527c7c : SUCCESS in 2h 33m 23s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25225380b48b436497f176c60bfe8276 : SUCCESS in 2h 27m 16s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/70b19f4d10d24003afc77bf79c650a9a : SUCCESS in 1h 19m 54s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a38dae8940b4db4afdb999493ee0666 : SUCCESS in 2h 39m 42s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c348269852754797b8c483d9a6ac1e3f : SUCCESS in 1h 35m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dd916a76c6014457a05f2b0f28a5c3fb : FAILURE in 8m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/904ffbd945f547b8b90fc8b436c49a57 : SUCCESS in 47m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47e00a7eadfb4b3e81ba2bdd2a3213ce : SUCCESS in 6m 31s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2b114d4997948fdae5feed9f5640c9d : SUCCESS in 1h 36m 08s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/897bc01b76c84cb692ced0df79283257 : SUCCESS in 2h 23m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d8d7f309b491f3055fafd7a0ca240fdee7319db3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 10:42:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06d516345090a7ba83b7e7da98a4a171303526cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 13:17:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/645f0a06c09c469ea2da86c3671e7e62\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78459907f953455a8dbcba422b05c4a6 : SUCCESS in 7m 45s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/b3f568ce1e634b5bbaeedb3dcfb71660 : SUCCESS in 57m 39s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/eee388d77d884b4a95fbcb3cc19e6023 : SUCCESS in 1h 44m 01s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0ee0eb9d7469451aa1200ce220f2e92c : SUCCESS in 54m 51s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/34f2b371aa1b4004a717a63449ced2f3 : SUCCESS in 54m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9072788eee664086870f9e8f57500756 : SUCCESS in 1h 32m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/add7e7fb040d4657a8744f8aafc4644f : SUCCESS in 1h 41m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9735b9cb67e34d67b971295a75589096 : SUCCESS in 1h 52m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9ea31011cff04317933969e4d7b6db19 : SUCCESS in 1h 55m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4211c6f0c4545cf91c0c63eb0c17e38 : SUCCESS in 54m 49s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4653f63cfe5845b8a77e46ba429a6aed : SUCCESS in 56m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a03f5c5754b4f6fa21d439f51b330b7 : SUCCESS in 3m 20s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c2ea8c3c782456785969ee7c47d6155 : SUCCESS in 2h 16m 43s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2c7b63ae8ceb4ed980e27e82779e7a54 : SUCCESS in 2h 30m 25s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d991d61ff5a14aaf901b389fc540f621 : FAILURE in 1h 05m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"97d8eecebd8c3e833b0fd2982858f688c188ad68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 13:17:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"607a404cb7541a454f1ccafe2946a0b9d044a993","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 13:18:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c547221431e4cdb9a6dd3fe5dd1c020\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e815a19b32c45fcb87f22f3b0b8bdca : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9acdf7582518983e2a9f7407d2983b87708d0744","revisions":{"9acdf7582518983e2a9f7407d2983b87708d0744":{"kind":"REWORK","_number":1,"created":"2026-04-30 12:27:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/89/986789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/89/986789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/89/986789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/89/986789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/89/986789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/89/986789/1"}}},"commit":{"parents":[{"commit":"0c929ac007ac0adafc57a9577cbcefbe031c66bb","subject":"Merge \"Add keystoneauth options to [neutron]\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/0c929ac007ac0adafc57a9577cbcefbe031c66bb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-23 06:27:31.000000000","tz":540},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-30 12:27:31.000000000","tz":0},"subject":"Switch to auth_section","message":"Switch to auth_section\n\n... to avoid duplicating options to the two sections.\n\nChange-Id: I2bf72f8ed15efc9750321876ba8dd02589481073\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n(cherry picked from commit 59f4ed6399563cd8d0b9283960c51c9142945060)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9acdf7582518983e2a9f7407d2983b87708d0744"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9acdf7582518983e2a9f7407d2983b87708d0744"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"ours":"0c929ac007ac0adafc57a9577cbcefbe031c66bb","theirs":"7174cf5b8c2a5a8329c057244c33e758d3f84f1d","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
