)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~987021","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~If18d19a2e52c84241eb4728ddbd1a0bcdf333e03","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"osa/functional_tests","attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-08-27 12:07:50.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-08-27 12:07:50.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-01 02:29:29.000000000","reason":"A robot voted negatively on a label"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-08-27 12:07:50.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"If18d19a2e52c84241eb4728ddbd1a0bcdf333e03","subject":"Drop functional tests","status":"NEW","created":"2026-05-02 18:28:18.000000000","updated":"2026-09-01 02:29:29.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5d854f5ed71ab25c6d0161e89d8d2943275864a2","_number":987021,"virtual_id_number":987021,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-01 02:29:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2026-08-31 23:14:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-02 19:07:44.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-05-02 21:46:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-11 22:58:52.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-08-27 12:07:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"77b13119a6ba894393989825263fdebdc2c7cd88","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-02 18:28:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9698ac1cc9970354e84288e55c5cada6b0ba5ce7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-02 18:30:16.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c6756db394e53b46a089b646260872d020158180","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-02 18:30:19.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"16f2065b8a432fc90562c65f26b408d4ea1884d8","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-05-02 19:07:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"800ff46cbb859d5439b89e0de6c9d63c54bf1c32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 21:46:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1bf22749d76435ea56bf501f95c8792\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d1881d7c6334ea3acf3c9bd1d371bbd : SUCCESS in 9m 17s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/8ac3ffead315480dabf04b3dbaedf0bc : SUCCESS in 1h 25m 02s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b73461cdf7cb4679a416a917abb40844 : SUCCESS in 1h 22m 09s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/9acb5bfb324f446791833d7c9390c29c : SUCCESS in 1h 21m 18s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c16bf8129fd1490db72e5eccb0ab72f3 : SUCCESS in 1h 01m 22s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2d6f832eccbb43f5aa4c747a516c16cb : SUCCESS in 1h 32m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b87f77619e054a5796555d41228dda18 : FAILURE in 2h 14m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d84d672485b24f50832641ad1141d736 : SUCCESS in 2h 20m 00s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fc4700c59835476980050e9a5aa44d10 : SUCCESS in 2h 43m 41s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2baaa513e0734288938429ddfa2a43ef : SUCCESS in 1h 33m 53s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ef3fa436c82433bad807949765b5672 : SUCCESS in 2h 04m 10s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a29fb9f82c742dfb75957da0813a28e : SUCCESS in 2h 50m 50s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e2acd51073a4ec7a21f31802c155031 : SUCCESS in 3h 08m 04s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/23a5f841a7c64279bf95d4d95048be56 : FAILURE in 8m 43s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aad584b87de9497f8d91b5403e20462c : FAILURE in 1h 02m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d835c6a5ac374374a8f1720e419b415d : SUCCESS in 4m 37s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/010bddaa46f04924b3be72475573dfa7 : SUCCESS in 1h 49m 56s","accounts_in_message":[],"_revision_number":2},{"id":"8b2e2e7467f9fd45d096f49de305360c9a94b511","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-03 13:25:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fbd4f5c33b22a333ce30a23956a515d7438e5303","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-03 13:25:20.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"3c2358dbe10d46d5f4e939499e0ca42560548720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-03 16:18:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9fcfe02288d40c089824c17d3132d43\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac85a29a5435472f8ff78595ece123c8 : SUCCESS in 7m 59s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/87e396e9b0084177a8d7d2ab384c5fca : SUCCESS in 1h 22m 33s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/75444af1f7c14e0c99acca621a77406f : FAILURE in 2h 24m 14s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/f4c556d02c28427e8127053057bcdcb2 : SUCCESS in 1h 47m 43s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e1b5710c429443d38810f3a6b36e6f2f : SUCCESS in 1h 50m 20s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/86a23e9626ef4ffd93ec285db2f99b9e : SUCCESS in 1h 56m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fed5bf2d55884cd6889ed4245b1727ac : SUCCESS in 1h 54m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85a93a2baec64593936cd237813d4a9f : SUCCESS in 2h 18m 18s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4440d21c7ccb45f58c8d51c0e10aae51 : SUCCESS in 1h 31m 20s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9ef21adb21674f21817d07207259a749 : SUCCESS in 2h 44m 41s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01bb6ce1f17d4019b0cfa1770ef59c32 : SUCCESS in 1h 19m 34s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d604b58894114886ab2632a12a856f10 : SUCCESS in 2h 02m 12s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47d3f988d3f3484aba21d708f3ffca0d : SUCCESS in 2h 46m 59s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eed2225938c24685aba294e603cd96ae : FAILURE in 10m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/703c570ed23f46a7843618289ba1bbb6 : FAILURE in 1h 04m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01ad2469dd6e40288980fa851ef1ceaf : SUCCESS in 7m 08s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/615c85cfb1fc4834b2372e63a899f951 : SUCCESS in 2h 40m 57s","accounts_in_message":[],"_revision_number":2},{"id":"4cf81a25a0e1b5a88bdce41c40d496e267988788","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-03 19:56:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4ac4ed600e105d617a767c7cdae5e7360c760977","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-03 22:49:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/65bec22cfa74469a820f4b269841a3ae\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ca2f973fb59442384c0f62c07527653 : SUCCESS in 7m 26s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/893d874a962e4f8f8ecebd8af94eaae3 : SUCCESS in 1h 36m 50s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/62252f99d0364dc1a632c1f74877863a : FAILURE in 1h 36m 09s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/a6d8f289fb064e02853dda4ba0defe43 : SUCCESS in 1h 22m 43s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/72135c30e2974cf48df096b53dffbe0c : SUCCESS in 1h 22m 32s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a08cdb2747b547218300f78fb493c6bd : SUCCESS in 1h 00m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82c3b12b84f742e88f45e6f239795b72 : SUCCESS in 1h 55m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/233403f2a6ee46f3bb5cd2be9a22afe0 : SUCCESS in 2h 03m 57s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c947c8753663497f910eeae79d00dca5 : SUCCESS in 1h 32m 52s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3af28cd0613a4e2287f90a772d8a91d5 : SUCCESS in 1h 21m 56s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a70264de004349bf94f997ecd286c318 : SUCCESS in 2h 45m 22s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/719074bd587a4fb0b84ef2c855084a7d : SUCCESS in 1h 20m 44s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/532b015e818743caa829cab6046b79ee : SUCCESS in 1h 34m 38s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fd1ca57b1e28403db76b878ec616fd5a : FAILURE in 7m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08d3d321fcaf4aea9378e1e687ea0366 : FAILURE in 1h 04m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7449ab5b8f4d44deb51b09178c67476b : SUCCESS in 3m 08s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06f32745f776412381a861c53c9ab3cf : SUCCESS in 2h 33m 25s","accounts_in_message":[],"_revision_number":2},{"id":"659172a614e12530b68e664b751567d6f4845f0e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-08 10:47:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"347c66230ce167322cba29929e621a4698692eea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 13:48:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/166f08febc064e99b28812e2fa5ac3c1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8159be33a6304a9aac8a53be33d740c6 : SUCCESS in 9m 11s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/e36a4d03dd5245a2891246a282b65bd4 : SUCCESS in 1h 00m 33s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f087c4b71b6841938595d990d7465950 : SUCCESS in 1h 02m 57s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/f542cb9c1a6a4f5885d3196c1ecfa34e : SUCCESS in 1h 34m 02s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dea695e7d20743bc914ad90bf94d9c5e : SUCCESS in 1h 41m 37s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0117f859950341f592b11e6969cbfc92 : SUCCESS in 1h 40m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a69f05ebe2344609eafd0ee185414a1 : SUCCESS in 1h 32m 28s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28c56a4216fe404c802b22e8442ab8b3 : SUCCESS in 2h 02m 12s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eb87bb3d24c441eabaea65c5205840c2 : SUCCESS in 2h 28m 05s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8bcb94595e644e289b89e62ddf95774d : SUCCESS in 1h 22m 33s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4a82f07e63c41a4a3161b7d12dfba7d : SUCCESS in 1h 56m 26s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02b22bc2adac47ab8324898695476bff : SUCCESS in 1h 22m 46s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f32437829d2a4989a00e58df93101d74 : SUCCESS in 2h 57m 23s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b5636d72390c46db9a09c7a20a6611ee : FAILURE in 9m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f363eed491a343c9a5f00f18d31b65cf : FAILURE in 52m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/379b00ad6ab54daf9c1557483f975763 : SUCCESS in 3m 38s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9de2c70795264bd0822c3b36bc03355a : FAILURE in 2h 13m 52s","accounts_in_message":[],"_revision_number":2},{"id":"2bbd8162600600f51dff892112d6a06c6df38067","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-11 19:56:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5dc9257864f643611a7ee9d5d2f1164a14980973","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 22:58:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fdbcf0b29cd4cdaa0dc4e70b4f2b4ab\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8aa7c6c02b3d4bcf86c96493857a50a9 : SUCCESS in 9m 22s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/858e7d9b7a754bfab0a950b2c708af26 : SUCCESS in 1h 07m 41s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c9704e8635914b12856b0cceab1c1ff1 : SUCCESS in 1h 49m 20s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/6791934b4c3f43e785f9e544abaff61f : SUCCESS in 1h 01m 26s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3b45f3cb3e2245d6a1b70f3228001fee : SUCCESS in 1h 44m 39s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bc1476fdd63249cf90e4de5c6fedc046 : SUCCESS in 1h 11m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b274946eca2b43d18e33fadf1b0758cf : SUCCESS in 1h 48m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53847eefd19943018571d724e778742b : SUCCESS in 2h 04m 19s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/60acab02176d4e2797b3d85812c56f52 : SUCCESS in 2h 50m 09s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1b9fdb0c35cb4f6d938aeb47928cbd2f : SUCCESS in 2h 18m 25s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2844c62f02814a75bd48c264d8626fa0 : SUCCESS in 2h 54m 55s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1760f4a1db4347d1a99098431fa72f19 : SUCCESS in 2h 47m 07s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2d7dc055a0046cb85e23f0e2abd2ab9 : SUCCESS in 1h 22m 14s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6de0112f6a22410a84be220d417c3381 : FAILURE in 8m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80b89bbf52984ea0b6baa86e3b608509 : FAILURE in 50m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1e4eeab1e6145618e3d6f3aab950ed4 : SUCCESS in 5m 08s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07f4fb033b7d411ea3e366ecf85ef972 : SUCCESS in 2h 06m 24s","accounts_in_message":[],"_revision_number":2},{"id":"0f8e6a11581c02eb0b14a72ca0e6758a42d59eb2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-11 22:58:52.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"e7e6b970c8fe9fb855f0b43db099a8828baba833","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-12 08:10:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4dde999d193e886be5ab3bbc39d9b8019f71b3de","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-12 10:41:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b669d4f42326c3c30f222d5c7397105c517afcca","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-12 10:41:11.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"e487eaa214682b23687d659bc2d466a127590c9d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-13 13:58:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"262d06fd1eaf0dd2cac50a5c6faec512f4e3ffb5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-26 14:53:53.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"7f9b34bd604c7bd7f08af17ceed1ab17cc7310be","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-26 14:54:04.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"0c0f48abb263455709baa73dea7ccae90c85f0b0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-26 14:55:48.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"c5e8252758d188dfd1b7fc0f19e5c0296e85741f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-26 14:59:09.000000000","message":"Topic set to osa/functional_tests","accounts_in_message":[],"_revision_number":5},{"id":"6feedec4885e15e47729ca40c8e87eac11838e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 18:17:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e0db8e44598418c871af5cef1a12318\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/afca747e748d4da18ab41854d5ef7edc : FAILURE in 1m 57s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d69dd5a5893c474fa0de4578d096db79 : SUCCESS in 9m 25s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/15e6ae39a7654247b3ae49eea39f64f0 : SUCCESS in 1h 29m 50s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/431280f08bc64a9091947f7dc8ede892 : SUCCESS in 1h 23m 55s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f868d3bf1d1941fd85c3d4e9f984199c : SUCCESS in 1h 28m 17s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/889b34e02ddc441591f2f1f2ce8c480a : SUCCESS in 1h 30m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b652ea863f44144b4a818fafc085e79 : SUCCESS in 2h 00m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e215e50d7315429d84a28d879cf227a2 : SUCCESS in 1h 22m 17s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/211b07b9a5e4401c9c4eef023083912c : SUCCESS in 2h 12m 40s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/50b83733d7bf4751ba8931c641c0528c : TIMED_OUT in 3h 08m 03s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1a01d2a56134c179fdc9e43bfec60a3 : SUCCESS in 2h 03m 04s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/401c26db177e453a911906697f4065b5 : TIMED_OUT in 3h 11m 32s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16586e537edf4769ae66166e5286b8fb : SUCCESS in 1h 25m 09s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d6c9423d167a4f3089d8bcc8381fe189 : FAILURE in 1h 49m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bccd6c8d2a14f94aa08ff37bd29af4a : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a7f6a26f1544fc4bed16ac5bdac8ca5 : SUCCESS in 4m 17s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73f74a91be4541609af882e67c4bb74c : SUCCESS in 1h 21m 26s","accounts_in_message":[],"_revision_number":5},{"id":"ec56441789957d16aba4bacf440bbdd14c5db863","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-27 08:29:02.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"1814099f48b0e4f19020bd825672fe09edd11150","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 11:46:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d5f8afda0524a69a77d2333fd2372ce\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7922629ead4049f4812e06ddbf0f9282 : SUCCESS in 12m 45s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/6f51f0ea1bb6435b8be9ea88b1313a77 : SUCCESS in 1h 20m 51s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/5466bc12c596450bb2f00223ad51d719 : SUCCESS in 1h 21m 56s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0d3b9552757141e88a07209d242fd533 : SUCCESS in 1h 01m 47s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8b2c4900dbb040d5badc82e35bdf99e8 : SUCCESS in 1h 03m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7007bc2d4ec84061957b426c5cfbba56 : SUCCESS in 1h 54m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/128556ebca8e4a82a105792d5c5e0e7f : SUCCESS in 54m 46s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6792991d9e24aa3933d1d4c858b5fcb : SUCCESS in 2h 08m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1b61bbdce0cd4abbbe6fc6db39ee4577 : SUCCESS in 2h 30m 59s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6ce2e2b3f1947dc8e392768b9359eeb : SUCCESS in 2h 53m 28s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3276780495c84af292567c809c8d2a7a : SUCCESS in 2h 40m 24s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e6c7ecb281f4a8aaf69550fc08d7429 : SUCCESS in 1h 02m 04s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/fa549571be1346f5aaf7a10eb1c73ac7 : FAILURE in 2h 07m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e154e8a66afd428982fedea5b8a23b92 : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fac3b349f6ca4a2a827db8785f918d7e : SUCCESS in 4m 22s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d146afff8fbb4bcf8d69e42154a97c00 : SUCCESS in 3h 01m 58s","accounts_in_message":[],"_revision_number":6},{"id":"e325735fc423059205b1aef6708fe509e8e82f37","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-27 12:07:42.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"3bd32a504f959e41b41d1bcd1101007da8b168b6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-27 12:07:50.000000000","message":"Patch Set 7: Backport-Candidate-1\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":7},{"id":"0acafd18257c9a4c3daddb2724539350aa85c2c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 15:13:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d331b3a0fb5430eab6a64aaa0dcdd9c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/471cb6628c7b4870ab2d14a121801695 : SUCCESS in 16m 23s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/c460d628f090413c989cbe93cce96df4 : SUCCESS in 57m 17s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/ae428683248a45389c0f39b12a8079b5 : SUCCESS in 1h 47m 13s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/719de57a4f024dd0ae799598bf09d41b : FAILURE in 1h 17m 01s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0f28155ea1114939a2769040bb78dbbe : SUCCESS in 1h 02m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/420ab91084824b47811c4a3edf8b14ef : SUCCESS in 2h 06m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/cc7fb703b92844e29c077a05809beb5a : SUCCESS in 1h 24m 28s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b83c989ff6b4b8aafe9c39c10743ed7 : SUCCESS in 1h 28m 13s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c87cbaa4c1684c6f8b16cdad81f970d3 : SUCCESS in 2h 54m 39s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd70cd8dfc1c4631944119dfd688d43f : SUCCESS in 1h 27m 55s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49527a55a2024c11b07881aec7222603 : SUCCESS in 1h 31m 01s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fa99aa235a64318acea2e41f798218c : SUCCESS in 1h 53m 15s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f4826d94e3ec4d60b159a2d6a01193ad : FAILURE in 1h 32m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ed1124f5aca450c81224b98bde1a6e0 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af70a47d7cd74eb4856db70354ad4cd2 : SUCCESS in 2m 46s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9da3964f269c4df2b472db677b5cab80 : SUCCESS in 2h 45m 19s","accounts_in_message":[],"_revision_number":7},{"id":"e6cdce635bb088cf763a24d87e9ef1defb426ae0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-31 23:14:01.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"5d854f5ed71ab25c6d0161e89d8d2943275864a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 02:29:29.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77819d8e6b0c444f8358ac629a7d1e6f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1db296bd2d3a437091b91e96c74c80cb : SUCCESS in 13m 50s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/7b81cdddff8245b7bbc8503e74de9640 : SUCCESS in 57m 40s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/69695914349649dca6dd1aceb1f64363 : SUCCESS in 1h 52m 17s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/332a6481d2c7484f8b4ee8e8c677574f : SUCCESS in 1h 55m 16s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f6ba2fdc433846419f3afd422d4d1ae5 : SUCCESS in 1h 11m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df93645dbe30431eaebc71b41fa7f93e : SUCCESS in 1h 58m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/870f6720851d48328248a94b4fc35d01 : FAILURE in 1h 56m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bacb2fe22f04c92abccdbd082ba2b31 : SUCCESS in 1h 10m 25s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/46afd67ec4524ffb9b23d17fd73b03c6 : SUCCESS in 1h 33m 14s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5acab71820a4d0ab3ec4534a3344bfd : SUCCESS in 3h 04m 51s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd3b12e4bd72499bbfde01713a27ffd0 : SUCCESS in 2h 58m 39s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2cc61d846709405d9b3deb733685289e : SUCCESS in 1h 35m 32s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d309eeeb4e3445be9e09910ca979eaa7 : FAILURE in 1h 31m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf39f98f716b4cc9b2760f59817a68a9 : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e90ce5c3334d42a68e3fe18daeb19513 : SUCCESS in 3m 42s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb68d5f947f94a59b044d826adb09e7b : SUCCESS in 1h 33m 45s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"1ddd81892e7904ce9b18dcaae8d0eabeb93d268f","revisions":{"b54e76ffb13e5ac4339ca195925c64e9ba6e2b44":{"kind":"REWORK","_number":1,"created":"2026-05-02 18:28:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/21/987021/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/21/987021/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/1"}}},"commit":{"parents":[{"commit":"3516f2f808141eaf971196510c1ccb37e51532a3","subject":"Updated from OpenStack Ansible Integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3516f2f808141eaf971196510c1ccb37e51532a3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 18:28:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 18:28:18.000000000","tz":0},"subject":"Cleanup retired roles","message":"Cleanup retired roles\n\nChange-Id: If18d19a2e52c84241eb4728ddbd1a0bcdf333e03\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b54e76ffb13e5ac4339ca195925c64e9ba6e2b44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b54e76ffb13e5ac4339ca195925c64e9ba6e2b44"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"cb712bc7e8bb98247b5a36a1d9b3844e1fd64cda":{"kind":"REWORK","_number":2,"created":"2026-05-02 18:30:16.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/21/987021/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/21/987021/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/2"}}},"commit":{"parents":[{"commit":"3516f2f808141eaf971196510c1ccb37e51532a3","subject":"Updated from OpenStack Ansible Integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3516f2f808141eaf971196510c1ccb37e51532a3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 18:28:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 18:30:16.000000000","tz":0},"subject":"Cleanup retired roles","message":"Cleanup retired roles\n\nChange-Id: If18d19a2e52c84241eb4728ddbd1a0bcdf333e03\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/cb712bc7e8bb98247b5a36a1d9b3844e1fd64cda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/cb712bc7e8bb98247b5a36a1d9b3844e1fd64cda"}]},"branch":"refs/heads/master"},"f09710c4ec045ca8fd4b4f331d41ecafe8a80931":{"kind":"REWORK","_number":3,"created":"2026-08-26 14:53:53.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/21/987021/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/21/987021/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/3"}}},"commit":{"parents":[{"commit":"3516f2f808141eaf971196510c1ccb37e51532a3","subject":"Updated from OpenStack Ansible Integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3516f2f808141eaf971196510c1ccb37e51532a3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 18:28:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-26 14:53:52.000000000","tz":0},"subject":"WIP: Drop tests","message":"WIP: Drop tests\n\nChange-Id: If18d19a2e52c84241eb4728ddbd1a0bcdf333e03\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f09710c4ec045ca8fd4b4f331d41ecafe8a80931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f09710c4ec045ca8fd4b4f331d41ecafe8a80931"}]},"branch":"refs/heads/master"},"8b52a3d41cbda0d847d0cd273cefc29f44dc9d37":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-08-26 14:54:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/21/987021/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/21/987021/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/4"}}},"commit":{"parents":[{"commit":"4bdb77626e0bf8b8723e4bddefcd4bfe740a2815","subject":"Set resolute distro jobs to NV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4bdb77626e0bf8b8723e4bddefcd4bfe740a2815"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 18:28:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-26 14:54:04.000000000","tz":0},"subject":"WIP: Drop tests","message":"WIP: Drop tests\n\nChange-Id: If18d19a2e52c84241eb4728ddbd1a0bcdf333e03\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8b52a3d41cbda0d847d0cd273cefc29f44dc9d37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8b52a3d41cbda0d847d0cd273cefc29f44dc9d37"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"3516f2f808141eaf971196510c1ccb37e51532a3","ours":"f09710c4ec045ca8fd4b4f331d41ecafe8a80931","theirs":"4bdb77626e0bf8b8723e4bddefcd4bfe740a2815","merge_strategy":"recursive","contains_conflicts":false}},"4ff5d23ef2848ba1aa0e855f2cb7b483092f1963":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-08-26 14:55:48.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/21/987021/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/21/987021/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/5"}}},"commit":{"parents":[{"commit":"4bdb77626e0bf8b8723e4bddefcd4bfe740a2815","subject":"Set resolute distro jobs to NV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4bdb77626e0bf8b8723e4bddefcd4bfe740a2815"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 18:28:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-26 14:55:48.000000000","tz":0},"subject":"WIP: Drop functional tests","message":"WIP: Drop functional tests\n\nChange-Id: If18d19a2e52c84241eb4728ddbd1a0bcdf333e03\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4ff5d23ef2848ba1aa0e855f2cb7b483092f1963"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4ff5d23ef2848ba1aa0e855f2cb7b483092f1963"}]},"branch":"refs/heads/master","description":"Edit commit message","conflicts":{"base":"3516f2f808141eaf971196510c1ccb37e51532a3","ours":"f09710c4ec045ca8fd4b4f331d41ecafe8a80931","theirs":"4bdb77626e0bf8b8723e4bddefcd4bfe740a2815","merge_strategy":"recursive","contains_conflicts":false}},"e013f20b83c71d02ea1276f91e3393549f8b4251":{"kind":"REWORK","_number":6,"created":"2026-08-27 08:29:02.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/21/987021/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/21/987021/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/6"}}},"commit":{"parents":[{"commit":"4bdb77626e0bf8b8723e4bddefcd4bfe740a2815","subject":"Set resolute distro jobs to NV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4bdb77626e0bf8b8723e4bddefcd4bfe740a2815"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 18:28:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-27 08:29:00.000000000","tz":0},"subject":"WIP: Drop functional tests","message":"WIP: Drop functional tests\n\nChange-Id: If18d19a2e52c84241eb4728ddbd1a0bcdf333e03\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e013f20b83c71d02ea1276f91e3393549f8b4251"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e013f20b83c71d02ea1276f91e3393549f8b4251"}]},"branch":"refs/heads/master"},"962d674d4ab0b2feb219d73ae43870e7fbdd0460":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-08-27 12:07:42.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/21/987021/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/21/987021/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/7"}}},"commit":{"parents":[{"commit":"4bdb77626e0bf8b8723e4bddefcd4bfe740a2815","subject":"Set resolute distro jobs to NV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4bdb77626e0bf8b8723e4bddefcd4bfe740a2815"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 18:28:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-27 12:07:42.000000000","tz":0},"subject":"Drop functional tests","message":"Drop functional tests\n\nChange-Id: If18d19a2e52c84241eb4728ddbd1a0bcdf333e03\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/962d674d4ab0b2feb219d73ae43870e7fbdd0460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/962d674d4ab0b2feb219d73ae43870e7fbdd0460"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1ddd81892e7904ce9b18dcaae8d0eabeb93d268f":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-08-31 23:14:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/21/987021/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/21/987021/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/21/987021/8"}}},"commit":{"parents":[{"commit":"766920bfd06fbe1e8b72d483476a8c56d0d3a56d","subject":"Return ZooKeeper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/766920bfd06fbe1e8b72d483476a8c56d0d3a56d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 18:28:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-31 23:14:02.000000000","tz":0},"subject":"Drop functional tests","message":"Drop functional tests\n\nChange-Id: If18d19a2e52c84241eb4728ddbd1a0bcdf333e03\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1ddd81892e7904ce9b18dcaae8d0eabeb93d268f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1ddd81892e7904ce9b18dcaae8d0eabeb93d268f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"4bdb77626e0bf8b8723e4bddefcd4bfe740a2815","ours":"962d674d4ab0b2feb219d73ae43870e7fbdd0460","theirs":"766920bfd06fbe1e8b72d483476a8c56d0d3a56d","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
