)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~992319","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I63c7b915aa52234a1b54c6ce070ec2495dd69bad","project":"openstack/openstack-ansible-os_octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-06-25 07:36:34.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-06-25 07:14:23.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-06-08 18:56:58.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I63c7b915aa52234a1b54c6ce070ec2495dd69bad","subject":"docs: add Cinder storage configuration section for Octavia Amphora","status":"MERGED","created":"2026-06-08 17:08:27.000000000","updated":"2026-06-25 07:37:32.000000000","submitted":"2026-06-25 07:36:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992319","meta_rev_id":"67a7fe4348ec8fdd302384b0d9853b8908377895","_number":992319,"virtual_id_number":992319,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-25 07:36:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-06-25 07:14:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-06-15 10:59:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-06-25 07:14:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-11 10:23:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-08 17:55:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-08 18:33:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-06-08 18:56:58.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-06-08 18:57:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-06-15 10:59:51.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"3a7ae2c829e17dd35fb4ce18e53a3e7f962d08e6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-08 17:08:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f56dd8901f933d31dce3c58cfe690eed1ccdb0d2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-08 17:37:55.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"892d7627ea287e70c0617db4d6e0506ab97a8b35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 17:55:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d0eba1482b34a06824efd79107ecf17\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/970b81f0a39b42cb83235b28d5ce0178 : SUCCESS in 13m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bda4e209cbe54143a2cfa459e5f19532 : SUCCESS in 3m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2ba201a410f46689b524c113257cb81 : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b78229fd909936b1b417994dd0baf472090fdfa4","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-08 18:31:21.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"35a3ef88d155033848e5cbec44b718a6f11e63bc","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-08 18:33:28.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"065c4dce850fd92ee295e381b5a23aea69ad0767","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-08 18:37:14.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"bc4dbdfcf6d951ae876c4ac366a05c830ab99def","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-08 18:37:36.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f4af54240035d89e8c315e2c9c6c363f9eefb03e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-08 18:39:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bf7d74a43563034f7e834474fde3e815553e86b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-08 18:48:02.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"8fb7c536dc1028ea66dd2b117f4896fe06992307","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-08 18:48:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"097b3c7e26b739ec19a49c6b4ca37fa47f6effcd","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-08 18:56:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3bf41f965c269356f9b6a3a549d98621f1ea876e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-08 18:57:07.000000000","message":"Patch Set 4: Backport-Candidate+1","accounts_in_message":[],"_revision_number":4},{"id":"8684232346277208b56e06ef13270742e7dccaa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 21:50:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/473c6af661034f7c81daff38a522e7b9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9866e82553044abb12e9d6872151461 : SUCCESS in 10m 06s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3eb3278834f6473894c75cab3259be32 : SUCCESS in 1h 40m 03s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/18812346f4f44ceeb10d3a2604a3811b : SUCCESS in 1h 52m 30s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/89cd75c0740044cd819bbe7fcbf0d040 : SUCCESS in 1h 19m 49s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f6df1259c8cb4d2b81d95f8688f4febe : SUCCESS in 1h 17m 08s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/50ffae0bd63b4f9f9e8873e216280581 : SUCCESS in 1h 45m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3d92d29a5f346cb9a2265f7735f6154 : SUCCESS in 1h 59m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/c5fdd23f64bc4cdbb32d137fa4d45275 : FAILURE in 2h 08m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fe586c59cae4c7d8b112c54b6f8f023 : SUCCESS in 1h 21m 30s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ae8fedf20a2b46a4ad28d0410e684cd6 : SUCCESS in 1h 49m 36s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bdf8f98620254ffebb7978cba6da2f46 : SUCCESS in 1h 49m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e289a534babf4a238b68177348cf4f05 : SUCCESS in 1h 48m 08s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/849e6bfb06bd475c8ce5d761cfb9bd29 : SUCCESS in 2h 50m 45s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc045365bfed424c940bfbad3296f270 : SUCCESS in 2h 56m 00s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea7194cf3dd54afe8fe5dec079e40577 : SUCCESS in 1h 48m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/175c77c1989f4140be3d549b8136b266 : FAILURE in 25m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c2d907432da4a49bca7f2665b60f935 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb0276bb71e74112957ff7f3cccbc7c7 : SUCCESS in 2m 57s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7566ac53d51741d7ad0fd1eaafb6ae65 : SUCCESS in 2h 37m 24s","accounts_in_message":[],"_revision_number":4},{"id":"0a91315495b9fdfc4279a7fb7529d7f291de16f0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-09 00:00:52.000000000","message":"Patch Set 4: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8ae3b471174321dc07b3838853cf051ff8fc74d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 02:59:14.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2367150c0a1045b7b340532cc6ed7264\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5dc28cc008a44481b7d4c2e1dedae2b1 : SUCCESS in 10m 09s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/c5939ac5bc9e4147918161498a5942b3 : SUCCESS in 1h 29m 27s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c0395e4356314ef38ae3c783e439a621 : SUCCESS in 59m 39s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/5c9266b0f18b49969143c14c7a2da475 : SUCCESS in 1h 04m 18s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8f48b00b0b2c4f74a7149aec636b5de2 : SUCCESS in 1h 35m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/03504f739a3244cea1eeab484c23c83e : SUCCESS in 1h 09m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e791856461a947938f1b19e8a8d634b1 : SUCCESS in 1h 36m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/62decf47b3214e3e8001967423df6fbc : FAILURE in 1h 49m 19s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7916b5c9e6d84ba2be1455e3c75bbdf0 : SUCCESS in 1h 40m 09s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b7a7f9ba010f4dae89888f4acb326a54 : SUCCESS in 1h 36m 18s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d064f2cbc9fb4ce2b3b5a80651300691 : SUCCESS in 1h 24m 29s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7574743300004b18801e576fa4ff5926 : SUCCESS in 2h 38m 47s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bc6f1e5b5204510aafe48a5234d7e88 : SUCCESS in 2h 48m 46s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3e29ca79a79494e9f192967440013ad : SUCCESS in 2h 47m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c38cfdd253e5408bab2dc71a6d7b43f8 : SUCCESS in 1h 02m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b4ba831929534025a5608020a4563f08 : FAILURE in 13m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a7ec8b3457e4b519eb6828a10a25ae5 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db1919a2acb14e23a1c2e01e7923effd : SUCCESS in 2m 16s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/735b7999f1174311857fb657e18a84c2 : SUCCESS in 1h 24m 15s","accounts_in_message":[],"_revision_number":4},{"id":"83402d2583ca41a307bb78238bbff78b66a9c05f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-09 13:01:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"680b24cd64700866943be07a77f53963732290d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 16:07:40.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcb07db3043645838aa5455bf5f49d70\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7219a90af3ff433b86d579fae3fec362 : SUCCESS in 12m 48s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/e169a0f7162547da81811f0502b46520 : SUCCESS in 1h 51m 38s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ade43f5a995b4f91852a67321df82a05 : FAILURE in 1h 58m 45s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/bd8e4a2fdf834512ab445d3db7cf18c7 : SUCCESS in 1h 51m 58s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7027fba6cc624737879f474975a1632b : SUCCESS in 2h 03m 11s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cc1513e59b2c4a49b2369ce7db4f72e8 : SUCCESS in 1h 43m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e63335ebc7e643fead1fe2fc69eb90b3 : SUCCESS in 2h 01m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/58e7e8c6a72347fa82ffe37c07b07acb : FAILURE in 1h 38m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bce70c185014f9a9b5284a0ec5486c5 : SUCCESS in 2h 10m 52s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/36ac1b556332453e988686c6a1223fd4 : SUCCESS in 2h 53m 18s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f6b4eaf2557043bf87212fa5f0474796 : SUCCESS in 1h 27m 58s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6b285789d7b404da7e31eafcb7a20e3 : SUCCESS in 2h 52m 11s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ed34feffed74d0499dcc592f329251e : SUCCESS in 2h 53m 10s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/294cfbede2dd4f5f9decddf600a99a86 : SUCCESS in 2h 52m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9a1e4fd5b6a447bbe66badecebab217 : SUCCESS in 1h 51m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e6655eb04b6e409ca924b5e13d9010fd : FAILURE in 15m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adf53e6a0489498ca86f7cb013ed4ee0 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edd3977fd194470291f88e01a3c0f146 : SUCCESS in 2m 33s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e3d5f45bab84935a8d819f3fb7cbb45 : SUCCESS in 2h 26m 43s","accounts_in_message":[],"_revision_number":4},{"id":"5437eebb4c14e305208d05276930cfde5a16b42a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-10 18:45:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6d46dcd15826abb148812a1fe5f3ab28aa5e66b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-11 10:23:03.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"044edf7bc582abdb510d7fdcf553f8ca33876e4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 10:41:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5746b123b714a209a83b74e167d3c94\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/604ba07275ed46e09e9b1399d580a27f : SUCCESS in 13m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fefafe2ce9e0437885db56c9d648a261 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46c9eb352482407aa2cefb550df09c9d : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":5},{"id":"78b8b7f8a45e14f8d0276e94c76c9c45728f7429","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-06-15 10:59:51.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"1023e8233096f15a86de39b4278a183127c3e364","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-25 07:14:23.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"10de7d9229dd7858bd3eab99fcc87b3aed703898","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 07:17:06.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ad3fbcef9925bdff696b17e4b3f74c4e3a752ab1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 07:36:33.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/552803bfd39d48feb7481fb52195c95a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/770abbe2cc644b50a52fca3df3f45556 : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffbca967a4cd47ddb439683437e72180 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8d6486ac933433693ae31e218d1a9bb : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":5},{"id":"513157a4d405f0e9f660243d8ceff152a4e002cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 07:36:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"67a7fe4348ec8fdd302384b0d9853b8908377895","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 07:37:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/237d7f97638c4c77badd3043a452c0e2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57af7459e7554bb19b6bb91aade40e13 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a2017a507ab1ef0734a2f5e34e8fb5faa6d6aa5e","revisions":{"8fa05290c9d8794b54ccf1ee9c122d084a7e27b6":{"kind":"REWORK","_number":1,"created":"2026-06-08 17:08:27.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/19/992319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/992319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/1"}}},"commit":{"parents":[{"commit":"ed442f1769b694c9b64ce43660dfa1af7a5eed2e","subject":"Merge \"Add details about disable Octavia coordination temporarily due to ZooKeeper instability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed442f1769b694c9b64ce43660dfa1af7a5eed2e"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 17:08:27.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 17:08:27.000000000","tz":0},"subject":"docs: WIP","message":"docs: WIP\n\nChange-Id: I63c7b915aa52234a1b54c6ce070ec2495dd69bad\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8fa05290c9d8794b54ccf1ee9c122d084a7e27b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8fa05290c9d8794b54ccf1ee9c122d084a7e27b6"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"2ccc8136afbcace6bf5e506da18233e23d859627":{"kind":"REWORK","_number":2,"created":"2026-06-08 17:37:55.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/19/992319/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/992319/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/2"}}},"commit":{"parents":[{"commit":"ed442f1769b694c9b64ce43660dfa1af7a5eed2e","subject":"Merge \"Add details about disable Octavia coordination temporarily due to ZooKeeper instability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed442f1769b694c9b64ce43660dfa1af7a5eed2e"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 17:08:27.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 17:37:49.000000000","tz":0},"subject":"docs: add Cinder storage configuration section for Octavia Amphora","message":"docs: add Cinder storage configuration section\nfor Octavia Amphora\n\nAdd documentation for configuring Cinder volumes\nas storage backend for Octavia Amphora instances,\nincluding the relevant user_variables.yml settings.\n\nChange-Id: I63c7b915aa52234a1b54c6ce070ec2495dd69bad\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2ccc8136afbcace6bf5e506da18233e23d859627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2ccc8136afbcace6bf5e506da18233e23d859627"}]},"branch":"refs/heads/master"},"4f5befff248f7c9d1e5fe818d5cdef52d90f6aac":{"kind":"REWORK","_number":3,"created":"2026-06-08 18:37:14.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/19/992319/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/992319/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/3"}}},"commit":{"parents":[{"commit":"ed442f1769b694c9b64ce43660dfa1af7a5eed2e","subject":"Merge \"Add details about disable Octavia coordination temporarily due to ZooKeeper instability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed442f1769b694c9b64ce43660dfa1af7a5eed2e"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 17:08:27.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 18:37:06.000000000","tz":0},"subject":"docs: add Cinder storage configuration section for Octavia Amphora","message":"docs: add Cinder storage configuration section\nfor Octavia Amphora\n\nAdd documentation for configuring Cinder volumes\nas storage backend for Octavia Amphora instances,\nincluding the relevant user_variables.yml settings.\n\nChange-Id: I63c7b915aa52234a1b54c6ce070ec2495dd69bad\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4f5befff248f7c9d1e5fe818d5cdef52d90f6aac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4f5befff248f7c9d1e5fe818d5cdef52d90f6aac"}]},"branch":"refs/heads/master"},"92b6e721d7cf7b3a4086815cd8e306210cea9b67":{"kind":"REWORK","_number":4,"created":"2026-06-08 18:48:02.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/19/992319/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/992319/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/4"}}},"commit":{"parents":[{"commit":"ed442f1769b694c9b64ce43660dfa1af7a5eed2e","subject":"Merge \"Add details about disable Octavia coordination temporarily due to ZooKeeper instability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed442f1769b694c9b64ce43660dfa1af7a5eed2e"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 17:08:27.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 18:47:50.000000000","tz":0},"subject":"docs: add Cinder storage configuration section for Octavia Amphora","message":"docs: add Cinder storage configuration section\nfor Octavia Amphora\n\nAdd documentation for configuring Cinder volumes\nas storage backend for Octavia Amphora instances,\nincluding the relevant user_variables.yml settings.\n\nChange-Id: I63c7b915aa52234a1b54c6ce070ec2495dd69bad\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/92b6e721d7cf7b3a4086815cd8e306210cea9b67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/92b6e721d7cf7b3a4086815cd8e306210cea9b67"}]},"branch":"refs/heads/master"},"a2017a507ab1ef0734a2f5e34e8fb5faa6d6aa5e":{"kind":"REWORK","_number":5,"created":"2026-06-11 10:23:03.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/19/992319/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/992319/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/992319/5"}}},"commit":{"parents":[{"commit":"ed442f1769b694c9b64ce43660dfa1af7a5eed2e","subject":"Merge \"Add details about disable Octavia coordination temporarily due to ZooKeeper instability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed442f1769b694c9b64ce43660dfa1af7a5eed2e"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 17:08:27.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-11 10:23:01.000000000","tz":0},"subject":"docs: add Cinder storage configuration section for Octavia Amphora","message":"docs: add Cinder storage configuration section\nfor Octavia Amphora\n\nAdd documentation for configuring Cinder volumes\nas storage backend for Octavia Amphora instances,\nincluding the relevant user_variables.yml settings.\n\nChange-Id: I63c7b915aa52234a1b54c6ce070ec2495dd69bad\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a2017a507ab1ef0734a2f5e34e8fb5faa6d6aa5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a2017a507ab1ef0734a2f5e34e8fb5faa6d6aa5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
