)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~992498","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I9fb2374e36c808f3c6d15c757c70f0a1c34904d9","project":"openstack/openstack-ansible-os_octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-08-28 11:23:51.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-08-28 08:51:34.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-08-28 11:23:51.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-08-27 13:12:50.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I9fb2374e36c808f3c6d15c757c70f0a1c34904d9","subject":"Return ZooKeeper","status":"MERGED","created":"2026-06-09 17:50:12.000000000","updated":"2026-08-28 11:24:45.000000000","submitted":"2026-08-28 11:23:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992498","meta_rev_id":"e4f85b6d5f92ff20b09b0e03b700c4292a1a125e","_number":992498,"virtual_id_number":992498,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-28 11:23:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-08-28 08:51:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-08-27 13:12:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-08-28 08:51:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-24 09:36:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-09 21:01:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-06 15:01:59.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-08-09 11:06:11.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"},{"updated":"2026-08-26 12:58:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-08-26 13:01:54.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"}],"messages":[{"id":"edb241e20ce9d77ff609e7d3e9a7248d462a24b5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-09 17:50:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa783818462e94914ac81cd4bba1dd7c99dc0121","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-09 17:50:48.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1f5463360df409b339250146f0771d44037b7813","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-09 17:52:27.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"bcb5c29ccaaf7b657ee4e80283b2cf8738b8d3c5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-09 17:54:33.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"2dae5d7e95e2954673270e7a6c8353a5be25dbd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 21:01:57.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-ansible-galera_server/+/992476 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"429c1cbb4007ca0b896205387a74825ea6de9c08","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-10 10:14:32.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"e41ef29e36d634a8c9819195cb9642f5cfd1cf72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 13:39:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f385fcbec9c143fb805398558554afd7\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/340bd920c3634df491be68765ff9a740 : SUCCESS in 10m 41s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/2947e57f61df4f6ebe42178bf0e533ec : SUCCESS in 1h 37m 30s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f506ddf7bd6b496787ec583d47e6cfe5 : FAILURE in 1h 26m 54s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/52d15913dd2242e99bd0b3cee02c1992 : SUCCESS in 2h 09m 17s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3d95a8d76c5749e79d7e1a3dfda0bc40 : SUCCESS in 1h 06m 25s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f65f4f0ae8f24d3caf43cea11c720482 : SUCCESS in 2h 09m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03ab96f2ab004803ab806ab618e865db : SUCCESS in 1h 04m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/aa1b29684a0c45b8a0a9b724b3778f4a : FAILURE in 1h 29m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b390cb6725b48f79bc9b2b57251d070 : SUCCESS in 2h 10m 48s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d7dde2ae35d040089fbbd73680f6eb93 : SUCCESS in 2h 17m 18s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3f7ed6d55e6a4523a9a11b423016365e : SUCCESS in 1h 29m 34s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be005d6e33b049519d0a6d92f3c74522 : SUCCESS in 2h 59m 08s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa3cce18ffbe45c388b7d25d7ede8e79 : SUCCESS in 2h 54m 16s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6696beed001646b5849a30222f2c3472 : SUCCESS in 2h 44m 51s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11df4b9210824ce38e7f73eea98ad882 : SUCCESS in 1h 54m 43s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/fcb1dcbbe13348c09b5050e143d59a16 : FAILURE in 2h 00m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/840382a401204ed4ac8cd3d2ac5db6cc : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4fd655d1ee249ecbb7ba603ff7d9f0a : SUCCESS in 4m 03s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de76d40ad7114ece83d41d33cbe18012 : SUCCESS in 2h 31m 45s","accounts_in_message":[],"_revision_number":5},{"id":"81d4f17687bb17a69e52c7b96496771efb90750a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-10 15:52:46.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"87c6d95f167dceb26ebbcb5ac547a274ee0ff666","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 19:10:14.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7f7cc6e33934740a41c63e523c32409\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/376c4263f212461c802de27ac1d34581 : SUCCESS in 11m 47s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/bf579a444a4f4b5db590a885e6743090 : SUCCESS in 1h 13m 11s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/234643542790461bae403170a3ce2156 : SUCCESS in 1h 59m 50s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/2367443183914dfaad550e8959f2cf53 : SUCCESS in 1h 51m 16s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bb16bcf61e4040df9b77fbb2c47f2285 : SUCCESS in 1h 46m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f2bfbf5015ed4cbaa562c4cce0b4e13e : SUCCESS in 1h 44m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0759ef1e91249589dcc9ca1e7979e01 : SUCCESS in 1h 49m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b4e551ae920e4099a68b67dde31a8c9b : FAILURE in 1h 30m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0bd18f2c9744452b37bc999538edc2e : SUCCESS in 2h 07m 48s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d911ae4a64f143deb9dbf1bd7a26048b : SUCCESS in 1h 35m 42s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b32d3b1f514f41f095f6c012f66b76cf : SUCCESS in 1h 27m 58s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62a9f0ad402f4479a6ec5597667894f6 : SUCCESS in 2h 51m 18s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30cdc655c29a450aaa6c933799047a6b : SUCCESS in 3h 08m 05s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7eda17ccd21d49578554d8552a2a7824 : SUCCESS in 1h 29m 18s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/146cab90f6b7454b9bcc921d3ac136f6 : SUCCESS in 1h 47m 44s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/078b2a629f114d57a56a62e4a45c09d6 : FAILURE in 2h 10m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e63d9750afd45d69dbddedb99353463 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bf9d600833d4ea29268b7482fbb1b96 : SUCCESS in 3m 55s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2dbec5715bf42d5af67c1e31b22a03b : SUCCESS in 2h 19m 57s","accounts_in_message":[],"_revision_number":6},{"id":"064975c5cc9238184deaa28082d6ab86ed2ac0b6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-11 12:06:28.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"b5e29e8bdc0b39ff9140da45d26149e8b39016f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 15:18:41.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/319cc3af7d0d42ce8f4edbc1efeacd3c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28254a71a41c48e49ad6362b9e6a1452 : SUCCESS in 12m 21s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/dba51b031eb247eba32bbe22fae3d496 : SUCCESS in 1h 02m 53s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7cfbf318529d4709af93c41dbe7f961b : SUCCESS in 58m 15s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/56e8f94c2dae48ffa437e4b585540517 : SUCCESS in 1h 46m 02s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ac5b1ed9e0d94f97a6b3ecf4f804f215 : SUCCESS in 1h 58m 23s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a75518e032694dd38d777c6a56e21249 : SUCCESS in 1h 51m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2af7494a64a4a98b821f4d474c1501a : SUCCESS in 1h 53m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/905ad2cad73d4778bf30e075f55f5bfa : FAILURE in 2h 15m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b70567f96a5e45f3991ae19da8341146 : SUCCESS in 1h 57m 57s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ad247bd134c64adfa49a28b1eeecbbdd : FAILURE in 1h 18m 33s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2f4fa57e1849489d9a5d0f9c0c948dbb : SUCCESS in 2h 17m 28s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5c4242edb804249bfdf6c7352d095f5 : SUCCESS in 2h 51m 41s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54540e525aed49349deb6c7886bb01e7 : SUCCESS in 3h 04m 11s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82c1190371eb49188f26717e2ffc90c2 : SUCCESS in 2h 53m 49s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a32e3351faa4fdb8db5b1030550aaaf : SUCCESS in 1h 49m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/9ea3a55d8fdc428b8085da544744b9d0 : FAILURE in 2h 03m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72b1323467284e80ab0cba439aa6c7bd : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7da3856bb60497783c77cc372bc0677 : SUCCESS in 3m 01s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3721e43aef54a3b892122b4eb1000ee : SUCCESS in 1h 25m 44s","accounts_in_message":[],"_revision_number":7},{"id":"7ff5b97513091a41c6247ee6571626220e20f0ea","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-30 12:12:35.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"bd429cff2be6a4b6baa985512d69d52192b80593","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-30 12:12:38.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"2daf6128a6b4caac9d469dc02fedd210c5cbbef6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 15:09:57.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0918d777d0f143d48e9b5be75f5b4a45\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ce36da38ff64c21a83e8d924aa18964 : SUCCESS in 9m 05s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/35b4e1dad21b4da9821083aedeb6cf1d : SUCCESS in 1h 36m 08s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/618a1194e0a44821a10f1979a99b26e4 : SUCCESS in 1h 03m 30s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/049b817528244b199494a406f5b4a0a6 : SUCCESS in 59m 35s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2808d87a146941c1ab65386c03329ae7 : SUCCESS in 2h 00m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa3fd05fa3094fa5b972a48e6b937dc9 : SUCCESS in 1h 33m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d81fb7d01c3e48558a5850fb35b43b5f : FAILURE in 1h 32m 53s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7df4d7bc2d0d4513a8e1d2abee8c7ab1 : SUCCESS in 2h 03m 42s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/46da4aff38b2499b949b84bd87f84a72 : SUCCESS in 1h 38m 11s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b0e78db01374eb8bfba16b51603fd62 : SUCCESS in 2h 53m 02s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f92bbcb1f4d4315aae7a89306a7c396 : SUCCESS in 2h 26m 32s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91a96c45b51445a7af0ecb9bc5758413 : SUCCESS in 1h 32m 43s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/caad9abd17774e86abca4f81062754f2 : FAILURE in 1h 28m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b3d0facb9f444258ddf720b2d0d407d : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8174e861dee44086981f63d584f38d9b : SUCCESS in 2m 31s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cafe570053df459aa7afebddb6393db1 : SUCCESS in 2h 41m 06s","accounts_in_message":[],"_revision_number":9},{"id":"544576061bb51b8aa321366dbc816341058d1e79","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-28 08:53:29.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"e2fadb66d9efbbf201ad2103567ac4846fccaace","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-28 08:54:36.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"0e8a76fdbd48f272fcb1ad013fb2d30e6fd3f03d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-28 08:56:00.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"b79b8fa5955774dc372827cef56d13ba6322f4a6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-28 08:56:20.000000000","message":"Patch Set 12:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":12},{"id":"c46a7fa2354e6970f803702dcddff3d79011abb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 13:31:33.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/abe725d22f8446e7be9a12ff647ebc4b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/581dc9ec86234d71a0bd81660f14fefb : SUCCESS in 11m 07s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f113bce0adc9461e930e79bc7512e90e : TIMED_OUT in 4h 29m 36s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/6829800d782c48b0a664ae59573b4f49 : SUCCESS in 1h 04m 50s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4540873fd3e64944a5245cc5486a96ad : SUCCESS in 1h 07m 31s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cc3aa29c12d547e698e0c565a1e5519d : SUCCESS in 1h 07m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e9e7a693e4244f2940df5502b9db7ca : SUCCESS in 1h 05m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b8248308aaec4f2d96abd48901a9b92f : FAILURE in 1h 34m 24s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02a8aa18574d476689f00be14cb38cef : FAILURE in 2h 04m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/978435792cd941faaee57d4aecf2a740 : SUCCESS in 1h 37m 08s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab43e6b9545d4968b7c5fe879a931871 : FAILURE in 2h 57m 09s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8766f26ad1e4729a412984eb78ae211 : SUCCESS in 2h 49m 15s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6deeac3445f44f7e8536d5758d46aa4f : SUCCESS in 1h 27m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/35a4dff4a09c41ef8e7baeef73f2758a : FAILURE in 1h 31m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29de051123f548b5847c9bd48469b138 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6076611915da49bd8698fd84981a8dd8 : SUCCESS in 4m 55s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f95f7863fffc493aa3284c432347ec7f : SUCCESS in 1h 55m 12s","accounts_in_message":[],"_revision_number":12},{"id":"b42fcd6183ed495480039aac95b6b9d57084d9b0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-06 15:01:59.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"7e1c34de3c29d2fa155803dfb42d0aa9a28b0027","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-06 15:02:38.000000000","message":"Patch Set 12: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"4e410e77702cabfa2f76c7781128e1c74628e435","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-08-09 11:06:06.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"7862671ad7d437ebb5c54c2b075c1d11bfc953bc","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-08-09 11:06:11.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"9996aa2854eff8fa9157a5a5e6a068681f3d8e75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-08-09 11:06:35.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"e984cc3e374b22475395bef2903d02bdcb242ffa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-09 14:23:18.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9f2ca52562c4bfd88e3b0bf6c5069ec\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e00e265dd3442078e627b8da4b9edd4 : SUCCESS in 8m 54s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/551853a610694957a07118c7b393ab4a : FAILURE in 2h 58m 05s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/d1ac02dafa7848098190bbe6a1e6a613 : SUCCESS in 1h 48m 55s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/138b743d34a0442b933c3f86f504a8cd : SUCCESS in 1h 26m 49s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3d7803ab3731450e8f892485feb2253d : SUCCESS in 1h 28m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08efd34d2ea44dc8992616d47d90cf1b : SUCCESS in 1h 50m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/89b73d72a152461aaa3c2579103f4ead : FAILURE in 2h 14m 16s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48b9cae5448d416fbbbe2761f81935a6 : SUCCESS in 1h 26m 04s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/606c421f51df46e48031490fb289e82d : SUCCESS in 2h 09m 01s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/455e395158a04649ae26698142640d00 : SUCCESS in 2h 36m 24s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad4ae0f312ec43a188d9aa6d2ef8791c : SUCCESS in 2h 56m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a93aa4bf972c40a2802846870373167e : SUCCESS in 1h 15m 32s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/39e4be50e55046c497a05cd94d114fa5 : FAILURE in 1h 25m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92ee4405c64645bcadf67ef020742d1d : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2726315ed3a54771a5d9099e5a995b49 : SUCCESS in 3m 58s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83ea28dda21443ad96c9538dbf66b627 : SUCCESS in 1h 50m 34s","accounts_in_message":[],"_revision_number":14},{"id":"a6a116e3a6f7d076e6d879b5e00698388c4b301a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-24 09:34:13.000000000","message":"Patch Set 15: Patch Set 14 was rebased on behalf of \u003cGERRIT_ACCOUNT_37632\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"_revision_number":15},{"id":"0ca0000361d50224b7dda8dcf4fed724a020bc2e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-24 09:36:46.000000000","message":"Patch Set 15: Backport-Candidate+1","accounts_in_message":[],"_revision_number":15},{"id":"1a962974ed0ad721707dcd0a62faa56caf2f264b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 12:55:50.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56830910b6244de39a2893910cd4dcd1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bee4cffa7b44283a3145a5d1e9a71aa : SUCCESS in 11m 28s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f9e38ecdc9044de28c5c2a8ae42a0401 : SUCCESS in 1h 03m 14s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/ab7e691b4b2e4503bf2ed86005ca3379 : SUCCESS in 59m 36s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/485a5025394542b6afeb8e1df340260a : SUCCESS in 1h 02m 43s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/580117accd724f189dd089f1f439adc6 : SUCCESS in 1h 39m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc0007aa618c490bba205d47f8e015d2 : SUCCESS in 1h 59m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/29cc25bd4a1845a2a9c5536855c74341 : FAILURE in 1h 49m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/066d24e8e1c34585ab482bd20b0ed375 : SUCCESS in 1h 59m 33s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b6bf429f1e1847f18194959bba9ee1c6 : SUCCESS in 2h 06m 17s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eac90df5f88c47d0bd1d6abe4873e827 : SUCCESS in 2h 48m 27s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ae184f801004d37b4fd3d6af73ed65c : SUCCESS in 3h 06m 05s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18bff00ac6154c62ac820a02e9f49395 : SUCCESS in 1h 49m 42s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/8eeb816806aa41f4a91653ad72218190 : FAILURE in 2h 11m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75bdf7a695b146d19e7d9751f4e7b1cb : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5c788eb0a70401d8e2c93cb2c8eb62d : SUCCESS in 2m 25s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f64c669cc2574d509840a19a22864a1a : SUCCESS in 2h 36m 25s","accounts_in_message":[],"_revision_number":15},{"id":"c6962c6359c667939895ce976fab1e550ecb6848","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-26 09:42:59.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"62921c3285f1194a0fb5a9b675ef4695ca9cd82e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 12:53:08.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61b057e44ada4582a18cd665a83c17c8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/219e049ce3924390ba06e38436342960 : SUCCESS in 14m 21s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/885583dfa4d14132878d0caa52788a8d : SUCCESS in 1h 27m 41s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/82f08ff88c7d4b1ebef080742ec47a0d : SUCCESS in 1h 48m 36s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e0b987812fc940908511a087c0ee3d47 : SUCCESS in 1h 03m 44s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7b8372a549584a699b5ab894cda5991d : SUCCESS in 2h 05m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f4cb6eeddf14f51b1f9d7ebff6cce1b : SUCCESS in 2h 01m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/72f19fc3fcad4dd38aa8f6a35bae9475 : SUCCESS in 54m 37s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93f28e1ba72345fda25e3df0136ed7ac : SUCCESS in 1h 59m 54s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/17e3004334bf40069124ec1a0245d38e : SUCCESS in 1h 33m 44s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1132ef5562b4b898f8c2feb8d5155c0 : SUCCESS in 2h 51m 28s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3393bb960e3d40579719ec76f039b385 : SUCCESS in 2h 59m 46s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad7d5f445d9f47b08655df9fa6e31021 : SUCCESS in 1h 58m 12s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/1a4e1be7ee854bc9807e134d3b77f112 : FAILURE in 1h 30m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3e5eb7b79f74e6d95f9baa6d52a92b5 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e8148e3b89240598ccd36ec6939d78a : SUCCESS in 3m 50s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d81ff08b4f44d95a3b6e5f7cf0fb66a : SUCCESS in 2h 33m 01s","accounts_in_message":[],"_revision_number":15},{"id":"804a048e62b9e48cfaf65604dbbdc69418d7559b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-08-27 13:12:50.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"07a9a1429fafac1d7c8503069711a99101452d1c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-28 08:51:34.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"bf04c8510eb23237114031a8059c22bca81c707c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 08:53:49.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"f573085e4f4ccde280305aa303fdb0eae405c807","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 11:23:51.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7266ce816caa4fbd838c9aafeb38cc49\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a60513331295444094bc28e0a5f4202d : SUCCESS in 13m 20s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/bdb0b424dbf44458bcb0e943da5bbfa7 : SUCCESS in 1h 54m 27s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/b85e26c2051c4e77a8e03ae97b13aad4 : SUCCESS in 1h 50m 09s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bbf6c31c41b64771b12047213962e063 : SUCCESS in 1h 34m 55s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1069619720504eb0aa63fb0a239e4b41 : SUCCESS in 1h 29m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc5d780e6edc4ceb949219311f389703 : SUCCESS in 1h 53m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/4754934420ff420aabd6adf0764b935f : SUCCESS in 1h 51m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ea6db9f07054439acfbeeb0325b34be : SUCCESS in 2h 21m 12s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9caba68bef24a138cfe68b13253a3f7 : SUCCESS in 1h 29m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d4177ba1d36648c2808826883a5a7f2c : FAILURE in 2h 08m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90d0074abcfd461e94246df31d72639d : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2268147b92084f379bd50370605ab813 : SUCCESS in 3m 57s\n- openstack-ansible-deploy-aio_lxc_ovnprovider-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6805695b3044a86ab59cb599a8db991 : SUCCESS in 1h 30m 51s","accounts_in_message":[],"_revision_number":15},{"id":"b9c7f3b118d87df41a604e553a48cf81b8ea35b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 11:23:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"e4f85b6d5f92ff20b09b0e03b700c4292a1a125e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 11:24:45.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0cb3c2f25b142aab8fe5986ac9c73b4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0646949e04e547b993c108be2244c8f7 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"766920bfd06fbe1e8b72d483476a8c56d0d3a56d","revisions":{"e3e380347846b7e7430d78a9bd15633643fb077d":{"kind":"REWORK","_number":1,"created":"2026-06-09 17:50:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/992498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/98/992498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/1"}}},"commit":{"parents":[{"commit":"ed442f1769b694c9b64ce43660dfa1af7a5eed2e","subject":"Merge \"Add details about disable Octavia coordination temporarily due to ZooKeeper instability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed442f1769b694c9b64ce43660dfa1af7a5eed2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:50:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:50:12.000000000","tz":0},"subject":"CI: test results","message":"CI: test results\n\nChange-Id: I9fb2374e36c808f3c6d15c757c70f0a1c34904d9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e3e380347846b7e7430d78a9bd15633643fb077d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e3e380347846b7e7430d78a9bd15633643fb077d"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"c78817102ace77a89c570683bfc57e5c7cece8a4":{"kind":"REWORK","_number":2,"created":"2026-06-09 17:50:48.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/992498/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/98/992498/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/2"}}},"commit":{"parents":[{"commit":"ed442f1769b694c9b64ce43660dfa1af7a5eed2e","subject":"Merge \"Add details about disable Octavia coordination temporarily due to ZooKeeper instability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed442f1769b694c9b64ce43660dfa1af7a5eed2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:50:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:50:48.000000000","tz":0},"subject":"CI: test results","message":"CI: test results\n\nChange-Id: I9fb2374e36c808f3c6d15c757c70f0a1c34904d9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/c78817102ace77a89c570683bfc57e5c7cece8a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/c78817102ace77a89c570683bfc57e5c7cece8a4"}]},"branch":"refs/heads/master"},"4a0077018fe64a895210f7292db35eb94bbd6622":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-06-09 17:52:27.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/992498/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/98/992498/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/3"}}},"commit":{"parents":[{"commit":"ed442f1769b694c9b64ce43660dfa1af7a5eed2e","subject":"Merge \"Add details about disable Octavia coordination temporarily due to ZooKeeper instability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed442f1769b694c9b64ce43660dfa1af7a5eed2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:50:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:52:27.000000000","tz":0},"subject":"CI: test results","message":"CI: test results\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/992418\nChange-Id: I9fb2374e36c808f3c6d15c757c70f0a1c34904d9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4a0077018fe64a895210f7292db35eb94bbd6622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4a0077018fe64a895210f7292db35eb94bbd6622"}]},"branch":"refs/heads/master","description":"Edit commit message"},"607a8930fe503ac224daccdbec167428eea0bf4a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-06-09 17:54:33.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/992498/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/98/992498/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/4"}}},"commit":{"parents":[{"commit":"ed442f1769b694c9b64ce43660dfa1af7a5eed2e","subject":"Merge \"Add details about disable Octavia coordination temporarily due to ZooKeeper instability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed442f1769b694c9b64ce43660dfa1af7a5eed2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:50:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:54:33.000000000","tz":0},"subject":"CI: test results","message":"CI: test results\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/992418\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-galera_server/+/992476\nChange-Id: I9fb2374e36c808f3c6d15c757c70f0a1c34904d9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/607a8930fe503ac224daccdbec167428eea0bf4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/607a8930fe503ac224daccdbec167428eea0bf4a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f80ee949db04f36636a92bce02516ce6d70fa7fb":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-06-10 10:14:32.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/992498/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/98/992498/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/5"}}},"commit":{"parents":[{"commit":"ed442f1769b694c9b64ce43660dfa1af7a5eed2e","subject":"Merge \"Add details about disable Octavia coordination temporarily due to ZooKeeper instability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed442f1769b694c9b64ce43660dfa1af7a5eed2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:50:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-10 10:14:32.000000000","tz":0},"subject":"CI: test results","message":"CI: test results\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-galera_server/+/992476\nChange-Id: I9fb2374e36c808f3c6d15c757c70f0a1c34904d9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f80ee949db04f36636a92bce02516ce6d70fa7fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f80ee949db04f36636a92bce02516ce6d70fa7fb"}]},"branch":"refs/heads/master","description":"Edit commit message"},"283cbe644cb3bfbbf76c048ad13d62bb63b6440b":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-06-10 15:52:46.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/992498/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/98/992498/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/6"}}},"commit":{"parents":[{"commit":"ed442f1769b694c9b64ce43660dfa1af7a5eed2e","subject":"Merge \"Add details about disable Octavia coordination temporarily due to ZooKeeper instability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed442f1769b694c9b64ce43660dfa1af7a5eed2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:50:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-10 15:52:46.000000000","tz":0},"subject":"CI: test results","message":"CI: test results\n\nChange-Id: I9fb2374e36c808f3c6d15c757c70f0a1c34904d9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/283cbe644cb3bfbbf76c048ad13d62bb63b6440b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/283cbe644cb3bfbbf76c048ad13d62bb63b6440b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"59ede226c9e121d00958ad8958a365e9769e8ba7":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-06-11 12:06:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/98/992498/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/98/992498/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/7"}}},"commit":{"parents":[{"commit":"ed442f1769b694c9b64ce43660dfa1af7a5eed2e","subject":"Merge \"Add details about disable Octavia coordination temporarily due to ZooKeeper instability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed442f1769b694c9b64ce43660dfa1af7a5eed2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:50:12.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-06-11 12:06:27.000000000","tz":0},"subject":"CI: test results","message":"CI: test results\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/992414\nChange-Id: I9fb2374e36c808f3c6d15c757c70f0a1c34904d9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/59ede226c9e121d00958ad8958a365e9769e8ba7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/59ede226c9e121d00958ad8958a365e9769e8ba7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9a942ba950f8e8c805a1714337d47b16c7a6b129":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2026-06-30 12:12:35.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/992498/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/98/992498/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/8"}}},"commit":{"parents":[{"commit":"ed442f1769b694c9b64ce43660dfa1af7a5eed2e","subject":"Merge \"Add details about disable Octavia coordination temporarily due to ZooKeeper instability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed442f1769b694c9b64ce43660dfa1af7a5eed2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:50:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-30 12:12:35.000000000","tz":0},"subject":"CI: test results","message":"CI: test results\n\nChange-Id: I9fb2374e36c808f3c6d15c757c70f0a1c34904d9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9a942ba950f8e8c805a1714337d47b16c7a6b129"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9a942ba950f8e8c805a1714337d47b16c7a6b129"}]},"branch":"refs/heads/master","description":"Edit commit message"},"df04260ef165ee7c7dd45989fa10baafdbaf2177":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-06-30 12:12:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/992498/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/98/992498/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/9"}}},"commit":{"parents":[{"commit":"a2017a507ab1ef0734a2f5e34e8fb5faa6d6aa5e","subject":"docs: add Cinder storage configuration section for Octavia Amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a2017a507ab1ef0734a2f5e34e8fb5faa6d6aa5e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:50:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-30 12:12:38.000000000","tz":0},"subject":"CI: test results","message":"CI: test results\n\nChange-Id: I9fb2374e36c808f3c6d15c757c70f0a1c34904d9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/df04260ef165ee7c7dd45989fa10baafdbaf2177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/df04260ef165ee7c7dd45989fa10baafdbaf2177"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"ed442f1769b694c9b64ce43660dfa1af7a5eed2e","ours":"9a942ba950f8e8c805a1714337d47b16c7a6b129","theirs":"a2017a507ab1ef0734a2f5e34e8fb5faa6d6aa5e","merge_strategy":"recursive","contains_conflicts":false}},"f92a689394f5cec51c6f06fdaefe8f0d23585c6a":{"kind":"REWORK","_number":10,"created":"2026-07-28 08:53:29.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/992498/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/98/992498/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/10"}}},"commit":{"parents":[{"commit":"a2017a507ab1ef0734a2f5e34e8fb5faa6d6aa5e","subject":"docs: add Cinder storage configuration section for Octavia Amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a2017a507ab1ef0734a2f5e34e8fb5faa6d6aa5e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:50:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-28 08:53:28.000000000","tz":0},"subject":"CI: test results","message":"CI: test results\n\nChange-Id: I9fb2374e36c808f3c6d15c757c70f0a1c34904d9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f92a689394f5cec51c6f06fdaefe8f0d23585c6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f92a689394f5cec51c6f06fdaefe8f0d23585c6a"}]},"branch":"refs/heads/master"},"7c7b3c4b718e4b31e33bc77111636aa6f24d8705":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2026-07-28 08:54:36.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/992498/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/98/992498/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/11"}}},"commit":{"parents":[{"commit":"a2017a507ab1ef0734a2f5e34e8fb5faa6d6aa5e","subject":"docs: add Cinder storage configuration section for Octavia Amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a2017a507ab1ef0734a2f5e34e8fb5faa6d6aa5e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:50:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-28 08:54:36.000000000","tz":0},"subject":"Return ZooKeeper","message":"Return ZooKeeper\n\nZooKeeper back to live from Octavia side.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/993556\nChange-Id: I9fb2374e36c808f3c6d15c757c70f0a1c34904d9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7c7b3c4b718e4b31e33bc77111636aa6f24d8705"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7c7b3c4b718e4b31e33bc77111636aa6f24d8705"}]},"branch":"refs/heads/master","description":"Edit commit message"},"63310f607a34611e8e8f68afd0c580065c2e1af5":{"kind":"REWORK","_number":12,"created":"2026-07-28 08:56:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/992498/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/98/992498/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/12"}}},"commit":{"parents":[{"commit":"a2017a507ab1ef0734a2f5e34e8fb5faa6d6aa5e","subject":"docs: add Cinder storage configuration section for Octavia Amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a2017a507ab1ef0734a2f5e34e8fb5faa6d6aa5e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:50:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-28 08:55:59.000000000","tz":0},"subject":"Return ZooKeeper","message":"Return ZooKeeper\n\nZooKeeper back to live from Octavia side.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/993556\nChange-Id: I9fb2374e36c808f3c6d15c757c70f0a1c34904d9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/63310f607a34611e8e8f68afd0c580065c2e1af5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/63310f607a34611e8e8f68afd0c580065c2e1af5"}]},"branch":"refs/heads/master"},"f356c058db926d1c6d2b8bc989014259c8b0abe8":{"kind":"REWORK","_number":13,"created":"2026-08-09 11:06:06.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/98/992498/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/98/992498/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/13"}}},"commit":{"parents":[{"commit":"a2017a507ab1ef0734a2f5e34e8fb5faa6d6aa5e","subject":"docs: add Cinder storage configuration section for Octavia Amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a2017a507ab1ef0734a2f5e34e8fb5faa6d6aa5e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:50:12.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-08-09 11:05:50.000000000","tz":0},"subject":"Return ZooKeeper","message":"Return ZooKeeper\n\nZooKeeper back to live from Octavia side.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/993556\nChange-Id: I9fb2374e36c808f3c6d15c757c70f0a1c34904d9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f356c058db926d1c6d2b8bc989014259c8b0abe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f356c058db926d1c6d2b8bc989014259c8b0abe8"}]},"branch":"refs/heads/master"},"416b4db04354252f36dfc5ee0221ecf8f9984c74":{"kind":"REWORK","_number":14,"created":"2026-08-09 11:06:35.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/98/992498/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/98/992498/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/14"}}},"commit":{"parents":[{"commit":"a2017a507ab1ef0734a2f5e34e8fb5faa6d6aa5e","subject":"docs: add Cinder storage configuration section for Octavia Amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a2017a507ab1ef0734a2f5e34e8fb5faa6d6aa5e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:50:12.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-08-09 11:06:31.000000000","tz":0},"subject":"Return ZooKeeper","message":"Return ZooKeeper\n\nZooKeeper back to live from Octavia side.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/993556\nChange-Id: I9fb2374e36c808f3c6d15c757c70f0a1c34904d9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/416b4db04354252f36dfc5ee0221ecf8f9984c74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/416b4db04354252f36dfc5ee0221ecf8f9984c74"}]},"branch":"refs/heads/master"},"766920bfd06fbe1e8b72d483476a8c56d0d3a56d":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-08-24 09:34:13.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/992498/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/98/992498/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/98/992498/15"}}},"commit":{"parents":[{"commit":"4bdb77626e0bf8b8723e4bddefcd4bfe740a2815","subject":"Set resolute distro jobs to NV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4bdb77626e0bf8b8723e4bddefcd4bfe740a2815"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 17:50:12.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-08-24 09:34:13.000000000","tz":0},"subject":"Return ZooKeeper","message":"Return ZooKeeper\n\nZooKeeper back to live from Octavia side.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/993556\nChange-Id: I9fb2374e36c808f3c6d15c757c70f0a1c34904d9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/766920bfd06fbe1e8b72d483476a8c56d0d3a56d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/766920bfd06fbe1e8b72d483476a8c56d0d3a56d"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"a2017a507ab1ef0734a2f5e34e8fb5faa6d6aa5e","ours":"416b4db04354252f36dfc5ee0221ecf8f9984c74","theirs":"4bdb77626e0bf8b8723e4bddefcd4bfe740a2815","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
