)]}'
{"id":"openstack%2Fopenstack-ansible-os_panko~587296","triplet_id":"openstack%2Fopenstack-ansible-os_panko~master~I9d02fdf53a6b73e9fbaba82438f9605c8a841af2","project":"openstack/openstack-ansible-os_panko","branch":"master","hashtags":[],"change_id":"I9d02fdf53a6b73e9fbaba82438f9605c8a841af2","subject":"remove unnecessary \u0027\u003d\u0027","status":"MERGED","created":"2018-07-31 06:25:08.000000000","updated":"2020-03-30 02:50:43.000000000","submitted":"2020-03-30 02:49:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"587296-1585536557294-96258ce0","meta_rev_id":"c53a8ca29b042f947e69097daf4b7b734e492b34","_number":587296,"virtual_id_number":587296,"owner":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-30 02:49:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2018-08-03 04:22:37.000000000","_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"value":0,"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2020-03-30 02:41:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"value":-1,"date":"2018-08-03 04:28:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2020-03-30 02:41:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"value":0,"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"value":0,"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-08-03 04:28:55.000000000","updated_by":{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"},"reviewer":{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"},"state":"REVIEWER"},{"updated":"2020-03-30 02:41:02.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-03-30 02:49:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"41017db10fd0b7cef49041809685410c3a167a95","author":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"date":"2018-07-31 06:25:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcb57046f990351dc1832b7ade4e1e5a2f0e3fa3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-31 06:32:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/96/587296/1/check/build-openstack-sphinx-docs/d032345/html/ : SUCCESS in 3m 25s\n- openstack-ansible-linters http://logs.openstack.org/96/587296/1/check/openstack-ansible-linters/56a0c62/ : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1bf7e203c2041bfb3c5816b8fbb9e6617ee18217","author":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"date":"2018-08-03 04:22:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1034acd269ced1a507770841024dc754b6e36dda","author":{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"},"date":"2018-08-03 04:28:55.000000000","message":"Patch Set 1: Code-Review-1\n\nI don\u0027t think it is a relevant fix.","accounts_in_message":[],"_revision_number":1},{"id":"2d868fcd6a0417a4bdd8f67f18c72abfd573193a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-03 04:30:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/96/587296/1/check/build-openstack-sphinx-docs/fbdf0a9/html/ : SUCCESS in 2m 35s\n- openstack-ansible-linters http://logs.openstack.org/96/587296/1/check/openstack-ansible-linters/ba694dc/ : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a9071e455660b505d184d6ba0aff83a35c4c7f7b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-03-30 02:41:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"60750d66c757c625ea34615f3925811509c709d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 02:41:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4ff565c593e02c01543eb11159c9e2e228acc6e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 02:49:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"419f2619cafa50e3aad42c78784468d377f2ff80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 02:49:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d9107e4e5d884ea698fb6198f14ff5a1 : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b58f3d40475841f69a1d35ed5baece31 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":1},{"id":"c53a8ca29b042f947e69097daf4b7b734e492b34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 02:50:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ea378d0140648129e26c17bf381f3d9 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a5d8f3538554fe491197d21029b4485d9e446ee","revisions":{"3a5d8f3538554fe491197d21029b4485d9e446ee":{"kind":"REWORK","_number":1,"created":"2018-07-31 06:25:08.000000000","uploader":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"ref":"refs/changes/96/587296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/96/587296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/96/587296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/96/587296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/96/587296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/96/587296/1"}}},"commit":{"parents":[{"commit":"7a7e2be17353e004b2710a1b6e187ca0b930b8fa","subject":"Merge \"Remove the unnecessary space\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/7a7e2be17353e004b2710a1b6e187ca0b930b8fa"}]}],"author":{"name":"wangqi","email":"wang.qi@99cloud.net","date":"2018-07-31 06:24:43.000000000","tz":0},"committer":{"name":"wangqi","email":"wang.qi@99cloud.net","date":"2018-07-31 06:25:06.000000000","tz":0},"subject":"remove unnecessary \u0027\u003d\u0027","message":"remove unnecessary \u0027\u003d\u0027\n\nChange-Id: I9d02fdf53a6b73e9fbaba82438f9605c8a841af2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/3a5d8f3538554fe491197d21029b4485d9e446ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/3a5d8f3538554fe491197d21029b4485d9e446ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
