)]}'
{"id":"openstack%2Fopenstack-ansible-os_panko~654528","triplet_id":"openstack%2Fopenstack-ansible-os_panko~stable%2Fstein~I7c5eccf226094941476e3c4c74f486538021f69d","project":"openstack/openstack-ansible-os_panko","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I7c5eccf226094941476e3c4c74f486538021f69d","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 15:39:34.000000000","updated":"2019-04-22 20:21:25.000000000","submitted":"2019-04-22 20:21:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654528-1555964485522-4f1731f4","meta_rev_id":"34280063c17288088bfdf56eebfc734a9a119543","_number":654528,"virtual_id_number":654528,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-04-22 20:21:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-04-22 16:15:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-04-22 16:15:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 16:15:45.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-04-22 20:21:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2982cfccb8d2dabfdebe944879b0c6ea486359b8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 15:39:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35095a8a9e7054a27229c35fab81a4e5786e1b76","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 16:15:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1ac813983a3ec226d62d4e3249c9f15f6242a39a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:33:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/28/654528/1/check/openstack-ansible-linters/ecb8724/ : SUCCESS in 5m 20s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/28/654528/1/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/6921128/ : FAILURE in 28m 29s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/28/654528/1/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/e90ef59/ : FAILURE in 21m 52s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/28/654528/1/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/fc8193c/ : FAILURE in 22m 11s\n- openstack-tox-docs http://logs.openstack.org/28/654528/1/check/openstack-tox-docs/aa4bf0a/html/ : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"047d6099a0d11238ba09f270e320e36710b1501b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 16:44:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d572c07f46c3d26e1c4f7d9900685d99303772a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:43:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/28/654528/1/check/openstack-ansible-linters/c021a96/ : SUCCESS in 6m 12s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/28/654528/1/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/1921399/ : SUCCESS in 1h 46m 26s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/28/654528/1/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/2ca3c20/ : SUCCESS in 1h 09m 10s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/28/654528/1/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/7162138/ : SUCCESS in 1h 07m 37s\n- openstack-tox-docs http://logs.openstack.org/28/654528/1/check/openstack-tox-docs/75a35cf/html/ : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ac3857ae12f0748df281bace610247bf572535c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:43:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"91e5cf0e2ce7401fbc78068bf9494b523003f121","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:21:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/28/654528/1/gate/openstack-ansible-linters/10730fa/ : SUCCESS in 5m 01s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/28/654528/1/gate/openstack-ansible-deploy-aio_telemetry_metal-centos-7/f1c4b27/ : SUCCESS in 1h 32m 13s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/28/654528/1/gate/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/c1d2553/ : SUCCESS in 1h 10m 10s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/28/654528/1/gate/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/9efc336/ : SUCCESS in 1h 10m 11s\n- openstack-tox-docs http://logs.openstack.org/28/654528/1/gate/openstack-tox-docs/4178bb2/html/ : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c0f33467484dc480e4212f7043de325d326f2da5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:21:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8be87262f89ac9d6edfe093e4e4ae0e929acd69b","revisions":{"8be87262f89ac9d6edfe093e4e4ae0e929acd69b":{"kind":"REWORK","_number":1,"created":"2019-04-22 15:39:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/654528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/28/654528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/28/654528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/28/654528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/28/654528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/28/654528/1"}}},"commit":{"parents":[{"commit":"44ddb4c2084b002a7de4c03d2e4988b8ab5a6a36","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/44ddb4c2084b002a7de4c03d2e4988b8ab5a6a36"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 15:39:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 15:39:32.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I7c5eccf226094941476e3c4c74f486538021f69d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/8be87262f89ac9d6edfe093e4e4ae0e929acd69b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/8be87262f89ac9d6edfe093e4e4ae0e929acd69b"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
