)]}'
{"id":"openstack%2Fopenstack-ansible-os_panko~657264","triplet_id":"openstack%2Fopenstack-ansible-os_panko~master~I460064f0fc17f2abe2b600d8722e7da2e56d9baf","project":"openstack/openstack-ansible-os_panko","branch":"master","hashtags":[],"change_id":"I460064f0fc17f2abe2b600d8722e7da2e56d9baf","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"ABANDONED","created":"2019-05-05 13:11:43.000000000","updated":"2019-08-15 00:39:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1abc8241a60a1b7d05c56813bbc3e46cce199a96","_number":657264,"virtual_id_number":657264,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-05-06 12:54:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-06 12:54:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"342b5cceb75d27311303ee647cd238737a85cdbb","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-05 13:11:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e1028b9cf2f544cda02cba88fc15a312d749747","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 15:00:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/64/657264/1/check/openstack-ansible-linters/19c5833/ : SUCCESS in 4m 31s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/64/657264/1/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/ccd5d6c/ : SUCCESS in 1h 43m 00s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/64/657264/1/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/8fe3101/ : SUCCESS in 1h 14m 41s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/64/657264/1/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/cb553d7/ : SUCCESS in 1h 11m 07s\n- openstack-tox-docs http://logs.openstack.org/64/657264/1/check/openstack-tox-docs/2ea8388/html/ : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":1},{"id":"93c2e8d60d79d4d40601cb6bc7574d27ce00300c","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-06 11:26:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1eb7df60e9a8c0d69c327899dd90a52e53cdea18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 12:54:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/64/657264/2/check/openstack-ansible-linters/0f08e58/ : SUCCESS in 5m 05s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/64/657264/2/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/5f01212/ : SUCCESS in 1h 19m 25s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/64/657264/2/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/8b48250/ : SUCCESS in 1h 03m 32s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/64/657264/2/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/56856b6/ : SUCCESS in 1h 01m 54s\n- openstack-tox-docs http://logs.openstack.org/64/657264/2/check/openstack-tox-docs/7ef169c/html/ : SUCCESS in 2m 34s\n- build-openstack-releasenotes http://logs.openstack.org/64/657264/2/check/build-openstack-releasenotes/cbc8c18/html/ : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":2},{"id":"1abc8241a60a1b7d05c56813bbc3e46cce199a96","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-15 00:39:16.000000000","message":"Abandoned\n\ndone by I352d0d3a91a3ef67d61e1c45d5b97d4f3a84972c","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4d4a628377900a87ec40d332bae9da5d391a215d","revisions":{"a4d47125b60a50b6293bb2fb2ee222f505d62301":{"kind":"REWORK","_number":1,"created":"2019-05-05 13:11:43.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/64/657264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/64/657264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/64/657264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/64/657264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/64/657264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/64/657264/1"}}},"commit":{"parents":[{"commit":"94ab4d0e32daa070240a42fae75f6c34d56c5f07","subject":"Merge \"Trivial: Fix the pep8 warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/94ab4d0e32daa070240a42fae75f6c34d56c5f07"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 13:11:17.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 13:11:41.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I460064f0fc17f2abe2b600d8722e7da2e56d9baf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/a4d47125b60a50b6293bb2fb2ee222f505d62301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/a4d47125b60a50b6293bb2fb2ee222f505d62301"}]},"branch":"refs/heads/master"},"4d4a628377900a87ec40d332bae9da5d391a215d":{"kind":"REWORK","_number":2,"created":"2019-05-06 11:26:37.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/64/657264/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/64/657264/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/64/657264/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/64/657264/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/64/657264/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/64/657264/2"}}},"commit":{"parents":[{"commit":"94ab4d0e32daa070240a42fae75f6c34d56c5f07","subject":"Merge \"Trivial: Fix the pep8 warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/94ab4d0e32daa070240a42fae75f6c34d56c5f07"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 13:11:17.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-06 11:26:22.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I460064f0fc17f2abe2b600d8722e7da2e56d9baf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/4d4a628377900a87ec40d332bae9da5d391a215d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/4d4a628377900a87ec40d332bae9da5d391a215d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
