)]}'
{"id":"openstack%2Fopenstack-ansible-os_panko~658610","triplet_id":"openstack%2Fopenstack-ansible-os_panko~stable%2Fstein~Ia00a2c0f2b32c201a62edbf3f3c9ce9b6d67580d","project":"openstack/openstack-ansible-os_panko","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ia00a2c0f2b32c201a62edbf3f3c9ce9b6d67580d","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-11 23:59:27.000000000","updated":"2019-05-12 12:08:39.000000000","submitted":"2019-05-12 12:08:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658610-1557662919214-0f9a02b3","meta_rev_id":"980c7c7f957cfe62e151fbabe3c793e53f1ecb30","_number":658610,"virtual_id_number":658610,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-12 08:10:01.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-12 12:08:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-05-12 01:56:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":1,"date":"2019-05-12 01:56:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-12 01:56:56.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-12 08:10:01.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-05-12 12:08:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fed343cb16a85ac98f6690446711e89d656cdd6f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-11 23:59:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a0bfd0d38c26cfa942d9cf98a15c4a0156ca0f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:36:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/10/658610/1/check/openstack-ansible-linters/f476ccc/ : SUCCESS in 6m 12s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/10/658610/1/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/aad5595/ : SUCCESS in 1h 31m 29s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/10/658610/1/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/e7b9cf6/ : SUCCESS in 1h 08m 08s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/10/658610/1/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/4b37604/ : SUCCESS in 1h 02m 26s\n- openstack-tox-docs http://logs.openstack.org/10/658610/1/check/openstack-tox-docs/c6f433b/html/ : SUCCESS in 3m 29s\n- build-openstack-releasenotes http://logs.openstack.org/10/658610/1/check/build-openstack-releasenotes/d34e548/html/ : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0dfd31b2f110974fb2b13621d3a1a044fb971381","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-12 01:56:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f7dc10a77152a460bb665161a69c87c3dcde769","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:57:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34d661ead07920680f6ee40a85ba66d26876b245","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 03:09:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/10/658610/1/gate/openstack-ansible-linters/6dd50eb/ : SUCCESS in 5m 14s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/10/658610/1/gate/openstack-ansible-deploy-aio_telemetry_metal-centos-7/0be1376/ : FAILURE in 37m 30s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/10/658610/1/gate/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/ab79deb/ : SUCCESS in 1h 05m 10s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/10/658610/1/gate/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/148b04d/ : SUCCESS in 1h 07m 01s\n- openstack-tox-docs http://logs.openstack.org/10/658610/1/gate/openstack-tox-docs/2463676/html/ : SUCCESS in 2m 33s\n- build-openstack-releasenotes http://logs.openstack.org/10/658610/1/gate/build-openstack-releasenotes/73a224f/html/ : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f8b010aa0cdbc19c7d33037a60cd219c997d47b7","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-12 08:10:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a9b886f4088db0d82e298280f61aec89eb5ddbb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 10:02:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/10/658610/1/check/openstack-ansible-linters/c47da79/ : SUCCESS in 5m 44s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/10/658610/1/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/a35df7b/ : SUCCESS in 1h 38m 54s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/10/658610/1/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/41363fa/ : SUCCESS in 1h 05m 03s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/10/658610/1/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/9611078/ : SUCCESS in 1h 07m 09s\n- openstack-tox-docs http://logs.openstack.org/10/658610/1/check/openstack-tox-docs/a4d50ad/html/ : SUCCESS in 2m 41s\n- build-openstack-releasenotes http://logs.openstack.org/10/658610/1/check/build-openstack-releasenotes/4cb44cf/html/ : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":1},{"id":"183f2f751400ccfab1f62566060ec7ed0b289c6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 10:02:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bcc4accef4a43e1b8b0f784acf2dbe84e01673cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 12:08:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/10/658610/1/gate/openstack-ansible-linters/2cfc4f1/ : SUCCESS in 4m 13s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/10/658610/1/gate/openstack-ansible-deploy-aio_telemetry_metal-centos-7/ae27d29/ : SUCCESS in 1h 59m 20s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/10/658610/1/gate/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/96509b6/ : SUCCESS in 1h 10m 07s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/10/658610/1/gate/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/91c17f5/ : SUCCESS in 1h 13m 49s\n- openstack-tox-docs http://logs.openstack.org/10/658610/1/gate/openstack-tox-docs/510f0eb/html/ : SUCCESS in 2m 23s\n- build-openstack-releasenotes http://logs.openstack.org/10/658610/1/gate/build-openstack-releasenotes/0e64866/html/ : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":1},{"id":"6e08d915cb544700fd4e6cf3e6315869e7165242","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 12:08:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f8684a54a8cba97d923bec0d4d18b674ca67269","revisions":{"0f8684a54a8cba97d923bec0d4d18b674ca67269":{"kind":"REWORK","_number":1,"created":"2019-05-11 23:59:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/10/658610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/10/658610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/10/658610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/10/658610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/10/658610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/10/658610/1"}}},"commit":{"parents":[{"commit":"8be87262f89ac9d6edfe093e4e4ae0e929acd69b","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/8be87262f89ac9d6edfe093e4e4ae0e929acd69b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:59:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:59:26.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ia00a2c0f2b32c201a62edbf3f3c9ce9b6d67580d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/0f8684a54a8cba97d923bec0d4d18b674ca67269"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/0f8684a54a8cba97d923bec0d4d18b674ca67269"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
