)]}'
{"id":"openstack%2Fopenstack-ansible-os_panko~664100","triplet_id":"openstack%2Fopenstack-ansible-os_panko~stable%2Fqueens~Ibe48b6de7f08c37f9502ba0c812583e0d6233d1f","project":"openstack/openstack-ansible-os_panko","branch":"stable/queens","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ibe48b6de7f08c37f9502ba0c812583e0d6233d1f","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-07 23:27:28.000000000","updated":"2019-06-08 18:07:49.000000000","submitted":"2019-06-08 18:07:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664100-1560017269512-22ee9161","meta_rev_id":"01f4a06c4831ddf9884770c056bfafa9fe1ea12f","_number":664100,"virtual_id_number":664100,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-06-08 18:07:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-06-08 17:17:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-06-08 17:17:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-08 17:17:35.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-08 18:07:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e15b711493551cd087a56c9f35cf730f417577d6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-07 23:27:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac22bd0fbe3c3176542ebc84865ba265627da843","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 00:22:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/00/664100/1/check/build-openstack-sphinx-docs/453d220/html/ : SUCCESS in 3m 19s\n- openstack-ansible-linters http://logs.openstack.org/00/664100/1/check/openstack-ansible-linters/9c5f771/ : SUCCESS in 4m 51s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/00/664100/1/check/openstack-ansible-functional-centos-7/4c9bcdf/ : SUCCESS in 42m 51s\n- openstack-ansible-functional-opensuse-423-nv http://logs.openstack.org/00/664100/1/check/openstack-ansible-functional-opensuse-423-nv/ee2baa0/ : FAILURE in 9m 05s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/00/664100/1/check/openstack-ansible-functional-ubuntu-xenial/4310f6a/ : SUCCESS in 22m 55s","accounts_in_message":[],"_revision_number":1},{"id":"fbf524930ee9a0130a757fe33d4fc13ca8134c1a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-08 17:17:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e84d7722ef925c057c78e166516fd65fd18f8d3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 17:17:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"127de46cba96e4a1d6c05ff030f5397bf0652819","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 18:07:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d0e5ebd27133398c4016466f54feadd8b6b47b02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 18:07:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/00/664100/1/gate/build-openstack-sphinx-docs/3fc9c8f/html/ : SUCCESS in 3m 17s\n- openstack-ansible-linters http://logs.openstack.org/00/664100/1/gate/openstack-ansible-linters/a94aaef/ : SUCCESS in 4m 42s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/00/664100/1/gate/openstack-ansible-functional-centos-7/a9bfd23/ : SUCCESS in 40m 44s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/00/664100/1/gate/openstack-ansible-functional-ubuntu-xenial/e2a065a/ : SUCCESS in 28m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f2d99b5388587e2ee1b6c0592008ea4fb45c650","revisions":{"5f2d99b5388587e2ee1b6c0592008ea4fb45c650":{"kind":"REWORK","_number":1,"created":"2019-06-07 23:27:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/664100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/00/664100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/00/664100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/00/664100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/00/664100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/00/664100/1"}}},"commit":{"parents":[{"commit":"c2eb3f9aaa08d5b65a3c6ec42e5b3665ad9ee65f","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/c2eb3f9aaa08d5b65a3c6ec42e5b3665ad9ee65f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:27:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:27:24.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ibe48b6de7f08c37f9502ba0c812583e0d6233d1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/5f2d99b5388587e2ee1b6c0592008ea4fb45c650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/5f2d99b5388587e2ee1b6c0592008ea4fb45c650"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
