)]}'
{"id":"openstack%2Fopenstack-ansible-os_panko~675159","triplet_id":"openstack%2Fopenstack-ansible-os_panko~master~I2d849e55a1bb300e3f9290ef173b259f71cf3b45","project":"openstack/openstack-ansible-os_panko","branch":"master","topic":"osa/service_refactor","hashtags":[],"change_id":"I2d849e55a1bb300e3f9290ef173b259f71cf3b45","subject":"service_setup: refactor service setup to a single file","status":"MERGED","created":"2019-08-07 17:23:31.000000000","updated":"2019-08-08 14:43:56.000000000","submitted":"2019-08-08 14:43:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675159-1565275436790-067e4e06","meta_rev_id":"609fef5efc0cf9a7cbde045b0df883efcbe52dd4","_number":675159,"virtual_id_number":675159,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-08-08 14:43:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-08-08 12:47:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-07 21:17:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-08-08 12:47:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 21:17:44.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-08-08 12:47:27.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-08 14:43:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd359f5bd59150298c31640e4e875725c9bf023a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-07 17:23:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6281dedbb068efc6b5d6a41db5b13537087caa77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 19:01:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://logs.opendev.org/59/675159/1/check/openstack-ansible-linters/ba81cce/ : SUCCESS in 5m 35s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/59/675159/1/check/openstack-ansible-deploy-aio_metal-debian-stable/c88bc76/ : SUCCESS in 1h 07m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/59/675159/1/check/openstack-ansible-deploy-aio_metal-centos-7/39e9882/ : SUCCESS in 1h 34m 01s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/59/675159/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/0ff5b63/ : SUCCESS in 1h 19m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/59/675159/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/6c43e2f/ : SUCCESS in 1h 08m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/59/675159/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/97bae7a/ : SUCCESS in 1h 06m 26s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://logs.opendev.org/59/675159/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/08dd16a/ : SUCCESS in 55m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/59/675159/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/9de2a00/ : SUCCESS in 59m 34s\n- openstack-tox-docs https://logs.opendev.org/59/675159/1/check/openstack-tox-docs/bcf2db0/html/ : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b995add59145a296766cc8613531b5bb278fd607","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-08-07 21:17:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cb7d7bac719b665ffbae893bb00f39cc68f074f0","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-08 12:47:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"758f45fa989af37e3dd0d8afa0eb19717cf69ce8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 12:47:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1b284990b6451918933250c794a9baed7428c29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 14:43:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"51e35350c6bd8b4c13e090e6a6cc067f1d6dd8e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 14:43:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://logs.opendev.org/59/675159/1/gate/openstack-ansible-linters/47cb42e/ : SUCCESS in 5m 34s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/59/675159/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/ac3062f/ : SUCCESS in 1h 02m 02s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/59/675159/1/gate/openstack-ansible-deploy-aio_metal-centos-7/91a8a40/ : SUCCESS in 1h 49m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/59/675159/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/03a10a8/ : SUCCESS in 1h 14m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/59/675159/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/106338e/ : SUCCESS in 1h 00m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/59/675159/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/de4b990/ : SUCCESS in 1h 00m 45s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/59/675159/1/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/9f8b46e/ : SUCCESS in 44m 55s\n- openstack-tox-docs https://logs.opendev.org/59/675159/1/gate/openstack-tox-docs/fa94673/html/ : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ece27edf705ea85698a44251b21f720437c85349","revisions":{"ece27edf705ea85698a44251b21f720437c85349":{"kind":"REWORK","_number":1,"created":"2019-08-07 17:23:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/59/675159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/59/675159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/59/675159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/59/675159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/59/675159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/59/675159/1"}}},"commit":{"parents":[{"commit":"50f1a48c6c4cd41cffc74c2751a50ff4cf9fe869","subject":"Migrate panko to use uWSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/50f1a48c6c4cd41cffc74c2751a50ff4cf9fe869"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-07 17:22:40.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-07 17:23:17.000000000","tz":180},"subject":"service_setup: refactor service setup to a single file","message":"service_setup: refactor service setup to a single file\n\nThis patch refactors the openstack user/service/endpoints creation to\nservice_setup.yml which will eventually be managed by\nopenstack-ansible-tests.\n\nChange-Id: I2d849e55a1bb300e3f9290ef173b259f71cf3b45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/ece27edf705ea85698a44251b21f720437c85349"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/ece27edf705ea85698a44251b21f720437c85349"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
