)]}'
{"id":"openstack%2Fopenstack-ansible-os_panko~677949","triplet_id":"openstack%2Fopenstack-ansible-os_panko~stable%2Fqueens~I06454d6eade42ddd76db059c2bb74438fe04f032","project":"openstack/openstack-ansible-os_panko","branch":"stable/queens","topic":"goodbye_42.3","hashtags":[],"change_id":"I06454d6eade42ddd76db059c2bb74438fe04f032","subject":"Remove openSUSE 42.3 job","status":"MERGED","created":"2019-08-22 10:25:44.000000000","updated":"2019-08-27 16:03:48.000000000","submitted":"2019-08-27 16:03:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677949-1566921828568-8c29c65b","meta_rev_id":"2e3a130aa1d045b21cb5a06ed82d33098a0df83f","_number":677949,"virtual_id_number":677949,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-08-27 16:03:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2019-08-27 08:44:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-08-27 12:29:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-08-27 13:02:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2019-08-27 13:02:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 08:44:40.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-08-27 12:29:21.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-08-27 13:02:02.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-27 16:03:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f845dca82cd9a0324089a03696735d758fd3d525","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-08-22 10:25:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"526b7861301720fed31b174dde9837d1d0694d62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 11:23:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f20e80b2757e48af9074d09127b1acb0 : SUCCESS in 4m 45s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b11439bba50c43019deb0a0c1e3ad9e4 : SUCCESS in 8m 08s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/f3037e1f660a4c3bbc812aac96339d9d : POST_FAILURE in 42m 23s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/c897b566f7b6447db90a02ad002a4cbd : SUCCESS in 22m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3045b9ee01a6c5a126eb27def7e3d03140c998c7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-27 08:44:40.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck - post_failure should be fixed now","accounts_in_message":[],"_revision_number":1},{"id":"fbe7eadee65933211407709e41570472f593bb47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 09:31:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0b95a61f151543dc952e9bf0b48de8b3 : SUCCESS in 4m 53s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cc1bcab4415b4667b2314d55aaa4fb5a : SUCCESS in 10m 01s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/08eecdc66c614077a46f350e1c97cf62 : SUCCESS in 46m 00s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/01f271a5b4df435da2ab3fc870b33cad : SUCCESS in 31m 38s","accounts_in_message":[],"_revision_number":1},{"id":"5bd4de50c4487c34f679f9f5c4b35fff6dda8ec0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-27 12:29:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"946b6beaafd00764129dca298bf5aa1e64d12194","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-27 13:02:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ef00222c5ed01eff1e8da8087cd8e3e4e704680f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:02:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"018ba418a695b7434aebfe174f6bdc05ee226758","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 14:10:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3907fcb6626e46468263fe731f40084b : SUCCESS in 3m 56s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c763109e2ef64db0a65415e700a2f431 : SUCCESS in 6m 54s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/b0e5d1d8ae8549da8efef779dd358606 : SUCCESS in 50m 23s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f6e88235b001418f85427464ac3ab849 : FAILURE in 57m 36s","accounts_in_message":[],"_revision_number":1},{"id":"39e74316746c8b5240a760908b3d0c561e536aea","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-27 14:10:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3a528978857be2b0913c7d73de6a69edda5903e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 15:04:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/021f2ad6ece545e6b7c7e79bfc8a5cbd : SUCCESS in 4m 57s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c3e142dd2e0747efa89da3ef8d65a7bb : SUCCESS in 12m 29s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/8ce0937289424be48806163d18af1ca6 : SUCCESS in 45m 50s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/3d4fb7aac4c84390b2f272c6ccd78945 : SUCCESS in 29m 56s","accounts_in_message":[],"_revision_number":1},{"id":"27760de5a37e9a23d70390170b02709a9340e5a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 15:04:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c7a094887b7bb947e2d977dc8fe7905d4c3ac4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 16:03:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"09d44e73888a3f173944b2f350e959d805764850","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 16:03:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4d8d97b1c2cd47189a067ddf968c0567 : SUCCESS in 4m 16s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/49643490a4af45e09b1a7f3b16fd2683 : SUCCESS in 6m 29s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/2e4073456fc645958cf208d14be51573 : SUCCESS in 56m 55s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/5205157e0d25428f9f7c9f9bf577a5f2 : SUCCESS in 29m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"61d61d3b9f2dbd26268fb03ded10f6e8dea4f1c0","revisions":{"61d61d3b9f2dbd26268fb03ded10f6e8dea4f1c0":{"kind":"REWORK","_number":1,"created":"2019-08-22 10:25:44.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/49/677949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/49/677949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/49/677949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/49/677949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/49/677949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/49/677949/1"}}},"commit":{"parents":[{"commit":"5f2d99b5388587e2ee1b6c0592008ea4fb45c650","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/5f2d99b5388587e2ee1b6c0592008ea4fb45c650"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-08-22 10:25:05.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-08-22 10:25:42.000000000","tz":120},"subject":"Remove openSUSE 42.3 job","message":"Remove openSUSE 42.3 job\n\n42.3 will not be maintained anymore and therefore\nthere is no reason to test it. We will remove 42.3 nodes from\nOpenDev CI soon.\n\nChange-Id: I06454d6eade42ddd76db059c2bb74438fe04f032\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/61d61d3b9f2dbd26268fb03ded10f6e8dea4f1c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/61d61d3b9f2dbd26268fb03ded10f6e8dea4f1c0"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
