)]}'
{"id":"openstack%2Fopenstack-ansible-os_panko~680055","triplet_id":"openstack%2Fopenstack-ansible-os_panko~master~I7e7c8c50e3d6cd0c97a8a6ef8f5854fce5d16e96","project":"openstack/openstack-ansible-os_panko","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I7e7c8c50e3d6cd0c97a8a6ef8f5854fce5d16e96","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-09-04 15:17:53.000000000","updated":"2019-09-06 12:31:06.000000000","submitted":"2019-09-06 12:29:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680055-1567772972857-b95e921d","meta_rev_id":"f3321752564319714eb67ed83232b492ba17c7c4","_number":680055,"virtual_id_number":680055,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-09-06 12:29:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-09-05 21:52:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-09-05 21:52:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-05 21:52:41.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-09-06 12:29:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47cec01af2debea24bb490f0c2f16a5d7a254a28","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-04 15:17:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9b7f8cebb1c0379c8b57ddc45d4648c7a6b84a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 17:30:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/67c64464a1b74484938cf92b55cca7d0 : SUCCESS in 9m 03s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/8bdee85e7a7a49f0a0ae88d0eaecc85c : POST_FAILURE in 1h 08m 51s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/90e709adbe9c4efdb22a53e87e1d6273 : SUCCESS in 1h 35m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8572135ca6b140e29e710f40cad8c9a8 : FAILURE in 50m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/738dcc2885864f31a0b885e1816d83e3 : POST_FAILURE in 1h 13m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f4a840f064bc4e889cf8b4f49d64ae3c : POST_FAILURE in 1h 15m 21s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e20a3034c9274381956c8e9680a01209 : RETRY_LIMIT in 5m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f494c89573114d8fbff8a836928bd5bc : POST_FAILURE in 52m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d08d76f6cab4bf3ad2cecc17c3957b9 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"65f3f7fa47d83fd03cb4169b94ce8fdf62f7b75d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 14:45:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b172de396a605982e0ccaec26cfebd2222da361b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 19:14:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c67cfec6e48744f98cab1f642b251242 : SUCCESS in 10m 17s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/83cae4ce84b444df88aed9bd053d9e53 : SUCCESS in 1h 25m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ff55f170102a42a59edd2891177f2127 : SUCCESS in 1h 59m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0d2b20de2da2469fbb7e29735afc67cd : SUCCESS in 1h 11m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1644dd1a1cf4cb6939a813bca5e7696 : SUCCESS in 1h 06m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a0875817ebeb4d21a6ea02f9bcfbb01e : SUCCESS in 1h 00m 52s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4ff8db89ee734a06b2607519a9f99bd5 : FAILURE in 31m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6eb62cd8dba349dcb542890e0477c6e4 : SUCCESS in 52m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59c45688e6f9430fa94de04039ee438b : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"6aba4909ef02e5950cf680d3ca8819cb4fe96ea8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 21:52:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"774a6991186881e45a0bb7fd32f129b8f0abb0a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 21:52:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4cb27e83eb7a62b2e02c3dcaaecee38c4dfd4b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 23:35:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3c2e45b02bf04620b14a1e87659ef0bb : SUCCESS in 6m 24s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/fc2e612e17044f5281e0aa2846ac5a7c : SUCCESS in 1h 24m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/30f186ec52e94715bf5a68786fe2b5a4 : SUCCESS in 1h 32m 54s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/895e0bbe65774e30af57d3ec65c0d441 : SUCCESS in 1h 33m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39b46dc316f946faa2eb56aac8ce236f : SUCCESS in 1h 14m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2a32a21c361d47d5af25e588cd67ea4d : SUCCESS in 1h 10m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f5abe0679c045e9a24515092143541c : FAILURE in 56m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c558c307c0c45458b86031c4a2aa08a : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"2491757186a2c66ed0e92e8df53cc2af7da45ebc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-06 07:09:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1ff92b72c9c0306efc724050aee09cb972aef85e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 09:09:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ec078c5b4f0a4261a5e8a626fd0e376d : SUCCESS in 5m 27s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/5a8f4c6c09ab4210b0b2d3bb75343d10 : SUCCESS in 1h 21m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/afea46aaaf9f4c1b921692e65b485add : FAILURE in 1h 01m 46s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9fb1be6d3d35458f8f54836d034238a4 : SUCCESS in 1h 17m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1248b7f9019413ba3a5415c3b8d2e0d : SUCCESS in 1h 00m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/eb3c1cadec3f45c0a419b0fcbe1255d8 : SUCCESS in 1h 04m 07s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7093dc2608624a6a9212192b3abb737c : FAILURE in 29m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/105fcbafb9624f009a88a132df7343c4 : SUCCESS in 43m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49420340f95643e48b9a19a6f8c16612 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a2bc463ae1b7bab36fcf42625d33e2d26c695136","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-06 09:23:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6e8e0819d096410b97c7d2f4761d84ad34296ea0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 10:49:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b8124b8365024bf7bbb91e0eaf9301bf : SUCCESS in 9m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/ad00bc9bcc2a4617aa3ff8e5725597c5 : SUCCESS in 1h 17m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f85e450af24b460599649c8ea4929348 : SUCCESS in 1h 18m 24s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/18f0db9f3d3643e682c36ee4afd678a7 : SUCCESS in 1h 09m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d2aa6d7bdf94b9681f7ad8a1e18220b : SUCCESS in 1h 04m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a1c9701805fc4cc69e9901df97529052 : SUCCESS in 1h 12m 31s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/bffb0e7e07794aa290cd1ec68228a6f4 : FAILURE in 31m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a78a278b9e7f4fd1963f24182f6e72d0 : SUCCESS in 48m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ce3fce8bed446029851ff5c3fea5545 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a9a4b96e089ecb79c9528814d3e9f65b22dd9da7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 10:49:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3381436b3d896588610015c6d47c6339ff267eed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 12:29:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/97efb75d60204945b73b8a33561d122d : SUCCESS in 6m 22s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/d09b469d91cb4769a5334ded735eba82 : SUCCESS in 1h 00m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4b022e014c734fa1af6e9769e381a120 : SUCCESS in 1h 34m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7ddf9effdb5249c9ac385438a257ee5e : SUCCESS in 1h 12m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a3b5141f04940f7aed6a74f94a12f24 : SUCCESS in 1h 05m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/70dc6db6eb844597bb8fd45967b99dd0 : SUCCESS in 1h 08m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0a1cd38f6814f979dcdad51a345bf8c : SUCCESS in 1h 02m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a8cb451b4074d58a2e3f7cb9f5161e2 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"77b24c845a94c1a9fe68ebb816b72e4582e5c2f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 12:29:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f13ee14d45f2fe49c353986c2edf4aae49540141","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 12:31:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6e08a30dcea4ab8a8a324b761ca4f5f : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b88e8ab09a96679c01bfb85f604a74a0ba12874","revisions":{"1b88e8ab09a96679c01bfb85f604a74a0ba12874":{"kind":"REWORK","_number":1,"created":"2019-09-04 15:17:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/55/680055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/55/680055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/55/680055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/55/680055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/55/680055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/55/680055/1"}}},"commit":{"parents":[{"commit":"93b2596a4ed4a44a823bfe3eba11b2db0ec87b56","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/93b2596a4ed4a44a823bfe3eba11b2db0ec87b56"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 15:17:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 15:17:48.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I7e7c8c50e3d6cd0c97a8a6ef8f5854fce5d16e96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/1b88e8ab09a96679c01bfb85f604a74a0ba12874"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/1b88e8ab09a96679c01bfb85f604a74a0ba12874"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
