)]}'
{"id":"openstack%2Fopenstack-ansible-os_panko~680416","triplet_id":"openstack%2Fopenstack-ansible-os_panko~master~Ied62269392fa8d7b0ddfe6e29f52c34ddbbbce20","project":"openstack/openstack-ansible-os_panko","branch":"master","topic":"ansible-role-uwsgi","hashtags":[],"change_id":"Ied62269392fa8d7b0ddfe6e29f52c34ddbbbce20","subject":"Start using uWSGI role","status":"MERGED","created":"2019-09-05 13:49:08.000000000","updated":"2019-09-06 15:48:04.000000000","submitted":"2019-09-06 15:46:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680416-1567784794164-b95e921d","meta_rev_id":"d0c5d09050a718912171e46f060cce75b4ea2ff1","_number":680416,"virtual_id_number":680416,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2019-09-06 15:46:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-09-06 02:03:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-06 13:55:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-06 13:55:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 02:03:35.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-09-06 13:55:13.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-09-06 15:46:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"655aa7df087863bb751e11f91f36b6bed0b94ec1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 13:49:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01bfa83719ba3935ea2da66fa06ff31bb1061629","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 17:52:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a9b8641963cf4927a6602a48ac9a5c40 : SUCCESS in 6m 46s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/811b2bc9b3ba4bce97537df6d6e956e6 : FAILURE in 1h 15m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/42bbf3c0c83f42029828ab9988831a53 : FAILURE in 1h 31m 42s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0e4d456e455847358a285fb1da6e6781 : FAILURE in 40m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1e2d12280b0468884307fa644e2c211 : FAILURE in 59m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ca95c0b77dd24e14b8e31ec387a01265 : FAILURE in 1h 06m 03s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6f66423cff554eb88e77da01886ce1cc : FAILURE in 38m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/899b9e81b8a44052b1bb75786b8e9bdd : FAILURE in 55m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/639db76955614a6ebdd0ebca93304fd6 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"241dc55108b56ceeb5517282f45265fae616c9c8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 19:41:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b0dbf995ea00c938fdf22750b40592d301dc5a04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 21:24:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/faacd82089624c2ba1aea24f0e1ada81 : SUCCESS in 7m 22s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/26aa5501e592498e9b09966732849a36 : FAILURE in 1h 00m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3f8142ecb5fa4e11bc9c2cc62825c4a4 : FAILURE in 1h 23m 51s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/caead512e18748b3b4feb5114de72530 : FAILURE in 1h 31m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13bf60bd707449059dc3682547720795 : FAILURE in 1h 06m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/41f8d68a28324df886a21ef94249baa0 : FAILURE in 58m 11s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a4b1275b16224711a30e0ebf6ba534fc : FAILURE in 32m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b3bdc19460745fcbe531c70b2ec3c46 : FAILURE in 47m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3fec6fae6d6419aad99c429d1a496f8 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":2},{"id":"7e4ab0a8f53651b05c8d70a7129ce90163ddd33a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 21:45:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dc83d1ea486a40062c4b1ca0f8fc2c37b206a840","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 00:19:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1ea2f1c66002456d90d59128779fd188 : SUCCESS in 5m 42s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/caf53b1e47c7417384f8aa4766ea9b72 : SUCCESS in 1h 15m 24s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/94273f0dccdf43d7bca3ea31c603faa2 : SUCCESS in 1h 35m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4fe576c5e2194005a9c3c3f6e2824cd2 : SUCCESS in 2h 15m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9b2be4bee3a4da1a5858ea915e98232 : SUCCESS in 2h 20m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a72aaf7f16334bc9ab1e6002b0b15f38 : SUCCESS in 1h 06m 22s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4d6ee53c7bea4b649ce5651e94ca3892 : FAILURE in 32m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c0b66eb1b7046c9b5c07584f68c0b4d : SUCCESS in 57m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b23dd26bfdf4425d932f75db5400e0b8 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":3},{"id":"0153f9a8c16c0c8323dd2759cb986ddb001067e0","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-09-06 02:03:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b131867fc7f06ed465e4fc654cdbd888b58faf04","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-06 13:55:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ddaf801b0b07ffd806a8128f449a46f58063b410","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 13:55:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7fc54dce33611ec5ec555c88eca33e957e2d28f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 15:46:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1f12d6c47d79467aaf9d02d07db85397 : SUCCESS in 5m 06s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/73c642d3e52a4e8ca05e1e2ef419bc4e : SUCCESS in 1h 27m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/04a78d8ab56a4cd8ac149c1bc7957524 : SUCCESS in 1h 48m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0ad69fb854c145a3a62e96a78503ff6d : SUCCESS in 1h 18m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9416ad9bd1e410ba5196b92fe73c091 : SUCCESS in 1h 09m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/75ea105353ba4c3d8d65bb0c121a1b43 : SUCCESS in 1h 07m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f5a4fe565ae4ff5b7b5deb20f05c05b : SUCCESS in 56m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04b70a84b0b6424599b18180c9f51af5 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":3},{"id":"a0165e21398ec325aada048a833d80cec0955b89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 15:46:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e63368ceb3ef036efd3bff18a7c7e9de9a032c04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 15:48:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d8737e2646d40f78b70b8dc345213dc : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2cb6ab181522d53c940cf45b5f0cd4787745e9b3","revisions":{"5ece2833503dc9b29519037aafc0effa0bbae3bf":{"kind":"REWORK","_number":1,"created":"2019-09-05 13:49:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/680416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/16/680416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/16/680416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/16/680416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/16/680416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/16/680416/1"}}},"commit":{"parents":[{"commit":"93b2596a4ed4a44a823bfe3eba11b2db0ec87b56","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/93b2596a4ed4a44a823bfe3eba11b2db0ec87b56"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-05 13:48:56.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-05 13:48:56.000000000","tz":180},"subject":"Start using uWSGI role","message":"Start using uWSGI role\n\nMove service to use uWSGI role instead of iternal task for uwsgi\ndeployment. This aims to ease the maintenance of uWSGI and speedup\nmetal deployments as the same uwsgi environment will be used\nacross all services.\n\nChange-Id: Ied62269392fa8d7b0ddfe6e29f52c34ddbbbce20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/5ece2833503dc9b29519037aafc0effa0bbae3bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/5ece2833503dc9b29519037aafc0effa0bbae3bf"}]},"branch":"refs/heads/master"},"ab66bbce05b08018db5a8c5b0877f219d6b560ec":{"kind":"REWORK","_number":2,"created":"2019-09-05 19:41:49.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/680416/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/16/680416/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/16/680416/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/16/680416/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/16/680416/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/16/680416/2"}}},"commit":{"parents":[{"commit":"93b2596a4ed4a44a823bfe3eba11b2db0ec87b56","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/93b2596a4ed4a44a823bfe3eba11b2db0ec87b56"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-05 13:48:56.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-05 19:40:39.000000000","tz":180},"subject":"Start using uWSGI role","message":"Start using uWSGI role\n\nMove service to use uWSGI role instead of iternal task for uwsgi\ndeployment. This aims to ease the maintenance of uWSGI and speedup\nmetal deployments as the same uwsgi environment will be used\nacross all services.\n\nChange-Id: Ied62269392fa8d7b0ddfe6e29f52c34ddbbbce20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/ab66bbce05b08018db5a8c5b0877f219d6b560ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/ab66bbce05b08018db5a8c5b0877f219d6b560ec"}]},"branch":"refs/heads/master"},"2cb6ab181522d53c940cf45b5f0cd4787745e9b3":{"kind":"REWORK","_number":3,"created":"2019-09-05 21:45:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/680416/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/16/680416/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/16/680416/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/16/680416/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/16/680416/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/16/680416/3"}}},"commit":{"parents":[{"commit":"93b2596a4ed4a44a823bfe3eba11b2db0ec87b56","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/93b2596a4ed4a44a823bfe3eba11b2db0ec87b56"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-05 13:48:56.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-05 21:45:45.000000000","tz":180},"subject":"Start using uWSGI role","message":"Start using uWSGI role\n\nMove service to use uWSGI role instead of iternal task for uwsgi\ndeployment. This aims to ease the maintenance of uWSGI and speedup\nmetal deployments as the same uwsgi environment will be used\nacross all services.\n\nChange-Id: Ied62269392fa8d7b0ddfe6e29f52c34ddbbbce20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/2cb6ab181522d53c940cf45b5f0cd4787745e9b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/2cb6ab181522d53c940cf45b5f0cd4787745e9b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
