)]}'
{"id":"openstack%2Fopenstack-ansible-os_panko~683145","triplet_id":"openstack%2Fopenstack-ansible-os_panko~master~Ib59be6310e061c2d30b94302e24e2f9a392c93f5","project":"openstack/openstack-ansible-os_panko","branch":"master","topic":"bind-to-mgmt","hashtags":[],"change_id":"Ib59be6310e061c2d30b94302e24e2f9a392c93f5","subject":"Add global override for service bind address","status":"MERGED","created":"2019-09-19 12:56:04.000000000","updated":"2019-10-02 00:41:02.000000000","submitted":"2019-10-02 00:40:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683145-1569976800825-b42f53c0","meta_rev_id":"540bb582331ccfb440cf83f08b797d447c50c9d9","_number":683145,"virtual_id_number":683145,"owner":{"_account_id":31001,"name":"Andrew Gibb","email":"andrew.gibb@bbc.co.uk","username":"andrewg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 00:40:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-09-19 15:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-01 17:54:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-01 17:54:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-19 15:39:56.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-10-01 17:54:37.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-10-02 00:40:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd6d8e3a81962aadb913258d8ab2e83beeb4b6c5","author":{"_account_id":31001,"name":"Andrew Gibb","email":"andrew.gibb@bbc.co.uk","username":"andrewg"},"date":"2019-09-19 12:56:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3305cde2f41cb26a319a75ddab316ef85b67e999","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-19 13:21:07.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"d8878ce682dc1fa94e8a9f3aac25e18aa00f7409","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-19 13:21:12.000000000","message":"Topic set to bind-to-mgmt","accounts_in_message":[],"_revision_number":1},{"id":"acff420383956b19f72fda7e12ee8a21ff52ed11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 15:03:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/903475ba10ef491c9d373f7bfea16749 : SUCCESS in 7m 12s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/8699dcf237334f3799417890e44d507e : SUCCESS in 1h 30m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6581fc4ba38e442195ba80579796664e : SUCCESS in 1h 49m 45s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/77800236e5f847108330153060852176 : SUCCESS in 1h 26m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79c5424ce6bb435da489f5009a46a947 : SUCCESS in 1h 17m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c4cdd4be90774cc79b1ce0d17d833244 : SUCCESS in 1h 11m 34s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/aba81ae3c490416aa90e789dcf573473 : FAILURE in 32m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93d0181c5717448b848ce6316d6babc6 : SUCCESS in 54m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37f1fda271a445d8831ddb8f5dfb0c72 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"70fd5c19ca1bb04f7725374ecd77ceb4d00679bb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-19 15:39:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"85a546c409e7ecc22262c6a78e8fd7539b1713d7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-01 16:18:17.000000000","message":"Removed Workflow-1 by Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"d2c074049b01abc64be4224038bf9b22024e174d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-01 17:54:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5dc3ba630fcce5d173b0d524c7c1453bfef62b27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 17:54:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1c084e1cb3bbf7218a338170e1c5c12f507a6be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 19:50:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0383411b3d82499aa471b38a0dd01cb6 : SUCCESS in 7m 26s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/21050721eef0436a9d5bffa63449f4b7 : SUCCESS in 1h 30m 13s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e8e5045528c84bc195f517310c27ead7 : SUCCESS in 1h 37m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/59bec63cf04d4f5791b2803aaa67aa80 : FAILURE in 1h 33m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d490c3d3cb3947a290401df053e1337c : SUCCESS in 1h 30m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/975105a832f14ba287fe3ba67d528e83 : SUCCESS in 1h 21m 05s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b1cc10664334784b4d0ce2268e1e8ac : SUCCESS in 58m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/882544b1f358475c9256a94373ede985 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"424cc1ed456db8e9ac36ab9e9c477090eab30e3c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-01 21:09:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"930dff0a6c157a34d6f7f71468eeee26727bb47a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 23:03:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/724f780d8dbd4c02b6ded3c86f7bd81b : SUCCESS in 6m 13s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/e331e42258fd478a9628f6eacd425b8f : SUCCESS in 1h 15m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0b937587136e4549a0ae3c2f4ad45692 : SUCCESS in 1h 40m 14s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3f799dd56749401995d4a1d20f0d8565 : SUCCESS in 1h 31m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c764080765d4825ab901216ec607dd6 : SUCCESS in 1h 20m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/da28df254a86437ca22131f670044e10 : SUCCESS in 1h 21m 55s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/94a86080b80a4d4180d58327f49a897d : SUCCESS in 1h 13m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ce71e1a9fc842c897e8c14b6e67e162 : SUCCESS in 52m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc4e882520424ccaa46571f311231bb7 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"03ada7f960d8f5c4fa119a1ebe38abaa1624f060","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 23:03:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"149860190d44ed38103b69217ea04b0865dc4664","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 00:40:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6fd6a6439fdb473e9000ccfb644cc6e8 : SUCCESS in 4m 35s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/adf5e1d6d12e438abf1e04c2000fbdf4 : SUCCESS in 1h 28m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2a35c7cb281e4b7c8c9cb989f25b8f49 : SUCCESS in 1h 25m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b64c6ee7f7024fe9a583b4eaf8de51e0 : SUCCESS in 1h 16m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25e74bdf646844ed9338c37d9703c0cb : SUCCESS in 1h 01m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a78b325703384e8dacc765f2d4954f76 : SUCCESS in 1h 18m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34acb3d0f36d430f9cbf7854b60515cf : SUCCESS in 58m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d688ff559bde4d54b8cf671c9519c5bd : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e96ab9c1b3e9e45a7b091f8a76c85425084bf358","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 00:40:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"540bb582331ccfb440cf83f08b797d447c50c9d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 00:41:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50e41042dd0d4c9b9c914916ce58a4fc : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c63add31271113ca06d6979f447fab6f1bd734f","revisions":{"5c63add31271113ca06d6979f447fab6f1bd734f":{"kind":"REWORK","_number":1,"created":"2019-09-19 12:56:04.000000000","uploader":{"_account_id":31001,"name":"Andrew Gibb","email":"andrew.gibb@bbc.co.uk","username":"andrewg"},"ref":"refs/changes/45/683145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/45/683145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/45/683145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/45/683145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/45/683145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/45/683145/1"}}},"commit":{"parents":[{"commit":"57d718385c72bb0bf068e6e25cf6c7f2bf4317e7","subject":"Allow venv python interpreter to be overridden","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/57d718385c72bb0bf068e6e25cf6c7f2bf4317e7"}]}],"author":{"name":"Andrew Gibb","email":"andrew.gibb@bbc.co.uk","date":"2019-09-19 12:55:28.000000000","tz":60},"committer":{"name":"Andrew Gibb","email":"andrew.gibb@bbc.co.uk","date":"2019-09-19 12:55:28.000000000","tz":60},"subject":"Add global override for service bind address","message":"Add global override for service bind address\n\nChange-Id: Ib59be6310e061c2d30b94302e24e2f9a392c93f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/5c63add31271113ca06d6979f447fab6f1bd734f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/5c63add31271113ca06d6979f447fab6f1bd734f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
