)]}'
{"id":"openstack%2Fopenstack-ansible-os_panko~694347","triplet_id":"openstack%2Fopenstack-ansible-os_panko~master~I5d2be87d6c44c4cbf9181803b1d1e5456212e30a","project":"openstack/openstack-ansible-os_panko","branch":"master","topic":"osa/opendev","hashtags":[],"change_id":"I5d2be87d6c44c4cbf9181803b1d1e5456212e30a","subject":"Replace git.openstack.org with opendev.org","status":"MERGED","created":"2019-11-14 16:28:58.000000000","updated":"2019-11-19 13:10:44.000000000","submitted":"2019-11-19 13:09:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694347-1574168975263-8d7acffc","meta_rev_id":"e8204068392555f732f8d0952e5f5037aa52df10","_number":694347,"virtual_id_number":694347,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-17 12:12:43.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-19 13:09:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-11-18 16:48:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-18 09:30:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2019-11-18 16:48:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 09:30:18.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-11-18 16:48:52.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-11-19 13:09:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8de082728c497e5da02076de6968ced89cb3cc40","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-14 16:28:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24b5eceb320771c85b0435bb484487d0e5d41654","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 19:24:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6179a0a5a08643eebc0752c4fc332fac : SUCCESS in 7m 07s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/25dd671722cc474b9ec0d3a134b7ce6e : SUCCESS in 1h 19m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d54e8295c86d49209acab78e67e9a4fe : SUCCESS in 1h 36m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/dad0d5ebe3dd48168a162cdca5e0c347 : FAILURE in 23m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/031c2de37c8544119e672cfb6e5aaebb : SUCCESS in 1h 08m 38s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cf303985fd52404aa6996f04c82dbcd3 : SUCCESS in 2h 49m 28s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62c1ed0658d3464b9baf5998f7a855d0 : SUCCESS in 1h 59m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f00903d66b9c47e4bdc7c72fe843268d : SUCCESS in 1h 11m 58s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0ce4819f4efa436c92d2333dd8092e71 : FAILURE in 19m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bcd523b22f624e1eb4c58f44c3a73345 : SUCCESS in 1h 04m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f32d8b0e26c43e8ad705534b62a0199 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3582546eac7e4c4ab284368292913033 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"786aa3f9defa0693f2487293990d0f056787ecb9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-17 12:12:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"af115f1de2158c308460c888aaa51ad44cb60b1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 15:08:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8479fd80efb24b01a6c70dd0a23c7c31 : SUCCESS in 6m 16s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/91a4bdc934254f048a4c6b2bcc2ff9e0 : SUCCESS in 1h 23m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fbb0e615529a48b0a2af4d272cb1b562 : SUCCESS in 1h 30m 09s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/677bda72e5c84469a88f067e9eb848de : SUCCESS in 1h 17m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d64f478d4b6d4ff1886095cbd701d079 : SUCCESS in 1h 24m 39s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/843be7bc615648dc8ad590e21bc95750 : SUCCESS in 2h 51m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eaec4956d12c46f1ad47ca21ee2b149c : SUCCESS in 1h 52m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/15d2abf0d786489e8b2cdee87e5a1105 : SUCCESS in 1h 02m 25s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e478a8a0ccbf49b5af75693d8e7225ae : SUCCESS in 1h 01m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02c9e5c1db3f4f3eae843ef71ab86f6e : SUCCESS in 1h 00m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc19d3af9caa4ab4891e78eb5b101c46 : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08a3f305de154fb6a1d3e6c80e17bb96 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"815224879807aaf43b326a83c9c9c93f492d336e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-18 09:30:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a8e8413fd6f16d60843f93de6a7dd47512458d21","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-11-18 16:48:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c8d6b83f61ca36c1501b38a908c0b512851d84db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 16:49:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d800a020e5ad9775010129f0f7d836980d0d3b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 20:09:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/be3e059620c745cf808a35fa0094204a : SUCCESS in 5m 09s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/67bca5f0e72c4a77906803c7e1ef6f8e : SUCCESS in 1h 27m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7cd9fae5d7014de4b007b7f277cf6642 : SUCCESS in 1h 56m 18s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/340436c521e245c4b200a1dab908445c : SUCCESS in 1h 33m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/885e91b7ec7e46149bcab3d39408f44b : SUCCESS in 1h 44m 58s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a478cde7588848d98abcaee4245e1d58 : TIMED_OUT in 3h 04m 15s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/447125edb2974d7f8c628c277558e522 : SUCCESS in 2h 06m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5ed9bf998b2a4d5192cd4923160def43 : SUCCESS in 1h 25m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e928d9c140054fca976d31b004587c6b : SUCCESS in 1h 00m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98186fd403f3440781fbf8d3f77563b3 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf55a75847d141e19ba1d0f309b8894e : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2248ae4b711ae556da5ef68b4c7ed71bcb1df2e6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-18 21:57:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f1824c9b942b7d2e52fca65ba6b539ad8de692e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 00:23:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3dfeab7da8d84870af342dc541ae7bcd : SUCCESS in 4m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/27ebf7545d6a431184a0f1e7002a01cf : SUCCESS in 2h 01m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/80b48ab3f2ad4445ac680baec55789f2 : SUCCESS in 1h 30m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4850b0d1c97a4131b007aaa54a7555ae : SUCCESS in 1h 32m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9ba1057f26c4c02bd88c1654f30d444 : SUCCESS in 1h 27m 28s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5569bcc0e677403298c66f31990932a4 : SUCCESS in 2h 25m 05s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53a7d0b2d3a34f2cb348dc5fa271e3ed : SUCCESS in 1h 47m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/310d9fa8bf21419bba802e64b0e5b162 : SUCCESS in 1h 14m 15s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cf7a02db1861415e9541e56262eeed3d : FAILURE in 1h 04m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/730e18854aac4063867487cdbdeaffc3 : SUCCESS in 1h 01m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1797bf4246724125a07cceca408cbebe : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/505bbb8988ec49de95cdc5fba143aaf3 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0dbafd50adb647bd2b703129d25c99dfcf623b6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 00:23:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c197706b9dafe007afc946c32c011669d905c289","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 03:27:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f1b343543aff45869d6e8bcfbe46d1cf : SUCCESS in 5m 41s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/fd61aadc586b48c6a77e9282a906160a : SUCCESS in 1h 24m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a40ea8837fad400b87d378ccfe09d70f : SUCCESS in 1h 43m 09s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/afe916673f5f4d1287cd0ffdeaec16e5 : SUCCESS in 1h 29m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b562def7fda543d4aa539a0a840ce4a6 : SUCCESS in 1h 09m 59s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/02c7a568d242400891621fcd498a34a6 : TIMED_OUT in 3h 02m 57s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a71353271ee4f2c91dc8f63eb38fa93 : SUCCESS in 1h 54m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1bdb8c17a1324df1aa72b5a1c00f6709 : SUCCESS in 1h 22m 50s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/024ac6390d2a45dabe69969b38fc4770 : SUCCESS in 1h 05m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/116351cd1562428aa2548e18c7484568 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64e18320f1524566b48f0acedd327507 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f6adecab7ba8ed5e2f02b86d970f13e2610bb2c0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-19 07:04:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"df954297d464b5edd9bc205f03b6e50c2c0e8200","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 10:08:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ca64e6cb1e034ad5a9ce241ed81b8124 : SUCCESS in 7m 53s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/6e8b94dcf369471e93bc0742f4a6b771 : SUCCESS in 1h 33m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2f405115bf1d4f4ebc0d582113eae97f : SUCCESS in 1h 59m 04s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/aee6e049af9b4ff28c7799cf0fbec582 : SUCCESS in 1h 32m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48f2f3a760374a51bf24c5c65753b735 : SUCCESS in 1h 23m 22s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/85d781e0170b4e369fff7fe1140aabac : SUCCESS in 2h 59m 06s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b1bad645b8d4c6da11fb9ca01be1bc0 : SUCCESS in 2h 01m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f08d916c2f70494fb853a970b40e2fab : SUCCESS in 1h 17m 09s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1f237960dc2f4046a96600b464a494bf : FAILURE in 1h 03m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4642c5976f724dd58983195c549a65a3 : SUCCESS in 1h 02m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e2412383f284fc88915aaaebda3b7e0 : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d23f632c44524054ab38e8cad65bb024 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a4a13b2d3073cf2d4a2b9f7435f39f06e3a13343","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 10:09:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"31ba8c88a6ef6d8d06c2ecdf7f252d92f2371140","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 13:09:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b1ca9382f33d3196845904d1eec89a0c908574e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 13:09:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/41017eded83f40f3be0ff0bcc289f305 : SUCCESS in 8m 08s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/035d8684b26b40859820d3d0f36c19b7 : SUCCESS in 1h 20m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a69b3152102a47e0b8223a0cd63ee1d1 : SUCCESS in 1h 57m 52s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9288a74c676f48aa8cb9aa1f098829ec : SUCCESS in 1h 35m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5f3fde3835b45ca8031f60398f76f28 : SUCCESS in 1h 21m 24s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9abab93d61054710836197cb7bc7f1d8 : SUCCESS in 2h 57m 52s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4ca41857d114b1fa6ee60ca247425d1 : SUCCESS in 2h 05m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bad7bb4531c84e3f8ebd4457449f48c4 : SUCCESS in 1h 19m 30s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aeaffee8fd8e416dbd9139307d8be681 : SUCCESS in 1h 08m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc9eaf48df8d4538bf9eeb01bd21fa44 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3661d24a2f34b33b855f8ae84c788bc : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"33fc75585a1ce5da560a3a68e2419a9c5c311205","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 13:10:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2fbb72b1b664899a42ace145fb24576 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbaf42eb345c4e1ea64645baf34529d0 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3df07db023c6f725f9abcf26892029834e558014","revisions":{"3df07db023c6f725f9abcf26892029834e558014":{"kind":"REWORK","_number":1,"created":"2019-11-14 16:28:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/47/694347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/47/694347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/47/694347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/47/694347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/47/694347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/47/694347/1"}}},"commit":{"parents":[{"commit":"25ecd7970909ea9b6b5f2a49ee11d7a57186d2ce","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/25ecd7970909ea9b6b5f2a49ee11d7a57186d2ce"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-14 16:28:47.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-14 16:28:47.000000000","tz":120},"subject":"Replace git.openstack.org with opendev.org","message":"Replace git.openstack.org with opendev.org\n\nThis patch replaces git.openstack.org with opendev.org as redirection\nfrom old path was enabled.\nAlso we change upper constraints url due to [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: I5d2be87d6c44c4cbf9181803b1d1e5456212e30a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/3df07db023c6f725f9abcf26892029834e558014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/3df07db023c6f725f9abcf26892029834e558014"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
