)]}'
{"id":"openstack%2Fopenstack-ansible-os_panko~768628","triplet_id":"openstack%2Fopenstack-ansible-os_panko~master~I66db54ded81cd423c861f38a88fae34ba2982a81","project":"openstack/openstack-ansible-os_panko","branch":"master","hashtags":[],"change_id":"I66db54ded81cd423c861f38a88fae34ba2982a81","subject":"Add Victoria release info","status":"ABANDONED","created":"2020-12-29 02:20:23.000000000","updated":"2021-01-26 17:30:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e6e1ba018b225d89876cf9e9f44fc17e7d529669","_number":768628,"virtual_id_number":768628,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-29 02:32:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-29 02:32:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a074f9ca42574cfe76b5d10543f614e15fd75b5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2020-12-29 02:20:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b1b3a0a9e58b59f05f2a5f604e9ad7deaac7f8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 02:32:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/64023d816901461698e75b69aab20234 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/341edfcb370b426fabf12a5933b51cc8 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6bebe7f9c7a45fd971ebc252e197320 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"e6e1ba018b225d89876cf9e9f44fc17e7d529669","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-26 17:30:59.000000000","message":"Abandoned\n\nWe are no longer publishing releasenotes for this repo","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5a71e398214dcbfd2981b986cdddcf4828d4425","revisions":{"b5a71e398214dcbfd2981b986cdddcf4828d4425":{"kind":"REWORK","_number":1,"created":"2020-12-29 02:20:23.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/28/768628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/28/768628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/28/768628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/28/768628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/28/768628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/28/768628/1"}}},"commit":{"parents":[{"commit":"518112a014c0784106accd5ea27f27fd901351ea","subject":"Merge \"Reduce number of processes on small systems\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/518112a014c0784106accd5ea27f27fd901351ea"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-12-29 02:19:42.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-12-29 02:20:17.000000000","tz":480},"subject":"Add Victoria release info","message":"Add Victoria release info\n\nAdd the lack of release information for Victoria, this patch added it.\n\nChange-Id: I66db54ded81cd423c861f38a88fae34ba2982a81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/b5a71e398214dcbfd2981b986cdddcf4828d4425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/b5a71e398214dcbfd2981b986cdddcf4828d4425"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
