)]}'
{"id":"openstack%2Fopenstack-ansible-os_panko~782285","triplet_id":"openstack%2Fopenstack-ansible-os_panko~master~Ib434e98cd4df42ea77bcb985eafe2e56b1ac6fe9","project":"openstack/openstack-ansible-os_panko","branch":"master","topic":"policy-json-to-yaml","hashtags":[],"change_id":"Ib434e98cd4df42ea77bcb985eafe2e56b1ac6fe9","subject":"[goal] Deprecate the JSON formatted policy file","status":"MERGED","created":"2021-03-22 16:04:45.000000000","updated":"2021-04-19 20:42:03.000000000","submitted":"2021-04-19 20:40:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"782285-policy-json-to-yaml","meta_rev_id":"eddcbc9a828900ea1df2f54a2a339779f56ab80d","_number":782285,"virtual_id_number":782285,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-19 20:40:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-04-19 09:05:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-13 16:47:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-04-19 09:05:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-22 18:23:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-30 15:23:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-04-19 09:05:45.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"b1fe747e2cbf06cac5dcfa80d8469ee0b7c58216","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-22 16:04:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb6c885bd8ba22f0b503e788b5e3b6ae8828b64a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 18:23:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/374ffd06c90847b69b079b3c4acadac3 : SUCCESS in 12m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d61b874a631640149d2db64661adb772 : SUCCESS in 1h 08m 27s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8b1a1dfae9cd443bb83f9e9953ef7622 : SUCCESS in 1h 21m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1825038ff491415a9c5170d5dfc442dc : SUCCESS in 1h 21m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7545871366ff4b048b2d1eeea5210140 : SUCCESS in 1h 12m 39s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d80788b853ba435eaa3ef973851852b4 : SUCCESS in 2h 06m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/466bd675961f4aceaed1f7e2e255ddcf : SUCCESS in 1h 45m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0a60fae0c2914e8db8473a074980674b : FAILURE in 39m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e1f52d4fa3e4be1b6f5ad18dbbc9403 : SUCCESS in 1h 16m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ca11768e74a4562bce86115c27d300b : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f54c17e5ccbcbdb8c11103a433aeb8152930d0e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-24 12:00:57.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"56b69c4469ceb6d1a9344dc2a358455e3325607a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 14:07:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5145c562630a48aba99b6ae24b805f3d : SUCCESS in 11m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ee1ac2e43e7d4e04b248233f8d696708 : SUCCESS in 1h 17m 18s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f79839a2b85a4614b6434626f2ced383 : SUCCESS in 1h 18m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/019e0943c4e64bc08b53afb71355ee1f : SUCCESS in 1h 25m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b96e8d57cd044d1fa73e2ec293855132 : SUCCESS in 59m 30s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cca95cb354c44d3aa90fbbf67e9edb81 : SUCCESS in 1h 59m 18s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5465e0dbd41349c0bf1f88db47cdc6de : SUCCESS in 1h 25m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9aec1a3a353b45a08bf794397d3ac8a4 : SUCCESS in 1h 05m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6912c1747fc4d4e86ed8d6850980a8e : SUCCESS in 55m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/299fa1842fed412b913c890ed7b1d7d2 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":2},{"id":"8dc3af0f727c0e1691382f0fd7f2a7602f8007b4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-30 15:23:26.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b3328265e27019249ab278a70e782cbf17754fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-30 16:43:33.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"50ad7459a0f6e3886000cdbd7628c49e1aae6ab3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 23:40:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/477c7a6915b34f0ca4e4675a8874455d : SUCCESS in 11m 24s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0a76f9efd8254748867d2bda40172fa4 : SUCCESS in 1h 12m 09s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f40d16f60de64a2f9be66345ca538cf9 : SUCCESS in 1h 26m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e422191ca90d459b857450506c65935b : SUCCESS in 1h 26m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8837fd54e70945d39f3d778542c6ead0 : SUCCESS in 1h 12m 29s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/eee950137ce34475a622c7f2f5b6eca4 : SUCCESS in 2h 20m 56s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/590b8cbc1e9f42c1b1985cf8713901e9 : FAILURE in 41m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/11c1806312274687854374f4a54fa7b7 : FAILURE in 34m 37s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cad713d51a834d8480a51575de91d0d4 : SUCCESS in 53m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e8748b5668d4e298d2c3f45d3aeb0d4 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":3},{"id":"3b2830c69c7eef091f1920fb48a3ef83222416d4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-01 19:22:46.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b31b541f9d2acf00958cbe6d77697afb1a4e5073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 21:52:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ffa89c7984bc42e48852091a025a07ae : SUCCESS in 14m 38s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1cb7b46ba88b4ec1907a7424bbbd6bf7 : SUCCESS in 1h 16m 47s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c0c16a7bebe547fe8c2aa383d676ed33 : SUCCESS in 1h 20m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d99754a40684ea9b4d7692ebcec5e25 : SUCCESS in 1h 17m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef248f2bfcf742a38df5175a1c2ee098 : SUCCESS in 1h 19m 37s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/735e3c40671f4bb2b88fab1ffe0d831c : SUCCESS in 2h 08m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fb3ec490cd24fa69f89ec5bf6604dc5 : SUCCESS in 1h 38m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2eb410af757b4f0b898c4bc0622223c3 : SUCCESS in 1h 02m 52s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61d1f518e53444e3bdb13fe9930b1ca3 : SUCCESS in 1h 05m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb8a13db14a3488c9d9c08862b7aabaf : SUCCESS in 8m 18s","accounts_in_message":[],"_revision_number":3},{"id":"d1f1db1f494f5251baf83dcbae484dc029cec87d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-02 06:24:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6c72265f7378fb2327a3c74b05a05b2056062937","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-06 11:34:17.000000000","message":"Removed Code-Review+2 by Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"6c375b2a09d7c5e40d402f45e5c60ee047166af5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2021-04-06 12:22:06.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"9e835a0637d673d43dcd25adbfe734f9e79b7d2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 14:28:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/559db64b9fba4e879838e6afba1173c6 : SUCCESS in 20m 55s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/766522e49dcb470c9eb34462e44775f1 : SUCCESS in 1h 10m 21s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2c8cab323ef34f22890691ebebf9ac51 : SUCCESS in 1h 24m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08eef6630a2b45168a5e0e37616839b2 : SUCCESS in 1h 22m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d50af1087d945a389f4499d1d9ecb1a : SUCCESS in 1h 13m 48s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7d7aa721e9c140a0956f6ae0ab02d0e0 : SUCCESS in 1h 55m 37s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c100baba4734f8db29f456f8b6880b5 : SUCCESS in 1h 38m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2cf4e551f62c4c0f82bd5a06f744c7ce : SUCCESS in 1h 00m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd11c4420df84fea883daa89cde740fc : SUCCESS in 1h 01m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d024b6ac919848c8aee06e536172935b : SUCCESS in 9m 25s","accounts_in_message":[],"_revision_number":4},{"id":"2d2f6be2b8f3c80aad7e332248d62776e82ded61","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-13 16:47:09.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8000e1e09aadf91e3733d7eaed8b68e80b80d0d8","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-04-19 09:05:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"584eac668ddb5145590bce8585d32da39fe43404","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 09:06:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e555e7d3533fe3bb8f715fbd48905087d380f693","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 10:37:56.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3e3a356a9bb843268ea1b281e550a16b : SUCCESS in 12m 56s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7276f45bfdc245eba6db8d743ea790b0 : SUCCESS in 1h 23m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73b46c29aa43478ab1c640434a092394 : SUCCESS in 1h 23m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b50199f4ddd845d3a6d47884c95802c9 : SUCCESS in 1h 12m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb1a92341c384899aced5ae43da7e340 : SUCCESS in 1h 04m 54s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d98f91e64ca495c825e949aaf1fce28 : FAILURE in 41m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/254dbeb0aa124e1295061ab891ed21ee : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":4},{"id":"ca7d54b52c8314b55b072697014e03d648f90e75","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 11:15:56.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"90d58939f2acebbdd774cc92082c77a10e0209a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 13:16:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/37ee74776e014b0fa90e2b82f0822a5e : SUCCESS in 11m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/aa0a10bbb59d4466bd6ceaab349e3b43 : SUCCESS in 1h 12m 42s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2ab49c30432f40aeb7c19a1fbdcaeb5e : SUCCESS in 1h 11m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d12d685b656c400b94a406251ec428c2 : SUCCESS in 1h 21m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7de313de8ca64dd7a82837f43af6b2d2 : SUCCESS in 1h 12m 28s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6ff26b5fecd44fd0b7da41276a682d61 : SUCCESS in 1h 54m 00s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af57d692fec94499b6d1f771e632e059 : SUCCESS in 1h 51m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b7fa175c12f94c20a57d016d9650a721 : SUCCESS in 1h 01m 41s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b031c9d5414b476ba99cdfa1b95f47a1 : SUCCESS in 1h 01m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d993012c65b411d8885d62ad3134f05 : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":4},{"id":"d4805985ef0fa5d6fa7a6902a42cdf9bea8fceb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 13:16:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"02898ad7e5a4452b81f6c11fb982e70ecb0aa52d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 14:48:42.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f673018213004fbda12adbf725e5b687 : FAILURE in 8m 01s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a54fac0dd9c04101b8ac98250afde698 : SUCCESS in 1h 25m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0aad6b6ff3ba4e3289b6311b205abc13 : SUCCESS in 1h 23m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/943aa1d9b5db448fa312b3bf85db11b7 : SUCCESS in 1h 10m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/562ac0884a67406dbb6084b3e781c7ec : SUCCESS in 1h 13m 15s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1caea15805f49a69deec910c3e4a778 : SUCCESS in 56m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af5a5ba8e7b548c38e4b5d07a95c6bb1 : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":4},{"id":"4efa16eff9746fa6ea2f0993e54b6e43cb6fc89f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 16:36:25.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"b01bfe1ccd19b3ac0332af3ea344ac8cc336f0a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 18:59:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1d4c536a71fa41be9af0c6ce510ad87a : SUCCESS in 10m 06s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8ec6c9f6e0fe49e68e0e89c538b8172a : SUCCESS in 1h 13m 42s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/275e9979646e4b8f845e59954581bf3a : SUCCESS in 1h 11m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c3b1ba4bde74fc687efcb6c4eda83d2 : SUCCESS in 1h 39m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb210e5de62542d480212938fd137eec : SUCCESS in 1h 15m 54s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/826118af21e940cbb8e171f97a6cfc7a : SUCCESS in 2h 00m 04s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a237bcc87df4b6f9ae64747bd3c578b : SUCCESS in 2h 10m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e188896e41147ac90d93a943308152e : SUCCESS in 1h 01m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/301af9ef546f470a8cebbbf87d13095f : SUCCESS in 59m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71e8d32d2da84cf1b4e90916f6026dec : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":4},{"id":"1ff845ec8f0b3720ad5ae827e53f622ab01ffa69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 18:59:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"68d369807d7981755a0fc792b3aba01b4ff13e01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 20:40:45.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b5bca1179234467f9afe0062bec8f2dc : SUCCESS in 13m 14s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4a6d032046944a4c8347deb6f02a55dd : SUCCESS in 1h 23m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7a3228f7cab42d691be1c7d29de996b : SUCCESS in 1h 17m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee57a6317d254c5a90f943a8d0377523 : SUCCESS in 1h 35m 10s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c6369936a9a48018ec76f1936ee3bb2 : SUCCESS in 1h 01m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58d5b212b63f451daa95066115c7214e : SUCCESS in 1h 00m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6658ca05fefe4bf89cf736d7625e124e : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":4},{"id":"615b0fc7a4f7929a1b0dc33b5675976d7ab759f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 20:40:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"eddcbc9a828900ea1df2f54a2a339779f56ab80d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 20:42:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50a808c98f44e0bb10dbb607b821f07 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a396d45942eae8a3bd66402aa7e3c7c86e4dd0c9","revisions":{"48c887ad626bc42480d96e2c0e16a5c756e6e577":{"kind":"REWORK","_number":1,"created":"2021-03-22 16:04:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/85/782285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/85/782285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/85/782285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/85/782285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/85/782285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/85/782285/1"}}},"commit":{"parents":[{"commit":"21f767d08b57cf69d57c091b515f04f67c39b465","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/21f767d08b57cf69d57c091b515f04f67c39b465"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-22 16:04:36.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-22 16:04:36.000000000","tz":120},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to replace policy.json to\npolicy.yaml and remove deprecated policy.json.\n\nconfig_template has been choosen instead of the copy, since it can\nproperly handle content that has been lookuped.\n\nWe make a separate task not to restart service when it\u0027s not needed.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: Ib434e98cd4df42ea77bcb985eafe2e56b1ac6fe9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/48c887ad626bc42480d96e2c0e16a5c756e6e577"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/48c887ad626bc42480d96e2c0e16a5c756e6e577"}]},"branch":"refs/heads/master"},"72421ed225b15421f7ff5efd9c43acb3be0e7e0c":{"kind":"REWORK","_number":2,"created":"2021-03-24 12:00:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/85/782285/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/85/782285/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/85/782285/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/85/782285/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/85/782285/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/85/782285/2"}}},"commit":{"parents":[{"commit":"21f767d08b57cf69d57c091b515f04f67c39b465","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/21f767d08b57cf69d57c091b515f04f67c39b465"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-22 16:04:36.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-24 12:00:51.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to replace policy.json to\npolicy.yaml and remove deprecated policy.json.\n\nconfig_template has been choosen instead of the copy, since it can\nproperly handle content that has been lookuped.\n\nWe make a separate task not to restart service when it\u0027s not needed.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: Ib434e98cd4df42ea77bcb985eafe2e56b1ac6fe9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/72421ed225b15421f7ff5efd9c43acb3be0e7e0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/72421ed225b15421f7ff5efd9c43acb3be0e7e0c"}]},"branch":"refs/heads/master"},"d5427bc3c59ef81397c19259a75d34aa7c9db384":{"kind":"REWORK","_number":3,"created":"2021-03-30 16:43:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/85/782285/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/85/782285/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/85/782285/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/85/782285/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/85/782285/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/85/782285/3"}}},"commit":{"parents":[{"commit":"21f767d08b57cf69d57c091b515f04f67c39b465","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/21f767d08b57cf69d57c091b515f04f67c39b465"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-22 16:04:36.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-30 16:43:29.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to replace policy.json to\npolicy.yaml and remove deprecated policy.json.\n\nconfig_template has been choosen instead of the copy, since it can\nproperly handle content that has been lookuped.\n\nWe make a separate task not to restart service when it\u0027s not needed.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: Ib434e98cd4df42ea77bcb985eafe2e56b1ac6fe9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/d5427bc3c59ef81397c19259a75d34aa7c9db384"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/d5427bc3c59ef81397c19259a75d34aa7c9db384"}]},"branch":"refs/heads/master"},"a396d45942eae8a3bd66402aa7e3c7c86e4dd0c9":{"kind":"REWORK","_number":4,"created":"2021-04-06 12:22:06.000000000","uploader":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"ref":"refs/changes/85/782285/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/85/782285/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/85/782285/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/85/782285/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/85/782285/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/85/782285/4"}}},"commit":{"parents":[{"commit":"21f767d08b57cf69d57c091b515f04f67c39b465","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/21f767d08b57cf69d57c091b515f04f67c39b465"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-22 16:04:36.000000000","tz":120},"committer":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2021-04-06 12:21:54.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to replace policy.json to\npolicy.yaml and remove deprecated policy.json.\n\nconfig_template has been choosen instead of the copy, since it can\nproperly handle content that has been lookuped.\n\nWe make a separate task not to restart service when it\u0027s not needed.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: Ib434e98cd4df42ea77bcb985eafe2e56b1ac6fe9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/a396d45942eae8a3bd66402aa7e3c7c86e4dd0c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/a396d45942eae8a3bd66402aa7e3c7c86e4dd0c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
