)]}'
{"id":"openstack%2Fopenstack-ansible-os_panko~827622","triplet_id":"openstack%2Fopenstack-ansible-os_panko~stable%2Fussuri~Iae032c759c6100eacedbd63b85aaf20a321ddae4","project":"openstack/openstack-ansible-os_panko","branch":"stable/ussuri","topic":"osa/remove-centos8","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-02-03 16:25:53.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-02-03 16:25:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iae032c759c6100eacedbd63b85aaf20a321ddae4","subject":"Remove legacy centos-8 CI jobs","status":"MERGED","created":"2022-02-03 10:32:19.000000000","updated":"2022-02-03 16:36:06.000000000","submitted":"2022-02-03 16:25:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827622-osa/remove-centos8","meta_rev_id":"666183832dd4e8193b441c37ca758f482062145b","_number":827622,"virtual_id_number":827622,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-03 16:25:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2022-02-03 14:41:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2022-02-03 14:35:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2022-02-03 14:41:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-03 13:59:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-03 14:35:08.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-02-03 14:41:03.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"b8c50bd1027837ab56f96c9420e5ee2566ef5ea4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-02-03 10:32:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a852e34e6cceaef555f93edb91ad7e4289028c8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-02-03 10:32:52.000000000","message":"Topic set to osa/remove-centos8","accounts_in_message":[],"_revision_number":1},{"id":"70ccb7c9f332784aa54d313705d5b194b544fe52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-03 11:15:35.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6ffc5a4584f61b64c358298e9c69de5ff1f941f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 13:59:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f50c6f63b3da4d5cadd40b78e0907d89 : SUCCESS in 8m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/97c2fb18080c48689c435a9a8171cc43 : SUCCESS in 1h 14m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/77d2c29dd2cb42f79bb113fa9c2304ff : SUCCESS in 1h 35m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a251d3ce8d6c425e83bbf3ab5a0eeb32 : SUCCESS in 1h 16m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7417928e39fb458f93b8df058775e852 : SUCCESS in 1h 10m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7361fd1d38fc445eb236939d1fb58377 : SUCCESS in 2h 37m 21s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb70612f261e4b088ba99bc794b97aa5 : SUCCESS in 1h 49m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c2130c2deee445feacce57d0a95f4c11 : SUCCESS in 1h 07m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f3154e6e7ab4d7abab83725ef419b57 : SUCCESS in 1h 01m 08s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/541490bb7aba4a79bbadb6bbc93b8f12 : SUCCESS in 55m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeb0a0984b4f415bbce1e052916e4477 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":2},{"id":"358a8cd9bd14a5fbfe9c832a22cbe235e850b305","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-03 14:35:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ca369155b26f4e2d3544d59f1aa526b96dd4bfbf","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-02-03 14:41:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"be57f1565da80363442b9eb0500601609288033d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 14:44:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"669f709ba9497dbe680064a2f4bac8d100ef0dc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 16:25:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a1d44f44a8f24d7582d2fdff5742dcd7 : SUCCESS in 11m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6bc12807dd5546859b7fe327f32dc195 : SUCCESS in 1h 17m 52s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/be836cedba6f4b14b9280746f7dd68fa : SUCCESS in 1h 40m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f7bf26d985c4e70961187395b4b4c69 : SUCCESS in 1h 33m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f773d69ddca422fa8a668f210dcdb68 : SUCCESS in 1h 18m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e625766188d54960b43f1043fb81c749 : SUCCESS in 1h 14m 05s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d9a06c553ba4602a2d9f006315fb274 : SUCCESS in 1h 05m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/642e251740dd493e8cd27057e5e2454b : SUCCESS in 1h 11m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71ab98cff05a4ffeb653132ff8c9a4ae : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":2},{"id":"beda59a0d67a47c6502510976efc0246d25268ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 16:25:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"666183832dd4e8193b441c37ca758f482062145b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 16:36:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18654bbe88754d25b396a0998f999e22 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6bb320d39d7259c22e69cebcd218f71e98cc7192","revisions":{"30ef0925071be06d5bde9f355745c51ce23a9c34":{"kind":"REWORK","_number":1,"created":"2022-02-03 10:32:19.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/22/827622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/22/827622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/22/827622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/22/827622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/22/827622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/22/827622/1"}}},"commit":{"parents":[{"commit":"ba15680dc4e8a8c2b8db5c24aaa919dd0b4ceed9","subject":"Fix wsgi_venv path for panko-api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/ba15680dc4e8a8c2b8db5c24aaa919dd0b4ceed9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-02-03 10:31:52.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-02-03 10:31:52.000000000","tz":0},"subject":"Remove legacy centos-8 CI jobs","message":"Remove legacy centos-8 CI jobs\n\nChange-Id: Iae032c759c6100eacedbd63b85aaf20a321ddae4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/30ef0925071be06d5bde9f355745c51ce23a9c34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/30ef0925071be06d5bde9f355745c51ce23a9c34"}]},"branch":"refs/heads/stable/ussuri"},"6bb320d39d7259c22e69cebcd218f71e98cc7192":{"kind":"REWORK","_number":2,"created":"2022-02-03 11:15:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/22/827622/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/22/827622/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/22/827622/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/22/827622/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/22/827622/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/22/827622/2"}}},"commit":{"parents":[{"commit":"ba15680dc4e8a8c2b8db5c24aaa919dd0b4ceed9","subject":"Fix wsgi_venv path for panko-api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/ba15680dc4e8a8c2b8db5c24aaa919dd0b4ceed9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-02-03 10:31:52.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-03 11:15:35.000000000","tz":0},"subject":"Remove legacy centos-8 CI jobs","message":"Remove legacy centos-8 CI jobs\n\nChange-Id: Iae032c759c6100eacedbd63b85aaf20a321ddae4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/6bb320d39d7259c22e69cebcd218f71e98cc7192"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/6bb320d39d7259c22e69cebcd218f71e98cc7192"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
