)]}'
{"id":"openstack%2Fopenstack-ansible-os_panko~913044","triplet_id":"openstack%2Fopenstack-ansible-os_panko~unmaintained%2Fvictoria~Idf42b428bef9eb6a1df0c07dcc564d0ebbf1117c","project":"openstack/openstack-ansible-os_panko","branch":"unmaintained/victoria","topic":"create-victoria","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-02-03 09:45:48.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idf42b428bef9eb6a1df0c07dcc564d0ebbf1117c","subject":"Update .gitreview for unmaintained/victoria","status":"ABANDONED","created":"2024-03-14 12:27:06.000000000","updated":"2025-02-03 09:45:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"849c2889e5a0746d95f60b74c531d573a09e96cf","_number":913044,"virtual_id_number":913044,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-04-17 22:03:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-14 12:59:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-03 09:45:48.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"063c8478360ab68868d32c5c7c581d72a8bd673f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-03-14 12:27:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af0b4c8f3e71fa36463321c35f0543bff093483e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 12:59:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b7c5b02d3154a208116ce8886482096\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cc29689d75934a78af63ac0e513bab64 : FAILURE in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac8dfc47459f441eb6be71840917feb2 : FAILURE in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"7f4c4cef7609dde40d66526ce2d8111b4a88e78e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-04-17 20:11:55.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bd8136640edac3d6477234c1828cfb4db018370c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 20:27:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eecf545bc68f44629fed11813e319cb4\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f6f08ed2879a4f5f98d6bcef9b162142 : SUCCESS in 9m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec68c1ff3cfd455c94b22bcf9dd0e82d : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7b27afb82b2445691db2c1a199e0427 : FAILURE in 2m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2529564720b4379f3f81f38b2c720d3192cc7eaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-04-17 20:58:56.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3b19e09d033b2fb8edfc65f713588f6a8aba39d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 22:03:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38746d0e62a04d4da3fd2cf92f303905\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3a4482192d0e445ea47d114d0539183f : SUCCESS in 7m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9c1063b65c0431d857b35510c19a30f : FAILURE in 29m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a3e922a5e824a11b517b64c500115a2 : FAILURE in 1h 00m 50s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/532ea8447f5a48e9a38248e4d860a272 : FAILURE in 12m 05s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d7d940a118f40fe8867e58b4f05ad59 : FAILURE in 52m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/638a4de5980b4dfba71c33806062fb2e : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":3},{"id":"849c2889e5a0746d95f60b74c531d573a09e96cf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-03 09:45:48.000000000","message":"Abandoned\n\nunmaintained/victoria branch of openstack/openstack-ansible-os_panko transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"640e433f10ae06883cd60a63901c095130371d04","revisions":{"deaec7864bbcade822ca17d33ea56468e69b866b":{"kind":"REWORK","_number":1,"created":"2024-03-14 12:27:06.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/44/913044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/44/913044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/44/913044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/44/913044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/44/913044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/44/913044/1"}}},"commit":{"parents":[{"commit":"dfb635b7f7a82ccbe23c45a281dda12c222ce663","subject":"Fix wsgi_venv path for panko-api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/dfb635b7f7a82ccbe23c45a281dda12c222ce663"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-14 12:27:06.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-14 12:27:06.000000000","tz":0},"subject":"Update .gitreview for unmaintained/victoria","message":"Update .gitreview for unmaintained/victoria\n\nChange-Id: Idf42b428bef9eb6a1df0c07dcc564d0ebbf1117c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/deaec7864bbcade822ca17d33ea56468e69b866b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/deaec7864bbcade822ca17d33ea56468e69b866b"}]},"branch":"refs/heads/unmaintained/victoria"},"036c8e35317fae7977f189bf0e39efb847300af8":{"kind":"REWORK","_number":2,"created":"2024-04-17 20:11:55.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/44/913044/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/44/913044/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/44/913044/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/44/913044/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/44/913044/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/44/913044/2"}}},"commit":{"parents":[{"commit":"dfb635b7f7a82ccbe23c45a281dda12c222ce663","subject":"Fix wsgi_venv path for panko-api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/dfb635b7f7a82ccbe23c45a281dda12c222ce663"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-14 12:27:06.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-04-17 20:11:55.000000000","tz":0},"subject":"Update .gitreview for unmaintained/victoria","message":"Update .gitreview for unmaintained/victoria\n\nChange-Id: Idf42b428bef9eb6a1df0c07dcc564d0ebbf1117c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/036c8e35317fae7977f189bf0e39efb847300af8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/036c8e35317fae7977f189bf0e39efb847300af8"}]},"branch":"refs/heads/unmaintained/victoria"},"640e433f10ae06883cd60a63901c095130371d04":{"kind":"REWORK","_number":3,"created":"2024-04-17 20:58:56.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/44/913044/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_panko","ref":"refs/changes/44/913044/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/44/913044/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/44/913044/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/44/913044/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_panko refs/changes/44/913044/3"}}},"commit":{"parents":[{"commit":"dfb635b7f7a82ccbe23c45a281dda12c222ce663","subject":"Fix wsgi_venv path for panko-api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/dfb635b7f7a82ccbe23c45a281dda12c222ce663"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-14 12:27:06.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-04-17 20:58:56.000000000","tz":0},"subject":"Update .gitreview for unmaintained/victoria","message":"Update .gitreview for unmaintained/victoria\n\nChange-Id: Idf42b428bef9eb6a1df0c07dcc564d0ebbf1117c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/640e433f10ae06883cd60a63901c095130371d04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_panko/commit/640e433f10ae06883cd60a63901c095130371d04"}]},"branch":"refs/heads/unmaintained/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
