)]}'
{"id":"openstack%2Fopenstack-ansible-os_placement~1003550","triplet_id":"openstack%2Fopenstack-ansible-os_placement~master~I64d1b3a7152786526a0071a2638e6c69d424b010","project":"openstack/openstack-ansible-os_placement","branch":"master","topic":"osa/galera_readonly","attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-18 09:27:14.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I64d1b3a7152786526a0071a2638e6c69d424b010","subject":"Implement readonly connection for MariaDB","status":"NEW","created":"2026-09-02 15:14:49.000000000","updated":"2026-09-18 09:27:14.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6db2bf075fe18f38d18d1446d560606062d33e3e","_number":1003550,"virtual_id_number":1003550,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-03 16:23:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-09-18 09:27:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-02 20:40:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-18 09:27:14.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"5a74bc18abd0e852ecb5c031398f8b7da881c9fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-02 15:14:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4faf8149cc921b8b0e96abd0d543df5450265909","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-02 15:48:05.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"042414f33ba79763bf016d2b488e42e179090e01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-02 17:11:07.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"540d591ab7403bebe7e8739d3b775d1b07b6da6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 20:40:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c85d61b36e254df38f245f6c655ef035\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a8df63b70a94098912bb7c8f6373dad : SUCCESS in 14m 32s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/bd1c80ed9fbf4764989f08cdeb191b31 : ERROR Failed to update project openstack/glance in 21s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/c5f74db4c0344aee9338ea8d48ca8711 : SUCCESS in 1h 22m 45s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f56da808458a4a7a963d96164acc99aa : SUCCESS in 1h 11m 39s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a9cbbfac15c546ed8cff34bb09685521 : SUCCESS in 51m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91a3220362fa40f6822110f4a41aab2e : SUCCESS in 1h 17m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/30bd4fe4b5ea449f963016a7d26e7e56 : SUCCESS in 1h 28m 36s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36435ce9ecb845a7b0451462da7a4325 : SUCCESS in 1h 24m 55s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c358925b34324b70a77dbb7f292938bc : ERROR Failed to update project openstack/glance in 12s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d50020e4b16142e888654659fc9c7888 : SUCCESS in 2h 23m 54s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3061f1cf59164d25860cc48de5e7c4b9 : SUCCESS in 2h 39m 04s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57be17ba015a46acb54ed62177ef1f11 : SUCCESS in 1h 30m 49s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/670674b8e76a42818769541b02d1830b : SUCCESS in 1h 24m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1704a5ef332f4eb0be7fa1f89d2e2112 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66d9df013b9649e09c478f06a4f1bbc6 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":3},{"id":"e34144230e852ca71b87588f09d99ec316fede47","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-03 13:33:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cbfa722cf272392af55ace171fa53b7776ad79c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 16:23:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a39a781fe20242508755b71eaa91e0b8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f54205ba5b8a410aafa5cfa9b5ad504f : SUCCESS in 12m 55s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/d6a5fb096cda48d3b8391daed7be4086 : SUCCESS in 53m 27s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/618b48775af54e31be62563c8e84380c : SUCCESS in 1h 27m 27s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/36276f6a91c143d5a6bc243c8504a29f : SUCCESS in 1h 30m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b823cf7c97ea4306b1038cd6688f24fc : SUCCESS in 51m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ace92589f9494666a4df086e9c6793b8 : SUCCESS in 1h 33m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/90f6f8665b40401c881b676a388331ba : SUCCESS in 46m 24s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f9d12925bd74bfc800c02b742a314bb : SUCCESS in 1h 39m 46s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e749c069913949458b7c0344a4138927 : SUCCESS in 2h 00m 45s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6bfb30b21a543b582f286c8e67f2726 : SUCCESS in 2h 17m 22s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/267761dab04f42feb2a255f130677e9d : SUCCESS in 1h 30m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/962954e6168e4a408ff2e3ac52139182 : SUCCESS in 1h 28m 06s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/33780d2d8dfc4532bd3bc7ecb0b876ec : SUCCESS in 1h 21m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce97a5f2b445434395c7f4135c9e9ad7 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10f7e1884ad949ef8016f161311e4679 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":3},{"id":"6db2bf075fe18f38d18d1446d560606062d33e3e","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-09-18 09:27:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3bbe90e126e61c36cdccb59f161fca5796087a64","revisions":{"979c06711e45310a4a59193882198ca4c6c660ba":{"kind":"REWORK","_number":1,"created":"2026-09-02 15:14:49.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/1003550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/50/1003550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/50/1003550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/50/1003550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/50/1003550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/50/1003550/1"}}},"commit":{"parents":[{"commit":"4348b6a3eef01771b00215d26fc13530c88f4d8a","subject":"Add support for Ubuntu 26.04","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/4348b6a3eef01771b00215d26fc13530c88f4d8a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-02 15:14:39.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-02 15:14:39.000000000","tz":120},"subject":"Implement readonly connection for MariaDB","message":"Implement readonly connection for MariaDB\n\nWhen MariaDB is run in a multu-node Galera cluster we add\nslave_connection to the [database] section.\nThis ensures that reads are pointed towards alternative port/frontend on\nHAProxy.\n\nsynchronous_reader is disabled whenever readonly connection is activated\nas otherwise synchronous reads will go through the connection and not\nslave_connection.\nWith that we are enabling wsrep_sync_wait to ensure consistency on\nMariaDB.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/1003535\nChange-Id: I64d1b3a7152786526a0071a2638e6c69d424b010\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/979c06711e45310a4a59193882198ca4c6c660ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/979c06711e45310a4a59193882198ca4c6c660ba"}]},"branch":"refs/heads/master"},"c711b6fdaa2069d7e578b392ce3d31fd77029f98":{"kind":"REWORK","_number":2,"created":"2026-09-02 15:48:05.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/1003550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/50/1003550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/50/1003550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/50/1003550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/50/1003550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/50/1003550/2"}}},"commit":{"parents":[{"commit":"4348b6a3eef01771b00215d26fc13530c88f4d8a","subject":"Add support for Ubuntu 26.04","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/4348b6a3eef01771b00215d26fc13530c88f4d8a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-02 15:14:39.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-02 15:48:04.000000000","tz":0},"subject":"Implement readonly connection for MariaDB","message":"Implement readonly connection for MariaDB\n\nWhen MariaDB is run in a multu-node Galera cluster we add\nslave_connection to the [database] section.\nThis ensures that reads are pointed towards alternative port/frontend on\nHAProxy.\n\nsynchronous_reader is disabled whenever readonly connection is activated\nas otherwise synchronous reads will go through the connection and not\nslave_connection.\nWith that we are enabling wsrep_sync_wait to ensure consistency on\nMariaDB.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/1003535\nChange-Id: I64d1b3a7152786526a0071a2638e6c69d424b010\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/c711b6fdaa2069d7e578b392ce3d31fd77029f98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/c711b6fdaa2069d7e578b392ce3d31fd77029f98"}]},"branch":"refs/heads/master"},"3bbe90e126e61c36cdccb59f161fca5796087a64":{"kind":"REWORK","_number":3,"created":"2026-09-02 17:11:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/1003550/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/50/1003550/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/50/1003550/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/50/1003550/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/50/1003550/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/50/1003550/3"}}},"commit":{"parents":[{"commit":"4348b6a3eef01771b00215d26fc13530c88f4d8a","subject":"Add support for Ubuntu 26.04","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/4348b6a3eef01771b00215d26fc13530c88f4d8a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-02 15:14:39.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-02 17:11:07.000000000","tz":0},"subject":"Implement readonly connection for MariaDB","message":"Implement readonly connection for MariaDB\n\nWhen MariaDB is run in a multu-node Galera cluster we add\nslave_connection to the [database] section.\nThis ensures that reads are pointed towards alternative port/frontend on\nHAProxy.\n\nsynchronous_reader is disabled whenever readonly connection is activated\nas otherwise synchronous reads will go through the connection and not\nslave_connection.\nWith that we are enabling wsrep_sync_wait to ensure consistency on\nMariaDB.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/1003535\nChange-Id: I64d1b3a7152786526a0071a2638e6c69d424b010\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/3bbe90e126e61c36cdccb59f161fca5796087a64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/3bbe90e126e61c36cdccb59f161fca5796087a64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
