)]}'
{"id":"openstack%2Fopenstack-ansible-os_placement~654374","triplet_id":"openstack%2Fopenstack-ansible-os_placement~master~I72d4ee80003900a9f939e7cdb0025d0b393a4370","project":"openstack/openstack-ansible-os_placement","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I72d4ee80003900a9f939e7cdb0025d0b393a4370","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 07:15:51.000000000","updated":"2019-04-22 11:39:54.000000000","submitted":"2019-04-22 11:39:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654374-1555933194826-61dee02a","meta_rev_id":"88b46548d86a63926de3da9b687da906aa47e9f1","_number":654374,"virtual_id_number":654374,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-04-22 11:39:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-04-22 10:55:14.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2019-04-22 10:55:14.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 10:55:14.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-04-22 11:39:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"81700edd9fb36e20cddbc4fe21e137710601b890","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 07:15:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f0892dc9acc02e78fb5a4525aff865b47a13138","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 08:31:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/74/654374/1/check/openstack-ansible-linters/8af4199/ : SUCCESS in 5m 53s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/74/654374/1/check/openstack-ansible-functional-debian-stable/38363df/ : SUCCESS in 23m 24s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/74/654374/1/check/openstack-ansible-functional-centos-7/eb0906d/ : SUCCESS in 42m 59s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/74/654374/1/check/openstack-ansible-functional-ubuntu-bionic/5d8e835/ : SUCCESS in 23m 00s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/74/654374/1/check/openstack-ansible-functional-distro_install-centos-7/49c0bb4/ : SUCCESS in 25m 37s\n- openstack-tox-docs http://logs.openstack.org/74/654374/1/check/openstack-tox-docs/356b393/html/ : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"437f014f8595dc440a40d1712a29542d330888ad","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-04-22 10:55:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"03836cd058ef4fb7985180c2a2912f91370b0c27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 10:55:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"609599ce32ebf79bf66fd1cf4e3c0a3d44b3d3a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 11:39:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f7324ba45ad26df0f9320bfbe2b2613f28890e18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 11:39:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/74/654374/1/gate/openstack-ansible-linters/7970508/ : SUCCESS in 5m 32s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/74/654374/1/gate/openstack-ansible-functional-debian-stable/cd8f910/ : SUCCESS in 24m 14s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/74/654374/1/gate/openstack-ansible-functional-centos-7/aa3f04b/ : SUCCESS in 29m 23s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/74/654374/1/gate/openstack-ansible-functional-ubuntu-bionic/0c814b1/ : SUCCESS in 22m 36s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/74/654374/1/gate/openstack-ansible-functional-distro_install-centos-7/9d50fec/ : SUCCESS in 24m 04s\n- openstack-tox-docs http://logs.openstack.org/74/654374/1/gate/openstack-tox-docs/ccb24eb/html/ : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b707288393dd49c4bca8539bc7de41d5ca04ea95","revisions":{"b707288393dd49c4bca8539bc7de41d5ca04ea95":{"kind":"REWORK","_number":1,"created":"2019-04-22 07:15:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/654374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/74/654374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/74/654374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/74/654374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/74/654374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/74/654374/1"}}},"commit":{"parents":[{"commit":"d49b184c4026d024f30cc484fd74faecc14b8304","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/d49b184c4026d024f30cc484fd74faecc14b8304"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:15:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:15:49.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I72d4ee80003900a9f939e7cdb0025d0b393a4370\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/b707288393dd49c4bca8539bc7de41d5ca04ea95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/b707288393dd49c4bca8539bc7de41d5ca04ea95"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
