)]}'
{"id":"openstack%2Fopenstack-ansible-os_placement~657088","triplet_id":"openstack%2Fopenstack-ansible-os_placement~master~Id6b11e94e57d45151deb0aeb57761e5117eadd86","project":"openstack/openstack-ansible-os_placement","branch":"master","topic":"placement_integration","hashtags":[],"change_id":"Id6b11e94e57d45151deb0aeb57761e5117eadd86","subject":"Switch authenticate_uri to use v3","status":"ABANDONED","created":"2019-05-04 00:30:15.000000000","updated":"2019-05-04 21:47:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fd6695e557747b4459692025e10a9f1789db7d4d","_number":657088,"virtual_id_number":657088,"owner":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-05-04 01:39:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-04 01:39:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f62efe3ce3963103ac477bff305089e7a216846","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-05-04 00:30:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f035639da005e8d752cec51e531396acc51ac8eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 01:39:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/88/657088/1/check/openstack-ansible-linters/2866055/ : SUCCESS in 4m 16s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/88/657088/1/check/openstack-ansible-functional-debian-stable/4ff6d79/ : SUCCESS in 25m 16s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/88/657088/1/check/openstack-ansible-functional-centos-7/c674303/ : SUCCESS in 35m 07s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/88/657088/1/check/openstack-ansible-functional-ubuntu-bionic/b3faa48/ : SUCCESS in 24m 06s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/88/657088/1/check/openstack-ansible-functional-distro_install-centos-7/2bbb187/ : SUCCESS in 23m 45s\n- openstack-tox-docs http://logs.openstack.org/88/657088/1/check/openstack-tox-docs/2490473/html/ : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3b6569bd663fb26dae2b83c0969972077ba3a632","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-04 21:47:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b77c2ec5dab533167d9ae923dcdd17e653dd704f","revisions":{"b77c2ec5dab533167d9ae923dcdd17e653dd704f":{"kind":"REWORK","_number":1,"created":"2019-05-04 00:30:15.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/88/657088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/88/657088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/88/657088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/88/657088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/88/657088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/88/657088/1"}}},"commit":{"parents":[{"commit":"49cbbc66bb10c928179ec9830934c16b796e75b1","subject":"Merge \"Fix author-email in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/49cbbc66bb10c928179ec9830934c16b796e75b1"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-05-04 00:29:35.000000000","tz":0},"committer":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-05-04 00:29:35.000000000","tz":0},"subject":"Switch authenticate_uri to use v3","message":"Switch authenticate_uri to use v3\n\nChange-Id: Id6b11e94e57d45151deb0aeb57761e5117eadd86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/b77c2ec5dab533167d9ae923dcdd17e653dd704f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/b77c2ec5dab533167d9ae923dcdd17e653dd704f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
