)]}'
{"id":"openstack%2Fopenstack-ansible-os_placement~665733","triplet_id":"openstack%2Fopenstack-ansible-os_placement~master~Ic9a12d7f04288f838cfc80c2b7725d20fec1dde9","project":"openstack/openstack-ansible-os_placement","branch":"master","topic":"osa/db-refactor","hashtags":[],"change_id":"Ic9a12d7f04288f838cfc80c2b7725d20fec1dde9","subject":"db_setup: refactor database setup to a common file","status":"MERGED","created":"2019-06-17 17:46:32.000000000","updated":"2019-06-27 14:16:25.000000000","submitted":"2019-06-27 14:16:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665733-1561644985719-e8d46088","meta_rev_id":"d8fa4db4d794f626c4fc9321213ee02f74ec6a9e","_number":665733,"virtual_id_number":665733,"owner":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-06-27 14:16:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2019-06-27 12:38:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2019-06-27 12:29:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2019-06-27 12:38:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 12:29:46.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-27 12:38:51.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2019-06-27 14:16:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f55aac8355ef06ca453658687ac48331326845aa","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-06-17 17:46:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ea610581faf9bed1bfece2686214c4f66602084","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 19:17:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/33/665733/1/check/openstack-ansible-linters/410c0e9/ : SUCCESS in 4m 32s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/33/665733/1/check/openstack-ansible-deploy-aio_metal-debian-stable/f0ac30f/ : SUCCESS in 1h 18m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/33/665733/1/check/openstack-ansible-deploy-aio_metal-centos-7/13fc7bf/ : SUCCESS in 1h 17m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/33/665733/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/5e08cdd/ : SUCCESS in 1h 05m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/33/665733/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/bd03081/ : SUCCESS in 58m 48s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/33/665733/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/7d11313/ : SUCCESS in 59m 28s\n- openstack-tox-docs http://logs.openstack.org/33/665733/1/check/openstack-tox-docs/5341429/html/ : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"10c22678da1defc5f06173ea72730337b12e32b2","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-27 12:29:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2f318f103004aa77b959936e65f3feb4dee964ea","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-06-27 12:38:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f29ec0c675fdad0b9a8ee3d555b320193b59fd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 12:39:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2cde843110c4ab223f789c4d07c157fa9deb9148","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:16:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9e3cfe136ab8664959d4139373068b56b4bd6b8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:16:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/33/665733/1/gate/openstack-ansible-linters/c36f536/ : SUCCESS in 7m 17s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/33/665733/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/5faf4b4/ : SUCCESS in 1h 03m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/33/665733/1/gate/openstack-ansible-deploy-aio_metal-centos-7/ad59902/ : SUCCESS in 1h 32m 35s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/33/665733/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/08c1464/ : SUCCESS in 1h 13m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/33/665733/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/ed40285/ : SUCCESS in 1h 13m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/33/665733/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/97511e0/ : SUCCESS in 1h 12m 33s\n- openstack-tox-docs http://logs.openstack.org/33/665733/1/gate/openstack-tox-docs/dbd2688/html/ : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2bdc2eebb248019e17a1cec80dc364cef529c780","revisions":{"2bdc2eebb248019e17a1cec80dc364cef529c780":{"kind":"REWORK","_number":1,"created":"2019-06-17 17:46:32.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/33/665733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/33/665733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/33/665733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/33/665733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/33/665733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/33/665733/1"}}},"commit":{"parents":[{"commit":"bbae385394aa1c1514c99c1c24e65cee18f6d016","subject":"Add Debian placement_service_distro_packages list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/bbae385394aa1c1514c99c1c24e65cee18f6d016"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-06-17 17:45:43.000000000","tz":0},"committer":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-06-17 17:45:43.000000000","tz":0},"subject":"db_setup: refactor database setup to a common file","message":"db_setup: refactor database setup to a common file\n\nThis patch refactors the database creation to db_setup.yml which\nwill eventually be managed by openstack-ansible-tests.\n\n This also re-orders the mq_setup to be done earlier so these system\nlevel dependencies are ready before service activation.\n\nChange-Id: Ic9a12d7f04288f838cfc80c2b7725d20fec1dde9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/2bdc2eebb248019e17a1cec80dc364cef529c780"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/2bdc2eebb248019e17a1cec80dc364cef529c780"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
