)]}'
{"id":"openstack%2Fopenstack-ansible-os_placement~666298","triplet_id":"openstack%2Fopenstack-ansible-os_placement~master~Iace7d8c0342e745b54a1a17dbeb349cf3217bc3e","project":"openstack/openstack-ansible-os_placement","branch":"master","hashtags":[],"change_id":"Iace7d8c0342e745b54a1a17dbeb349cf3217bc3e","subject":"Fix aio_distro_metal jobs for openSUSE","status":"MERGED","created":"2019-06-19 14:04:01.000000000","updated":"2019-06-27 14:32:33.000000000","submitted":"2019-06-27 14:32:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666298-1561645953976-1ef4935a","meta_rev_id":"5c69a6dd32088f634af284d59f66a7fe38a7e4e0","_number":666298,"virtual_id_number":666298,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-27 11:35:01.000000000","_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-27 14:32:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-06-27 10:21:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-27 09:46:52.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2019-06-27 10:50:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 09:46:52.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-06-27 10:50:04.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-27 14:32:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1aabd20e61160987428fa0b6b21edd36674a7c4d","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-19 14:04:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f3c9cf932432ff3db34e4e3442df313ef4a228e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 15:32:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/98/666298/1/check/openstack-ansible-linters/3f23efe/ : SUCCESS in 5m 46s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/98/666298/1/check/openstack-ansible-deploy-aio_metal-debian-stable/c6b2e5f/ : SUCCESS in 1h 03m 17s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/98/666298/1/check/openstack-ansible-deploy-aio_metal-centos-7/9d9c81a/ : SUCCESS in 1h 17m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/98/666298/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/5de1680/ : FAILURE in 14m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/98/666298/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/3b6c9c6/ : SUCCESS in 58m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/98/666298/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/cc82279/ : SUCCESS in 1h 04m 24s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/98/666298/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/f3ed55c/ : FAILURE in 13m 44s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/98/666298/1/check/openstack-tox-docs/fe412d0/html/ : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"9e9c627460a7fff27bb6d8e323bf52b1f8c4ad1f","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-26 18:37:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d48c7204bf4253267dac64d349ed9524cdab0119","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 20:02:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/98/666298/1/check/openstack-ansible-linters/1b29241/ : SUCCESS in 5m 11s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/98/666298/1/check/openstack-ansible-deploy-aio_metal-debian-stable/bb56a45/ : SUCCESS in 1h 03m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/98/666298/1/check/openstack-ansible-deploy-aio_metal-centos-7/ec7727d/ : SUCCESS in 1h 18m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/98/666298/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/93e55f9/ : FAILURE in 19m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/98/666298/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/4438ab2/ : SUCCESS in 59m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/98/666298/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/5537e73/ : SUCCESS in 55m 27s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/98/666298/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/22b8f19/ : FAILURE in 18m 21s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/98/666298/1/check/openstack-tox-docs/ab1a9c2/html/ : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1},{"id":"67c3e9e83afdfa820eb1a099a2e1d29322964514","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-27 07:03:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"33cfcbc9547f36b13a7ac241a1363e3aaf8301c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 08:36:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/98/666298/1/check/openstack-ansible-linters/2ddcb7c/ : SUCCESS in 5m 01s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/98/666298/1/check/openstack-ansible-deploy-aio_metal-debian-stable/0b6c2e6/ : SUCCESS in 59m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/98/666298/1/check/openstack-ansible-deploy-aio_metal-centos-7/c7d5fc6/ : SUCCESS in 1h 23m 19s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/98/666298/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/977a017/ : SUCCESS in 1h 08m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/98/666298/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/3a773d6/ : SUCCESS in 56m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/98/666298/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/e824413/ : SUCCESS in 57m 38s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/98/666298/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/a3aa743/ : SUCCESS in 47m 39s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/98/666298/1/check/openstack-tox-docs/a0ba923/html/ : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"fd83a4476eaaa71220f986b7b221a0f55b5e24bc","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-27 09:43:05.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"15d3814c70d21bbe12d5f316fccc80e6f70488f9","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-27 09:46:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a89150a268b1b524edec32647fb70d5be1b1a0a2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-27 10:21:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c97f115918595cca6f95117a094684c4a5ee98f5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-27 10:31:41.000000000","message":"Removed Workflow+1 by Dmitriy Rabotyagov (noonedeadpunk) \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"d77bc1874a0a4f70420f4c54653929c15df232b9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-27 10:50:04.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"56ee57c85744b1ff8d815d6dc844e66e67eab4bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 11:13:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/98/666298/2/check/openstack-ansible-linters/8e5ad52/ : SUCCESS in 5m 19s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/98/666298/2/check/openstack-ansible-deploy-aio_metal-debian-stable/aad40bc/ : SUCCESS in 1h 00m 22s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/98/666298/2/check/openstack-ansible-deploy-aio_metal-centos-7/486859d/ : SUCCESS in 1h 28m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/98/666298/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/5c18903/ : SUCCESS in 1h 09m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/98/666298/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/f313868/ : SUCCESS in 59m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/98/666298/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/7e3580d/ : FAILURE in 18m 29s\n- openstack-tox-docs http://logs.openstack.org/98/666298/2/check/openstack-tox-docs/644dad8/html/ : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"02c398c4a10f463210ad5545bdc5c315d275753c","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-27 11:35:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4612602afdce7dbb6f4f7d157aed2de896e814d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:07:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/98/666298/2/check/openstack-ansible-linters/0835be4/ : SUCCESS in 4m 41s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/98/666298/2/check/openstack-ansible-deploy-aio_metal-debian-stable/96a8db6/ : SUCCESS in 1h 00m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/98/666298/2/check/openstack-ansible-deploy-aio_metal-centos-7/d780f26/ : SUCCESS in 1h 19m 08s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/98/666298/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/4fc46eb/ : SUCCESS in 1h 14m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/98/666298/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/151cb8f/ : SUCCESS in 58m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/98/666298/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/94e0dad/ : SUCCESS in 58m 05s\n- openstack-tox-docs http://logs.openstack.org/98/666298/2/check/openstack-tox-docs/70c0e48/html/ : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":2},{"id":"d96dfca1069f4c5a53a2f533d55cf3fdd1bd63b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:07:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0d4a880462ed5fd66a8afcd431c507a0a15cf1f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:32:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5c69a6dd32088f634af284d59f66a7fe38a7e4e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:32:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/98/666298/2/gate/openstack-ansible-linters/7455607/ : SUCCESS in 6m 07s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/98/666298/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/ca7e10b/ : SUCCESS in 1h 05m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/98/666298/2/gate/openstack-ansible-deploy-aio_metal-centos-7/bdaf7aa/ : SUCCESS in 1h 22m 47s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/98/666298/2/gate/openstack-ansible-deploy-aio_metal-opensuse-150/fd2a86c/ : SUCCESS in 1h 08m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/98/666298/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/a8950e7/ : SUCCESS in 1h 09m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/98/666298/2/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/5f86e36/ : SUCCESS in 1h 00m 14s\n- openstack-tox-docs http://logs.openstack.org/98/666298/2/gate/openstack-tox-docs/d2ef5cb/html/ : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"259099f208047a19689b1e8f9b0b1da0013d6406","revisions":{"840aeda48d881b5bed1b639ad7f5e35e407708c8":{"kind":"REWORK","_number":1,"created":"2019-06-19 14:04:01.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/98/666298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/98/666298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/98/666298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/98/666298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/98/666298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/98/666298/1"}}},"commit":{"parents":[{"commit":"bbae385394aa1c1514c99c1c24e65cee18f6d016","subject":"Add Debian placement_service_distro_packages list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/bbae385394aa1c1514c99c1c24e65cee18f6d016"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-19 14:02:43.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-19 14:03:32.000000000","tz":120},"subject":"Fix aio_distro_metal jobs for openSUSE","message":"Fix aio_distro_metal jobs for openSUSE\n\nPackage list was missing the placement-api dependency.\n\nDepends-On: https://review.opendev.org/#/c/666058/\nChange-Id: Iace7d8c0342e745b54a1a17dbeb349cf3217bc3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/840aeda48d881b5bed1b639ad7f5e35e407708c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/840aeda48d881b5bed1b639ad7f5e35e407708c8"}]},"branch":"refs/heads/master"},"259099f208047a19689b1e8f9b0b1da0013d6406":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-27 09:43:05.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/98/666298/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/98/666298/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/98/666298/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/98/666298/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/98/666298/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/98/666298/2"}}},"commit":{"parents":[{"commit":"bbae385394aa1c1514c99c1c24e65cee18f6d016","subject":"Add Debian placement_service_distro_packages list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/bbae385394aa1c1514c99c1c24e65cee18f6d016"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-19 14:02:43.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-27 09:42:55.000000000","tz":120},"subject":"Fix aio_distro_metal jobs for openSUSE","message":"Fix aio_distro_metal jobs for openSUSE\n\nPackage list was missing the placement-api dependency.\n\nChange-Id: Iace7d8c0342e745b54a1a17dbeb349cf3217bc3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/259099f208047a19689b1e8f9b0b1da0013d6406"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/259099f208047a19689b1e8f9b0b1da0013d6406"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
