)]}'
{"id":"openstack%2Fopenstack-ansible-os_placement~666360","triplet_id":"openstack%2Fopenstack-ansible-os_placement~master~I96c77834f121a0a7fa03d49d6f1da7531aa3b834","project":"openstack/openstack-ansible-os_placement","branch":"master","hashtags":[],"change_id":"I96c77834f121a0a7fa03d49d6f1da7531aa3b834","subject":"Sync Sphinx requirement","status":"ABANDONED","created":"2019-06-19 16:00:39.000000000","updated":"2019-07-18 13:30:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"20feafb2a56661188c17bac5f725571a5da38859","_number":666360,"virtual_id_number":666360,"owner":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":1,"date":"2019-06-20 14:13:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":1,"date":"2019-07-14 11:43:56.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-16 07:54:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 14:13:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-14 11:43:56.000000000","updated_by":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"reviewer":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"state":"REVIEWER"},{"updated":"2019-07-16 07:54:22.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"ad1b7fee7a8aac9978217a41843780d16b534407","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-06-19 16:00:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d614694f4f2c4e54046999b5921c1cdea0fcdff0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 16:20:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/60/666360/1/check/requirements-check/756c77c/ : SUCCESS in 2m 21s\n- openstack-ansible-linters http://logs.openstack.org/60/666360/1/check/openstack-ansible-linters/ea1b553/ : SUCCESS in 6m 26s\n- openstack-tox-docs http://logs.openstack.org/60/666360/1/check/openstack-tox-docs/3c2b356/html/ : SUCCESS in 2m 39s\n- build-openstack-releasenotes http://logs.openstack.org/60/666360/1/check/build-openstack-releasenotes/b4496a4/html/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2ce41a5a43e77adce80a54d2e81ab42e2ea226cf","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-06-20 14:02:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"61595084130c8ab0d8cabc7fac6448b4bea3d3e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 14:13:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/60/666360/2/check/requirements-check/0864304/ : SUCCESS in 1m 58s\n- openstack-ansible-linters http://logs.openstack.org/60/666360/2/check/openstack-ansible-linters/ad25949/ : SUCCESS in 5m 12s\n- openstack-tox-docs http://logs.openstack.org/60/666360/2/check/openstack-tox-docs/6b08230/html/ : SUCCESS in 2m 55s\n- build-openstack-releasenotes http://logs.openstack.org/60/666360/2/check/build-openstack-releasenotes/cee0af7/html/ : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":2},{"id":"2e6d2064f0a99c96f626563dc3391a409fe06e1b","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-07-14 11:43:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c0c9435879c944c6a5e408e3987faff46d8b997f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-16 07:54:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"20feafb2a56661188c17bac5f725571a5da38859","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-18 13:30:02.000000000","message":"Abandoned\n\nalready merged in other patch","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"23b8ed5cb891179f5f95bd7d685e1b2af796b1e3","revisions":{"e664475d20634165507b3a597bbe3e889e55db9e":{"kind":"REWORK","_number":1,"created":"2019-06-19 16:00:39.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/60/666360/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/60/666360/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/60/666360/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/60/666360/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/60/666360/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/60/666360/1"}}},"commit":{"parents":[{"commit":"bbae385394aa1c1514c99c1c24e65cee18f6d016","subject":"Add Debian placement_service_distro_packages list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/bbae385394aa1c1514c99c1c24e65cee18f6d016"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-06-19 16:00:23.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-06-19 16:00:36.000000000","tz":480},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\nChange-Id: I96c77834f121a0a7fa03d49d6f1da7531aa3b834\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/e664475d20634165507b3a597bbe3e889e55db9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/e664475d20634165507b3a597bbe3e889e55db9e"}]},"branch":"refs/heads/master"},"23b8ed5cb891179f5f95bd7d685e1b2af796b1e3":{"kind":"REWORK","_number":2,"created":"2019-06-20 14:02:27.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/60/666360/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/60/666360/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/60/666360/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/60/666360/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/60/666360/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/60/666360/2"}}},"commit":{"parents":[{"commit":"bbae385394aa1c1514c99c1c24e65cee18f6d016","subject":"Add Debian placement_service_distro_packages list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/bbae385394aa1c1514c99c1c24e65cee18f6d016"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-06-19 16:00:23.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-06-20 14:01:49.000000000","tz":480},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\n1. Sync sphinx dependency with global requirements. It caps python 2 since\nsphinx 2.0 no longer supports Python 2.7.\n2. Remove unncessary \"\u003d\u003d\"\n\nChange-Id: I96c77834f121a0a7fa03d49d6f1da7531aa3b834\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/23b8ed5cb891179f5f95bd7d685e1b2af796b1e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/23b8ed5cb891179f5f95bd7d685e1b2af796b1e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
