)]}'
{"id":"openstack%2Fopenstack-ansible-os_placement~669691","triplet_id":"openstack%2Fopenstack-ansible-os_placement~master~Ic9ea32ea34af157b127ab84517c1ecefaac28ab8","project":"openstack/openstack-ansible-os_placement","branch":"master","topic":"deprecation-warning","hashtags":[],"change_id":"Ic9ea32ea34af157b127ab84517c1ecefaac28ab8","subject":"Fix deprecation warning","status":"MERGED","created":"2019-07-08 14:45:59.000000000","updated":"2019-07-10 17:58:31.000000000","submitted":"2019-07-10 17:58:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669691-1562781511443-b7e4bc0b","meta_rev_id":"fac36ef998cd0d29e2c25759500ad945ad8f3a35","_number":669691,"virtual_id_number":669691,"owner":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-07-10 17:58:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-10 16:00:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-10 12:47:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-10 16:00:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 12:47:42.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-10 16:00:43.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-10 17:58:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"696fa03563a54800a2954532993f0cacd7d76083","author":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"date":"2019-07-08 14:45:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f956427c31d9a373a7c518ef41b8fae4ff8939cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 16:32:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/91/669691/1/check/openstack-ansible-linters/3e93475/ : SUCCESS in 4m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/91/669691/1/check/openstack-ansible-deploy-aio_metal-debian-stable/e00cfcb/ : SUCCESS in 1h 02m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/91/669691/1/check/openstack-ansible-deploy-aio_metal-centos-7/a9bed9a/ : SUCCESS in 1h 26m 52s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/91/669691/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/b211712/ : SUCCESS in 1h 23m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/91/669691/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/4006b4e/ : SUCCESS in 1h 04m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/91/669691/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/5ce9b6a/ : SUCCESS in 1h 09m 08s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/91/669691/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/118e053/ : SUCCESS in 1h 03m 51s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/91/669691/1/check/openstack-tox-docs/298371d/html/ : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d6d5c9933a1657a22d2bd13199d7dde32a9045d0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-10 10:36:43.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d509e901322e8d46ea39a6bffb5c8119bdc5a889","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-10 10:43:14.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"dce5f93d1853a48377b443925d5735a1041ec9ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 12:34:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/91/669691/3/check/openstack-ansible-linters/fb23b21/ : SUCCESS in 8m 25s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/91/669691/3/check/openstack-ansible-deploy-aio_metal-debian-stable/9000469/ : SUCCESS in 1h 15m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/91/669691/3/check/openstack-ansible-deploy-aio_metal-centos-7/2f12ebb/ : SUCCESS in 1h 21m 32s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/91/669691/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/ddc6300/ : SUCCESS in 1h 24m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/91/669691/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/e1557bd/ : SUCCESS in 1h 11m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/91/669691/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/66c8bd0/ : SUCCESS in 1h 07m 48s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/91/669691/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/cab088c/ : FAILURE in 31m 54s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/91/669691/3/check/openstack-tox-docs/ccc8464/html/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":3},{"id":"fd1830039799b3135f3b6724d9292aca6fc0d10b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-10 12:47:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ca3e9548ca1335c170646aea33c1740f652c2dfd","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-10 16:00:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"029b969d1bf05286dc920b6d839e3fce4a167ef8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:00:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c8d8f24cb7f9a62421cc60d5cef94a833428663d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 17:58:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/91/669691/3/gate/openstack-ansible-linters/d3ec415/ : SUCCESS in 7m 06s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/91/669691/3/gate/openstack-ansible-deploy-aio_metal-debian-stable/d703156/ : SUCCESS in 1h 01m 22s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/91/669691/3/gate/openstack-ansible-deploy-aio_metal-centos-7/fd156da/ : SUCCESS in 1h 09m 50s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/91/669691/3/gate/openstack-ansible-deploy-aio_metal-opensuse-15/a3a13d6/ : SUCCESS in 1h 17m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/91/669691/3/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/b36bbd8/ : SUCCESS in 1h 00m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/91/669691/3/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/91a65d6/ : SUCCESS in 58m 32s\n- openstack-tox-docs http://logs.openstack.org/91/669691/3/gate/openstack-tox-docs/66cda9e/html/ : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":3},{"id":"f14791231b02782aa404739d3eeb20d395738be2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 17:58:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"77e709668675c02e8f988b36bb4b1eea94913436","revisions":{"73ea94fd8e014ebc0720c0e595391efc6af53552":{"kind":"REWORK","_number":1,"created":"2019-07-08 14:45:59.000000000","uploader":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"ref":"refs/changes/91/669691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/91/669691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/91/669691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/91/669691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/91/669691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/91/669691/1"}}},"commit":{"parents":[{"commit":"07f52166339dbe3e1e0a9306289cac1ed90f4002","subject":"Merge \"Fix aio_distro_metal jobs for openSUSE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/07f52166339dbe3e1e0a9306289cac1ed90f4002"}]}],"author":{"name":"Jonathan Herlin","email":"jonte@jherlin.se","date":"2019-07-08 14:45:06.000000000","tz":120},"committer":{"name":"Jonathan Herlin","email":"jonte@jherlin.se","date":"2019-07-08 14:46:14.000000000","tz":120},"subject":"Fix deprecation warning: ----","message":"Fix deprecation warning:\n----\n\nTASK [os_placement : Record the need for a service restart] ********************\n[DEPRECATION WARNING]: Using tests as filters is deprecated. Instead of using\n`result|changed` use `result is changed`. This feature will be removed in\nversion 2.9. Deprecation warnings can be disabled by setting\ndeprecation_warnings\u003dFalse in ansible.cfg.\n\nChange-Id: Ic9ea32ea34af157b127ab84517c1ecefaac28ab8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/73ea94fd8e014ebc0720c0e595391efc6af53552"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/73ea94fd8e014ebc0720c0e595391efc6af53552"}]},"branch":"refs/heads/master"},"ddc68baa869098e223c5e450c964dac9619f9f0c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-10 10:36:43.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/91/669691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/91/669691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/91/669691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/91/669691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/91/669691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/91/669691/2"}}},"commit":{"parents":[{"commit":"07f52166339dbe3e1e0a9306289cac1ed90f4002","subject":"Merge \"Fix aio_distro_metal jobs for openSUSE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/07f52166339dbe3e1e0a9306289cac1ed90f4002"}]}],"author":{"name":"Jonathan Herlin","email":"jonte@jherlin.se","date":"2019-07-08 14:45:06.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-10 10:36:41.000000000","tz":0},"subject":"Fix deprecation warning:","message":"Fix deprecation warning:\n\nTASK [os_placement : Record the need for a service restart] ********************\n[DEPRECATION WARNING]: Using tests as filters is deprecated. Instead of using\n`result|changed` use `result is changed`. This feature will be removed in\nversion 2.9. Deprecation warnings can be disabled by setting\ndeprecation_warnings\u003dFalse in ansible.cfg.\n\nChange-Id: Ic9ea32ea34af157b127ab84517c1ecefaac28ab8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/ddc68baa869098e223c5e450c964dac9619f9f0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/ddc68baa869098e223c5e450c964dac9619f9f0c"}]},"branch":"refs/heads/master"},"77e709668675c02e8f988b36bb4b1eea94913436":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-07-10 10:43:14.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/91/669691/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/91/669691/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/91/669691/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/91/669691/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/91/669691/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/91/669691/3"}}},"commit":{"parents":[{"commit":"07f52166339dbe3e1e0a9306289cac1ed90f4002","subject":"Merge \"Fix aio_distro_metal jobs for openSUSE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/07f52166339dbe3e1e0a9306289cac1ed90f4002"}]}],"author":{"name":"Jonathan Herlin","email":"jonte@jherlin.se","date":"2019-07-08 14:45:06.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-10 10:43:12.000000000","tz":0},"subject":"Fix deprecation warning","message":"Fix deprecation warning\n\nTASK [os_placement : Record the need for a service restart] ********************\n[DEPRECATION WARNING]: Using tests as filters is deprecated. Instead of using\n`result|changed` use `result is changed`. This feature will be removed in\nversion 2.9. Deprecation warnings can be disabled by setting\ndeprecation_warnings\u003dFalse in ansible.cfg.\n\nChange-Id: Ic9ea32ea34af157b127ab84517c1ecefaac28ab8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/77e709668675c02e8f988b36bb4b1eea94913436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/77e709668675c02e8f988b36bb4b1eea94913436"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
